)]}'
{"id":"openstack%2Fkayobe~797497","triplet_id":"openstack%2Fkayobe~master~I764ff019a18266b593add7ab80ee095d7d07a869","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I764ff019a18266b593add7ab80ee095d7d07a869","subject":"Support Ansible collections","status":"MERGED","created":"2021-06-22 15:14:15.000000000","updated":"2021-09-29 18:22:42.000000000","submitted":"2021-09-29 18:19:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797497","meta_rev_id":"efc4d3a8c24e60b7e21729ce8d7bf8c72b797cec","_number":797497,"virtual_id_number":797497,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-29 18:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-29 16:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-27 13:12:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-09-29 16:23:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-22 15:22:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2021-07-23 12:57:22.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2021-07-26 17:22:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-16 10:26:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-09-23 08:50:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-09-27 07:33:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"42844bdda188dd599c783fb3ead0c0d4acd73358","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-22 15:14:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c0d81c316ed845d182df22aea74b8b1a308f1ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-23 11:37:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5560fe4170db9ec9908a13ff16559a00be6c9e5b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-23 12:13:34.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"73478509fece8072ad3110b3b54d2c93da25ca64","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 15:31:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3168627b7273cc865faef2d00991b044a3eebf92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 17:22:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ff84d028c184772b008898e22c008cb : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23dbbe9939a645018f47862e484c34d4 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a757ad76dcae43b1be1fc83e55f627b8 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4614b2defaa4bdf9e00784ed7dc80b6 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edc3019567994f0395cefc82f0b4ad8a : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e3c1ed12db04027b46fdc133e852d58 : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dd737b2c0e149f1b42bfdf23241cce7 : FAILURE in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f670ddb84e4e49f7a1505c98d2aa4e86 : SUCCESS in 7m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/95457ff4ab53402799b2b0d32c336573 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/831d5058b8854ec080bd5a176f42d159 : SUCCESS in 20m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/55a316e750644809840e2781277d6d95 : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/037885ee96f14fd2a4d584560a78f164 : SUCCESS in 47m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9a256bbe1b464ff2a1c3fc70e2cb1cfa : SUCCESS in 1h 20m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6aed950ad07d4db3949c026ccd5cc43b : SUCCESS in 20m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0a983edb2414044a96a7c929a41489c : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e07dfc91391340a2b3f823015840b9b6 : SUCCESS in 1h 49m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/18831313a79a4b7498a950fd2cecf977 : SUCCESS in 33m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e1e0ff8c31642b9b04648ac3a0f42f6 : SUCCESS in 28m 17s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b18bf2bd189d4ed4a10e18564c8e8c9d : SUCCESS in 27m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7147428ade1a40e291ccd0420b6a81af : SUCCESS in 12m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/520556f1ec3f4d858a27f7644748228c : SUCCESS in 10m 00s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"c3f99a2912cda3929ffc48b4801d68ac50042497","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 08:17:46.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"afa78857cc2d99990f76c447b9db7f150e6b8eff","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 08:18:00.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"2d82e8c3293241f88513e74efdef982d88247bd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 10:15:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c11379f1a55841478acb598804bef328 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8e77a27308141daba022a8221f2b7f8 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/050eaab2a98f4818b0b088077ea95218 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/854dfd68073d470690060a87cb174e5f : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2629c80368154287972effa98bbcfdd2 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ed9358679f48c3945b3f282903f7ed : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4ccf151116a46f39de9533d2d56474e : SUCCESS in 4m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/012254c5aa4744d68e39f5a087fb28a6 : SUCCESS in 7m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/59f75aef6e314bf0a6cd98938d63108a : SUCCESS in 11m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/41ce7edf5ae8465abf320b7fb86f6cd4 : SUCCESS in 21m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ab1bfd62e8e742a581a454788240a20d : SUCCESS in 1h 08m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2286123b9a894e3a9da9540a9966450b : SUCCESS in 48m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7e338ad112e6458a8664ff754fa0c6fe : SUCCESS in 1h 11m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c7a190c8dbb847b9807b173d0f38b246 : SUCCESS in 17m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a270ee151c6c4adcafedb800769fbf65 : SUCCESS in 15m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/77b95c537d444fea8daac41d8d8d61f1 : SUCCESS in 1h 52m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6d3f509225914f02991be92a38b9a88a : FAILURE in 16m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bb761f793e5463caab3dfcb5de2104d : SUCCESS in 27m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8c0568666a6f4919848613dd8a47ce7b : SUCCESS in 26m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/be30f122e1194d1d978521507fb86ce5 : SUCCESS in 13m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37a6231e617a4428ad9ebb36db7d2b62 : SUCCESS in 10m 12s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":4},{"id":"a55044524b7d1b5fad55c3b97fe97cf07664c7b2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 11:00:48.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9dca317b02445a4a986cc86f0064ddf07f8301c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 11:50:47.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3565b2838d5f49818007d03dbb61b396 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/087d8684e0bd494091accd5925e3d726 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/558b76b197b54bd4ae6331235b0d4d80 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77470dbd5ef843eb98e8bb3ae41dbcad : SUCCESS in 8m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0905bdece9b7470a845c5fa79f38ba7c : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53512125a02240d8966d829c0a533537 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe3e8b9a46ba43e18a5da5af74dd492b : SUCCESS in 4m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f0d1187045f4b14abce40d5103d5c36 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f8bb685349e042f88054f6cb39c83405 : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1545955ce18f4591ae3138e2c5e655ae : SUCCESS in 18m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0dc2b4c814eb4ac8b82383c86522498b : FAILURE in 44m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f62df4f58ee4e19b174dc3ed20df028 : SUCCESS in 42m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/524e2d948e4948899c253473b9d68e73 : FAILURE in 22m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c58f4a0e42bb407caf3f7cd06f2c0893 : SUCCESS in 17m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50f09a4152304905b1711e771f27bae4 : SUCCESS in 12m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b4034a503fc643929fcbfdafc1d1e83e : FAILURE in 43m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9cc331fd5b2044e296c2fa3c7340a635 : FAILURE in 13m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b746eaae0076434fb02c4ac79fb387fa : SUCCESS in 30m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/95f1d5180bae4fd293edc7e515abdc58 : FAILURE in 17m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/919091984e8e4b8eab913c9af76b90cb : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/595c3718cc884208bfe3f23cf87a6ff2 : SUCCESS in 10m 50s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":4},{"id":"a8ac31b275e477291a98f7165275fc31a1fac625","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-28 13:26:15.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"53b14ae891ba37b3d848db4f78e5621a53101934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 15:07:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d420e1a361da40cdbed2b46cdc485271 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b193dc93e846457cb5cedb3d92dd70be : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/356fa645ba584c0ab47fcf7c20c66e79 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d5b7ef9e98f4e3e8958c38033810a66 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f03fca2cd7ab48c5871c9a860738e9c3 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a7d7f1fa5a4a34bfe4fe6a7fe0fb1b : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dc843457978402bbcbe651c0b8a74c7 : SUCCESS in 4m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d3d0b27f2d9f46e9a2814b0b9113cfd8 : SUCCESS in 7m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c718403684fe42a39cdbe2fe5fad37cd : SUCCESS in 11m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/107d513452e24dc985e8d5953d785f76 : SUCCESS in 20m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b6bda7c754b54da9866759580390e35c : SUCCESS in 1h 12m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c8b2ca7f4314089b7f10ce7592c6893 : SUCCESS in 51m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/30a7992a4d444e7fa30778dca2f43bdb : SUCCESS in 1h 16m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9351b6e114e14bd9bcb6e4d2d84d5bb4 : SUCCESS in 18m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04d75bc527ee46eaadb59023b3e8a558 : SUCCESS in 13m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1d6f3478afba4cf4b7f3f3c40fa81777 : SUCCESS in 1h 37m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fc2b1406cc8b426f932311822e0e7415 : SUCCESS in 35m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bb7d56a163c44bcbdb9c9211d5791c4 : SUCCESS in 30m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8519376b69fa4c72baf48b7e12afab05 : SUCCESS in 29m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3e03031b16b442ec96eccf270b991889 : SUCCESS in 12m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a164a322d82493abc6da012bb4fcae6 : SUCCESS in 11m 00s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":4},{"id":"8b40d4b99e4f27c759aced0293d30906bd67639c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-02 07:39:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e69a3e7af03413f3ad9d5c969d9ab40038fa376f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-02 07:51:40.000000000","message":"Patch Set 5:\n\nFixed merge conflicts.","accounts_in_message":[],"_revision_number":5},{"id":"3a3b5c14ae48729c00c31232e9fc6469043280b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 09:01:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75484cb10a194183af9c488ff8d36133 : TIMED_OUT in 51m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889911f0940e49c2afcfcc32b5530779 : TIMED_OUT in 31m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb8d32b8016a4a0b9f228c839fd24128 : TIMED_OUT in 41m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3bacdd9913849c8841d39a8111c83d7 : TIMED_OUT in 41m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24fcaa4513414a07ba79cebfaad4c1ab : TIMED_OUT in 41m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b9d5f0537d4ff3b277004737d1e8c7 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fd78f92c7cd429094f91cf80a593bc5 : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef607af25f69472a840718565f52c5dc : TIMED_OUT in 31m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/687377597dab4a4ca3e34d1276e4cdb7 : TIMED_OUT in 31m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c04576203c749bcab058a43dbb6613c : TIMED_OUT in 1h 01m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d62e243509ad47fda90c508863c38f87 : FAILURE in 27m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db4068f48f6c4804a00f015088292131 : FAILURE in 28m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0b6301059e2b49f294ea14e1aafe3f24 : FAILURE in 29m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/34d59c690f834627b63caf0d6f19f46b : SUCCESS in 21m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6616c6e5fa9480d87cc1de8b42598c1 : SUCCESS in 13m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/61a811c9f4924edabb5d04785a7ca0f0 : FAILURE in 1h 16m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6516ed44f9664461afa502dd81a5cff1 : SUCCESS in 34m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d638b488003471e879111a866338b13 : SUCCESS in 29m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/29aa9542a43445e5ad1e04b876eb3de8 : SUCCESS in 28m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/79635560871c480fb0eb774c5f40989b : SUCCESS in 15m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35ec4d1a0fbe48b8a2063bfc436541f1 : SUCCESS in 12m 10s","accounts_in_message":[],"_revision_number":5},{"id":"94b236acf7d56153c3f17ba8c05441e233c2f74a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-02 12:07:38.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"731be0c200d3fba687174264f2ea176d66eb3f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 13:36:30.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb8ae11422c6475ab420a935789f1d3a : TIMED_OUT in 51m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7f47387583b4e40a828027e57366a4d : TIMED_OUT in 31m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/770598bc2e1947b1ada2e05603daa3b8 : TIMED_OUT in 41m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dd9bb928b2643c8872ba650dad7379d : TIMED_OUT in 41m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e90ea80e75544455b99f6e73c13ceb25 : TIMED_OUT in 41m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cfb6d3534ee443cb683ba2ec190868a : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/399162b4782d4f589749c4cbf4103093 : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/04436360f24342bbbcad15c9225590ed : TIMED_OUT in 31m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2bab7bc0c9c1444da0ce6c399ef068d9 : TIMED_OUT in 31m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/abec5666f7eb4d6694656c0d11edd99c : TIMED_OUT in 1h 01m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1a86cae4955344b68809e8da2c54c697 : FAILURE in 31m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c91ca3c2296249d3a6137c184d89e4e8 : FAILURE in 23m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/62601721d8f54466a6e14108247364db : FAILURE in 29m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e77e37fa1e944d2ba7c9d599cd2c3cd7 : SUCCESS in 16m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54c198b8f0e24a67812dd20f5a6886ed : SUCCESS in 13m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f143e93577234092b29712b2731ec42d : FAILURE in 1h 23m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/75366c4eb74343ef9beded06c865b8ae : SUCCESS in 35m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bda66e1eaad7441397588d4fe8516fc5 : SUCCESS in 29m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/53b9c93f27cf4084b88a6c378c0ec141 : SUCCESS in 27m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f35a0190e2b040989c5b7dd7b3726ddb : SUCCESS in 13m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02a076277b1e486ca1f78e9801e8b686 : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":5},{"id":"c5b3c24f388536e087694a22da201de36bc95fed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-16 10:26:38.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2764abf20a888f59aba06aa781c68404b2efb3fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 12:12:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ae568c6476941b49f6326a2e352764d : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa8696940f3747339cda1b9b5c40b322 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/858668bfecd247978d81ddbf041b702c : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4f0d81d13a44f9eb38d3f57bc4b81d9 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d718001f30dd4f3aa6edd5dedd02e3a1 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f3b40f0b0a40f1ab4870c01f4db410 : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a22c730acf54cd58f54ee37115e1eb2 : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28c3a552b92f42ab98ed12c0e4a07b50 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9aebf67cc4c146469063cd7ae2668e22 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ce5f3397e78740449e4c1f36fbb28ad2 : SUCCESS in 32m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1dc744d362c04b37ac15da531d11742d : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a07e8e12b61f42859ef9c8e0abe8058b : SUCCESS in 43m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9de20e75d8754ba9a1bb37b3d75aa007 : SUCCESS in 1h 09m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bb8809207741475782fe589b9494194f : SUCCESS in 20m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f22e57cd231b49ec9ad4b0607061cb1e : SUCCESS in 13m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e9e3721d69ea4af5bc8d97de70fb7af6 : SUCCESS in 1h 43m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/013850ab83f34483aacd604ecf6a7c63 : SUCCESS in 38m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/becbd7adc91e4ab185be3b3eedec05e9 : SUCCESS in 27m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c47a9ebb3fc740008412bea495c7ac99 : SUCCESS in 27m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4f159e52076a4bc9a1213e6f0b6fc3c7 : SUCCESS in 13m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99c15efc41ce486a9a95ed3dba86aca1 : SUCCESS in 10m 35s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":5},{"id":"002d573eef919e6cf01ef2f603aafe27a15eb671","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-23 08:50:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"62aff037088098bac227cf0ed9afa06bc65b4d48","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-27 07:33:45.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nDo users need to / should refactor their requirements.yml file to use roles and collections keys?","accounts_in_message":[],"_revision_number":5},{"id":"6ff8a6057c5dd0ae6a9162067e477229df83b64e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-27 09:08:59.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Do users need to / should refactor their requirements.yml file to use roles and collections keys?\n\nOnly if they want to use collections. The old format is still accepted.","accounts_in_message":[],"_revision_number":5},{"id":"ac3e623d4f24f2fe414676cb41d1313d49834032","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-27 09:10:53.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"19e0f5a34c03a0b344a9c516728b6ba9d2996ffa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-27 09:11:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e3feb86763ed39966cbe8aa25227b34b120e9677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:23:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e5aad49c0bf48748473aa4b526fef22 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/700c004b223f48c196c1f2a78dc6cbd1 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4dd7b928f194eacbe7af2e2378dd618 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cbcb811750340b5ac41887e5da45616 : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27e13af21b5e4c979df0a9880b44c942 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c89ecd0ce954864ba687d5983fdd46a : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24e81e53b07440ee9e95e25cc421acdd : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61270ee6e82f48069ad532dd271ca999 : SUCCESS in 6m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/74c237a527f9406f956c1a6aa60e5448 : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e59b8b498e1c43e9a9522aaaed9413ee : SUCCESS in 34m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/48262f633cc44cd3adb6d73f0be83184 : SUCCESS in 1h 28m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc2bd7578eef48ac9f6399214c2b6de2 : SUCCESS in 59m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fd63e113fe184de49f57f04337e4ff23 : SUCCESS in 1h 32m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4988ec5c4157465f987421750700e7e5 : SUCCESS in 21m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9305919466d4092a79316cf92dfe164 : SUCCESS in 15m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4c4f4634137f4deea7fa4b69968a2d4f : SUCCESS in 2h 11m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/407e58e7f0b1470ab54cf48a05bc26f7 : SUCCESS in 39m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3b176f5573a46a3855e34999f259dd5 : SUCCESS in 34m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8602ef4f9b8947fbb77697c71785c3fc : SUCCESS in 35m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/154354362a704f5f8f3c53348fca78f0 : SUCCESS in 15m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10ec898d4552416199017d7aa80d588e : SUCCESS in 13m 31s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":6},{"id":"3e24b0625db2af5586bfe256beab930fc102a9f4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-27 13:12:20.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":6},{"id":"f69b61b4760402d585acbc46f7df3e7803d9d626","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-29 10:18:05.000000000","message":"Patch Set 6:\n\nThere is a limitation of this approach - it is not possible to use plugins from collections in kayobe-config unless you\u0027re executing a custom playbook. For example, we might want to use a lookup or filter plugin from a collection in one of the extra vars files.\n\nI think that could be addressed as a follow up.","accounts_in_message":[],"_revision_number":6},{"id":"fa44ed307b32f31349232b47d2763171ea0337a1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-29 10:22:45.000000000","message":"Patch Set 6:\n\nAnother limitation is that it is not possible to use plugins from collections in Kolla Ansible.","accounts_in_message":[],"_revision_number":6},{"id":"6ec22f6ff7e40a0f20eed291bf52b28c87366ea9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-29 16:23:08.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"718aa6c1cb32397b66a48ae2f5d31c4941a8dbfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 16:23:47.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8fab054e93871bbcc5e2e249dcd445b3a36c081b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 18:19:44.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23288adf8bd34d1ba3ecc84687f701e2 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3c48ab9b40947599258a71f3f49d7ad : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cfc04ff23554a0aad53a633c495c7e7 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/774f7f8617294d2fab6eba6ea6c1a572 : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deec7be8c3a443ac8cd757fa2547d984 : SUCCESS in 4m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb45ef510cc244389027aebe5ccfefd4 : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f3da9529b694487894e58cff964d2536 : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1880866f31ac45e48207ef4b8c5b6c09 : SUCCESS in 31m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/42b0f61f081346cca4fe64cecf7d85c0 : SUCCESS in 1h 09m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aa9c69e0b5b44f98b06d1ce23710dff : SUCCESS in 48m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/acf0704aabaa4ed3a725ede07e77a45f : SUCCESS in 1h 12m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ef06c563967b41b9b6f7882b01ae2593 : SUCCESS in 22m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/059c24d65ff74f6a996af2e5a21f0888 : SUCCESS in 17m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/37a60f0c564946a69c1ccdb987ff6654 : SUCCESS in 1h 54m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4e3a8fcb85f24270b3c0d826b7940a9a : SUCCESS in 40m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0faf41d8f284d278155c2b4d8787a8f : SUCCESS in 30m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8a1f2604e6bf404bbb442dcda822cbe2 : SUCCESS in 28m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4ce824be326f4e4b970fbb26df0a77c0 : SUCCESS in 17m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e84898a37f63404fab88c49a6e76e5ee : SUCCESS in 11m 46s","accounts_in_message":[],"_revision_number":6},{"id":"931721a4ca43778fe9d1cb51db29f80895f58652","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 18:19:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"efc4d3a8c24e60b7e21729ce8d7bf8c72b797cec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 18:22:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58cc1fb7606f4f2cb799267f4a175304 : SUCCESS in 2m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e6c8c97392c40a1a4d4f58312595b90 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5535832c100fb73492d8b4f9b5e611e3b94a2fe1","revisions":{"af47cf0362501a48bbeaea64ed4f68c37408da4f":{"kind":"REWORK","_number":1,"created":"2021-06-22 15:14:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/97/797497/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/797497/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/797497/1"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-22 15:10:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-22 15:13:51.000000000","tz":0},"subject":"WIP: Support Ansible collections","message":"WIP: Support Ansible collections\n\nStory: 2008391\nTask: 41315\n\nChange-Id: I764ff019a18266b593add7ab80ee095d7d07a869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af47cf0362501a48bbeaea64ed4f68c37408da4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af47cf0362501a48bbeaea64ed4f68c37408da4f"}]},"branch":"refs/heads/master"},"8a2de27388af3342dae57f29a5eaa9f28d042a6a":{"kind":"REWORK","_number":2,"created":"2021-06-23 11:37:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/97/797497/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/797497/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/797497/2"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-22 15:10:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 11:37:17.000000000","tz":60},"subject":"Support Ansible collections","message":"Support Ansible collections\n\nThis change adds support for installing Ansible collections via\nrequirements.yml in Kayobe or Kayobe config.\n\nStory: 2008391\nTask: 41315\n\nChange-Id: I764ff019a18266b593add7ab80ee095d7d07a869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a2de27388af3342dae57f29a5eaa9f28d042a6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a2de27388af3342dae57f29a5eaa9f28d042a6a"}]},"branch":"refs/heads/master"},"5a743224be70c9028faa7f1e8099eaff6185aa5f":{"kind":"REWORK","_number":3,"created":"2021-07-26 15:31:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/97/797497/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/797497/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/797497/3"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-22 15:10:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-26 15:31:04.000000000","tz":60},"subject":"Support Ansible collections","message":"Support Ansible collections\n\nThis change adds support for installing Ansible collections via\nrequirements.yml in Kayobe or Kayobe config.\n\nStory: 2008391\nTask: 41315\n\nChange-Id: I764ff019a18266b593add7ab80ee095d7d07a869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a743224be70c9028faa7f1e8099eaff6185aa5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a743224be70c9028faa7f1e8099eaff6185aa5f"}]},"branch":"refs/heads/master"},"ae5e023003ec6b7f386b338fbd1483ba5f634d2e":{"kind":"REWORK","_number":4,"created":"2021-07-27 08:17:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/97/797497/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/797497/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/797497/4"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-22 15:10:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-27 08:17:43.000000000","tz":0},"subject":"Support Ansible collections","message":"Support Ansible collections\n\nThis change adds support for installing Ansible collections via\nrequirements.yml in Kayobe or Kayobe config.\n\nStory: 2008391\nTask: 41315\n\nChange-Id: I764ff019a18266b593add7ab80ee095d7d07a869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae5e023003ec6b7f386b338fbd1483ba5f634d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae5e023003ec6b7f386b338fbd1483ba5f634d2e"}]},"branch":"refs/heads/master"},"c91333b7d54f1621de496f916607e05d2b4325cf":{"kind":"REWORK","_number":5,"created":"2021-09-02 07:39:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/97/797497/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/797497/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/797497/5"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-22 15:10:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-02 07:39:43.000000000","tz":60},"subject":"Support Ansible collections","message":"Support Ansible collections\n\nThis change adds support for installing Ansible collections via\nrequirements.yml in Kayobe or Kayobe config.\n\nStory: 2008391\nTask: 41315\n\nChange-Id: I764ff019a18266b593add7ab80ee095d7d07a869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c91333b7d54f1621de496f916607e05d2b4325cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c91333b7d54f1621de496f916607e05d2b4325cf"}]},"branch":"refs/heads/master"},"5535832c100fb73492d8b4f9b5e611e3b94a2fe1":{"kind":"REWORK","_number":6,"created":"2021-09-27 09:10:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/97/797497/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/797497/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/797497/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/797497/6"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-22 15:10:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-27 09:10:50.000000000","tz":0},"subject":"Support Ansible collections","message":"Support Ansible collections\n\nThis change adds support for installing Ansible collections via\nrequirements.yml in Kayobe or Kayobe config.\n\nStory: 2008391\nTask: 41315\n\nChange-Id: I764ff019a18266b593add7ab80ee095d7d07a869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5535832c100fb73492d8b4f9b5e611e3b94a2fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5535832c100fb73492d8b4f9b5e611e3b94a2fe1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
