)]}'
{"id":"openstack%2Fkayobe~797626","triplet_id":"openstack%2Fkayobe~master~I69868dca6e1f3e4a1bf123c9e7f1f3412162b264","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-11-10 09:18:11.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I69868dca6e1f3e4a1bf123c9e7f1f3412162b264","subject":"Ubuntu: add upgrade jobs in CI","status":"MERGED","created":"2021-06-23 12:12:16.000000000","updated":"2021-11-10 09:18:11.000000000","submitted":"2021-10-04 09:45:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"797626","meta_rev_id":"23846f4e20d53df7ccc06515f131774bc1eebb72","_number":797626,"virtual_id_number":797626,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-04 09:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-04 07:58:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-02 06:59:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-10-04 07:58:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-11-10 09:18:11.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-10 09:18:11.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-23 12:24:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-23 12:25:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-09-29 16:48:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-09-30 07:57:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"4b31017b5e8d37d90b759f189b87d27d4552a85f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-23 12:12:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91f35fe9a32036931894e29e84779d999a79c806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 12:24:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e65a0663e566459eb16da426e85dd3e8 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f31348eda52547618b5c9adfee260bb3 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cdb3be29b824504af3587a28a3873ef : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36795edc178442db81ba6630b5591c7e : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e257cd478914f769222f018f4da551e : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9186425002e54ec4a413b4c01093aec1 : SUCCESS in 9m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a849668b586b49b78c873ea0824a6ae1 : FAILURE in 4m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e2fab01c30146c6adde3e0171582424 : FAILURE in 5m 09s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"8938563f76d7202cd52dcd8a6aecde3c123384bb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-23 12:25:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a30a242f636d303c7dcc874b55445c895037f0e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-23 12:55:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"af6feac588c83ad8bccc630759395fb047d2bd4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 14:57:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a1fce80d6bb413caec7f44baf7762f8 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db4700f5c6ab488e85cf4d92572ad43e : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58664b9289804d1aa1d6a73b21720ceb : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/335f6bd35255401aa81b44ee8e71e9cf : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d0dfd96f8ee41e28603ab4bbbd5b224 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e71ae866900046ff85f5eb0c8b0035ee : SUCCESS in 8m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3967552be011404f845997d1612099f1 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ad5b2f3720c4787bcb997fec169ff87 : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a8c34724a85048ee93d9481a4964f61d : SUCCESS in 18m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a8a51007f716483782a91bd4fe7508a5 : SUCCESS in 1h 06m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6eafa7802ab42daad20de4e113a8d0a : SUCCESS in 43m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b880d4f16b754a9ba0426c78f01a7b04 : SUCCESS in 1h 08m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/922f80e88b734ea79fe8195b394e1edc : SUCCESS in 20m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18c20655a4174dc7be42841628190301 : SUCCESS in 14m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9e1b8b28a21841e3834ac2e4383758e4 : SUCCESS in 2h 01m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca1b94c4ab9746cc995d83c14d0d449e : FAILURE in 41m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/98b90a814dd64ccfbf056e5c6d0765de : FAILURE in 15m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca57e0cc7b9244c5a01c5ea1e1c7d544 : SUCCESS in 31m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fb8722c367084ccf9d8ef91374e46bcc : SUCCESS in 24m 49s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40b45c32dd1b4fa4864b4c53dd0de450 : SUCCESS in 21m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6ff0115127d345a380c54d58b58bd059 : SUCCESS in 12m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/964b8effecf7496bbdde94915d155cee : SUCCESS in 9m 50s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"0cd262b882cb4e84449be37d57b348cf849861ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-23 16:01:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ccf573ec41b68cf94ac2a9b228ad1b52a62ff7a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 17:52:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78f85596445245bf813f09ab0e62b056 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151cf3371ff44219ab203d613895cd99 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef95462ef21c4b79a0df6ad5dc78da5b : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b387389310c24d9dbc215318c27a0c71 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3993b2f05f6d47b1a838230a78a62214 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cceafa0492464f8b9d9dfef67ac49fd7 : SUCCESS in 8m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b1149daf7c2f412f8b1eca42f899d19b : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d36cf2601165457fa65017c4337a7702 : SUCCESS in 11m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/55807c7f88c34e4690219302b081902e : SUCCESS in 21m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4a6be305f28548f3bb3e1e176bf75dc1 : SUCCESS in 1h 14m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02c6c3242fb146f9902411f91ccfb6a7 : SUCCESS in 51m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/eb0d93f713c14bfda04cc79b03b2258a : SUCCESS in 1h 21m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eed591fbeb8445868272ecacb3a97edc : SUCCESS in 19m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/337bb6858b294c02b0770f6c6503811b : SUCCESS in 15m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a0d3143165c746eda85284d2d411ceef : SUCCESS in 1h 45m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b2925f05e9547e7b5c017cbdf5e764d : FAILURE in 1h 24m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bb55da4fbf994568b866a1ab6808f307 : SUCCESS in 35m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dc19c8b6ace4444ba3690f43dc8dd09 : SUCCESS in 28m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8da79942b8f54481ad5dd2ab5e0bb4ce : SUCCESS in 26m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a82effb8c5df4a7585023784b90f6739 : SUCCESS in 23m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b4057195284a4bdf8385c192a83b97ee : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9558f8a70ba4a11be8718b06af35634 : SUCCESS in 10m 23s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"ebc965205afaddc3d3e796c79d9e839dabbb92d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 16:49:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e477d4ec8e799a7dea7d8dbd91c99d5e108375e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 18:14:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a08c75642e7d4619aedbbc7b5b8aa93e : FAILURE in 1h 22m 37s","accounts_in_message":[],"_revision_number":4},{"id":"870c9e74d8bd8997785f9b3755014cb8b3ca897b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 08:05:30.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"b1917a391bfbc073a64aa8045ad52f9d50b7816e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:29:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7252c6a3db3478ebbd889430100f843 : SUCCESS in 1h 22m 51s","accounts_in_message":[],"_revision_number":5},{"id":"77ff39424f819c435d25622bf1e47b8214e9177c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 10:00:44.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6f818ddfa7c611e257be5cb0218497b89af31907","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 10:01:35.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"a37333478dc8039e0c25c53f09619e1355598efb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 11:54:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baf5763f5e864d39a39edbb2328bc380 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cae1025d712049d59471688f362c15f8 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca92c22cae5e4324af9b423d9664f886 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deda9d52c4b944848e8403afa1e62390 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e0e04b5dd914a77b4ea593777142234 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95090b7135940d5aecb0a62315dab6e : SUCCESS in 10m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ed916dfb6a6b426bb3dd2b16aff10bca : SUCCESS in 7m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f170878893284500bc9ddad24dbea2a1 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e65e56587794f139dc6d45eef2c071a : SUCCESS in 19m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c781b814d6c441f48899a1bb03bba184 : SUCCESS in 1h 03m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05c1a4bfe4ea4aed91e74bb055b133fb : SUCCESS in 45m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bf3b3ec0b74345f1be74114390916e62 : SUCCESS in 1h 10m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ce2da9a3b20e4fa3bd27c4d931680a0b : SUCCESS in 19m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc9b77ac83544a8db22ebee44fac2f74 : SUCCESS in 15m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/18019663086e474d9c81f8ef5b44d8a2 : SUCCESS in 1h 51m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99bf6f359805432db3588d0aba6b66d5 : SUCCESS in 1h 21m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/76c652c8aaa34549be512985eba4c9d1 : SUCCESS in 37m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77f11bdff1fe49238fd03a30553d02ff : SUCCESS in 28m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/344573abde314be19c06218a15d9bb78 : SUCCESS in 28m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f5a31d648b84bdab579b6d32c987c5a : SUCCESS in 22m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e99e137e115b4914b62017861a17de18 : SUCCESS in 14m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/489d9e61f9e14493918a962e95c96d49 : SUCCESS in 12m 19s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":7},{"id":"f6689719e5210be8a4cd268b1a5d730ddb2c217c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 13:06:33.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"df4f08046c2b55929b5570f5eca5ab029413adad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 15:10:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da3eaeecd9684c49b007aad42f99f1cd : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32f67ce3fffb4fa3bc2f70b4f1e0788a : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c16360819fb64dec8124fe5e3f55616a : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8a126eb56264d1082666c845f2edcc1 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1da937dde2754aaebd07a1497ce3f9d5 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9fdf49072174704bb1d60230bfaf3f1 : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fba6788a8513411da072072e0a495e04 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fba9dc8051114b4c9c4357aafcaff4d1 : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d885766b76cf4518a7fd8ddce5372b32 : FAILURE in 7m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a6a343bbd692493c9e78efef28552c54 : SUCCESS in 1h 08m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42e582e3a1c04003855c48fb64bebbda : SUCCESS in 48m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c1b10dca64cf48c4ac86194a3be7c013 : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/896949161ec646208889738492a38388 : SUCCESS in 19m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/803320cbfd6949f488448c4708e9f850 : SUCCESS in 12m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/61e0938b03224764a996ed4bf940d042 : SUCCESS in 1h 55m 17s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e311d6e960234448a8b16692483e8aaa : SUCCESS in 1h 10m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b114579e390e47418e9a50cbe14399eb : SUCCESS in 35m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92ebc12623d84b119ea8fac652aef3f8 : SUCCESS in 28m 03s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/54bfb92bbb1047be9be8ecaac3df7428 : SUCCESS in 26m 21s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07da24a3572942e8a54ba88e3d618f8c : SUCCESS in 21m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1922054c64634b20ad3d111c9262c81c : SUCCESS in 14m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6434067f82994535ba296af9b23fef86 : SUCCESS in 10m 56s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":8},{"id":"e16ad26be5419e90f4ba5e18f85c8e0134a8fde3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-20 08:18:06.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"ff61728ddd9d0e600d2b49b2a518ad8da8b25808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 10:18:53.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57c926429b3a4abf94e8c76610c33226 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/296824125d7c4c1691c499333c118d8c : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbff97b96c0b4f00bc31fef90f955627 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c288a4667be4fac8edade0657d52674 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a62d79af57104bdf8f0c1ca542783173 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a4ac850ba34480dbf409779d84928ee : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f96023974e84cd3a8ec7d8cd8a01958 : SUCCESS in 6m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b092847070d14528a72e03bb73446862 : SUCCESS in 10m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e53bb0cf7464ef59283e73b0b574657 : SUCCESS in 31m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4183921021cc4bc594502bec391e9de5 : SUCCESS in 1h 10m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68074fa3b4714f5096aac9f1b37b3347 : SUCCESS in 47m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0752191c36614f4aaa93940d00a62596 : SUCCESS in 1h 06m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0afdd34517b740b4a49e38236c03614d : SUCCESS in 18m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcf5577c79b843388c9ed315b9c11c68 : SUCCESS in 15m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4d59041b7b62424087f7ce1d6287a580 : SUCCESS in 1h 48m 45s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fbae3b630154c14afa2cfd7b0e753a7 : DISK_FULL in 1h 25m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/49760b233b0043108d84abc1ec2718ba : SUCCESS in 34m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/172cb59aa698434aab6bcc95d28c880f : SUCCESS in 30m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/98c3ef48aedd4637a5a775963989fbe1 : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffc3e664951943b8a7f22733bb2dcc82 : SUCCESS in 20m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2bfaa9678e7041c0ad06739f153abeef : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff9e203912eb44878015da66fdef48d0 : SUCCESS in 10m 28s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":9},{"id":"73d3c40272f140ca429be118914fcc88eec891df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-28 14:14:21.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"ec92759eb864f5debeb4b24bb760e65ca7db8706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 16:02:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/875f2e69e71642f1812684f43a4ef587 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14d02c37a054999b27af5677dc4e9af : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d689546ca61429f816d0a6ae879236b : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e11687b04083454fbe806bba4958e809 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15d13eddf931401592855bb50ea09fa2 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8007a420fee44420b154258d215def49 : SUCCESS in 7m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e61ae5c240b14584b82a3e9b988b21f0 : SUCCESS in 6m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9cc141184dc04951b33ccf1914759921 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e784f57c5dd944d48273fbd5191e756b : SUCCESS in 29m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c7469274d4aa44bfb1141979c9041d83 : SUCCESS in 1h 07m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0235b6ad39a4f08878bff523153d3c9 : SUCCESS in 47m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/48ff6042f37a4c0ea69b3fa5ec3fbb78 : SUCCESS in 1h 25m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/596c6b7aa9614774b34a9bc0947d48f6 : SUCCESS in 19m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e90185d72964451ba1d5a8297c7f6d4a : SUCCESS in 12m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/070838e5f63f418f930fded8e7f13e1b : SUCCESS in 1h 45m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a9bfb95a6c04e4b817c8c821d35fdf2 : DISK_FULL in 1h 15m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/51e60e4ac6ec486b9585cf95f77d795a : SUCCESS in 35m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a02af867cbbc4361acd30ad83ddd174d : SUCCESS in 29m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/19bc039b82004b53946643e5d578b42a : SUCCESS in 29m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23a13c2322504b09b821717a68f2e089 : SUCCESS in 19m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/57d166bd64da485981b48bbae66ba550 : SUCCESS in 15m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98d77d7db3134e6aa1b80e6e9739abbd : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":10},{"id":"582ae39dedfe2ed3be7e96a0d2e084cd88f9c8b3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 07:26:58.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"2666ca05657ac899606e2a055341f930968a94af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 08:03:54.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"abe874432f137b8c12dbebb3faeabe9253f2f72f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 09:59:27.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e1250748cdb44468f207fb8ab4d1b0a : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752ace6debb7478fa0c1853f84f127ee : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/698c84d6a2324a91ade2e022eeaba235 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e6b122531c64cb0b4fadc636f160544 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3a5dadd0cbf4a4da9a25f3654c30105 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c905613f58040a9a870f05ad29e81ea : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/deb67d043e25428eba071af5b3190994 : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9b48d78b96b24fdcab42a5ad9f83980a : SUCCESS in 12m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/76da97d5277d4846b3000acd6d5a0ae1 : SUCCESS in 30m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4173e12eb17343ed8ffca8802d7c9ed1 : SUCCESS in 1h 03m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07951b95b41f49f4beb1df02899a34c0 : SUCCESS in 43m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/54e30376e21f434db2d7814286deb7b4 : SUCCESS in 1h 39m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4755f01f685c45e1857a70c00031f59e : SUCCESS in 21m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60dece54fc344c28913e1a35975f1ab7 : SUCCESS in 14m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7803d8b64d42447e88e3d5aad217d0e4 : FAILURE in 1h 07m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63c191c99c884140904bebaa306e0e09 : SUCCESS in 1h 20m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e1903376bf73403589f72ecc98fdf2c1 : SUCCESS in 39m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e70e6a1acf4e44eb8dd51102b7d3808f : SUCCESS in 32m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2323530a5cf42499add91bb93b761e8 : SUCCESS in 26m 20s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b541234fe0cf478d806732f413ced914 : SUCCESS in 27m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2f94cd836c7e469c821b4bf14096b5c6 : SUCCESS in 14m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bae6db9fc4a14fd485e717daac3ea891 : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":11},{"id":"5e70d33697d68336f0b779d8c1f974c8174365f2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 14:42:11.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"c48df32581e598c301005f0ca8dd34883f30b5e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 16:45:50.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb5193c4c8154451a16e397eb032f3f9 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e741139b8e0d4bbda5ba2022ce445e13 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acf556b713014eef9105844424b0032c : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acd5246aea5c4ae9ad64d921ea1aa1ae : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0fb658817a0422385b5d42620fa8ed5 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38377e6f1e3c4edebb0780d78836db5f : SUCCESS in 10m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9d34868ee6284aaabba0f3ca513b6587 : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d36bce6703d47e098ed3e751a7ad0ea : SUCCESS in 11m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4e9a10bd66e4e3fbd082790c9c2f3c0 : SUCCESS in 33m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/40f303fb7c444d4b826b98ccc2095751 : SUCCESS in 1h 06m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/440be336310c4c9a9069899be459efa4 : SUCCESS in 45m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/056aa7b642614e1081c78bbd17dc4f55 : SUCCESS in 1h 35m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0cb13de1ac9c4025ad8dbc7298e1c4f3 : SUCCESS in 21m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18ab62ebd9b046eda0056b07cc2a6348 : SUCCESS in 17m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/77319b91303c49cd935b13bb4f7cee29 : SUCCESS in 2h 00m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be4c50a1812647ee8acffc07af1048a6 : SUCCESS in 1h 15m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dae8cf7111dd462eaa94ab68995151ad : SUCCESS in 38m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe8183ee6dd041f7967993231931835e : SUCCESS in 30m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8c79e77f67f343cf909a701ef3254cf1 : SUCCESS in 28m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24555b41550e4f3a99d74d631f0fb2c3 : SUCCESS in 23m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/207594871d5c401883b0678dcf424735 : SUCCESS in 13m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f2ca78578974aa481196fa77bf374ef : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":11},{"id":"1dca837495a4a52dbb8cc9fff6176a7654466feb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 16:48:00.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"b6de7c031977ff96c8ad2e3bb56f44f39e61973f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 07:57:03.000000000","message":"Patch Set 11: Review-Priority+1","accounts_in_message":[],"_revision_number":11},{"id":"6042ddbea6056d6a276b5a7cec18ce6bad525957","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 07:35:06.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"45f051950cc064152950726ccd4246748196d264","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 10:57:43.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"502029ab836197409987c4de1ffebead489da44d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 10:58:00.000000000","message":"Patch Set 12: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":12},{"id":"1ae4c5be99f73ae61d357adcebc02c9efda1c977","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-01 11:33:01.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"6e12d8fca053f66646dd477f98900892ea200161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 12:51:54.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/781d09eff12242b4a37e36d4c623da3e : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bf476ea4fb347748b0ef92f2794aee9 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cc5e3e395444fad83e074eafede47d9 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/425a9276c3e64c13834ee240245e1e89 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5078a1034dc487dadde9a92331304be : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95dce8bf0789418a9e159b94c8d5f11e : SUCCESS in 9m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba06c538c452416a9177c842075fbf2c : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28de630ba3be43bb9fa30b576901ea82 : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c2f49e9dc6c84703b27ee46c4cdfb35b : SUCCESS in 34m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/666ae40a02a249e09ce4efac48fd6d29 : SUCCESS in 1h 06m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbab13d59b7141d2815edf6cc48588c8 : SUCCESS in 45m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9f39718aff514f18a2845ad76f3d74fd : SUCCESS in 1h 04m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5753b29095f141c58b68d52f3feeea15 : SUCCESS in 20m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/336651db35ba4422a1dbd54551126e0a : SUCCESS in 15m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d2ca075461364814b6f12b9ae2c999e3 : SUCCESS in 1h 49m 47s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5e7e09cb3c84ffe9eafe6afbe35c65c : FAILURE in 17m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/42ae8f57e2e04e8383e65c33725a5df4 : SUCCESS in 36m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2300a0ee5a84c05959e1a83210ab2ea : SUCCESS in 32m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/558c986f677545e4bfa58668365dbade : SUCCESS in 29m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97f8e6a200824019ad05107a98070d62 : SUCCESS in 23m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8285a6cd84b949a0bf7be57004e6bbd1 : SUCCESS in 16m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4118aeef09f46108a4307787cd07938 : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":12},{"id":"f38320efabbda94a60fd69dab83548ff4b8a13aa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 12:56:46.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"dd28573b3a0ec3be2b14a05188f6317c8b14ae60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 13:49:52.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"395c72ef5486cbe8beadf42d5e38c4303d058341","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-01 13:53:53.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"f4cdc1e718fad3a044de6a534fd977ca8e265b31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 14:06:56.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1d3702a635a86575d16fc43f5e7ecb0f133e3d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 16:02:45.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92d69e83a9054d0ab9d0e2ef13cc6fc9 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1365a72cc0f4390a2b0bfeb3b01e231 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc09a3bb65ab4d5c861d1147bf4a244b : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b4718702de34a69b43e316744c52d9b : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b10134d20fc54cd48bfbb0b4ddd78b03 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/525ac271d404458bbe016f9821e74349 : SUCCESS in 9m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6dd73b91e6624fc09f39412fb3eba812 : SUCCESS in 7m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0f76e9b46c1c46f2820edf5e1ff15d83 : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/33ddf79b612b424198f5232d70e65b8d : SUCCESS in 29m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/37665859990d47ebac39c8bf483a7726 : SUCCESS in 1h 14m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae5c0842d3c847579e9ea56e77f260d7 : SUCCESS in 49m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e0c5a8ac66654367bda3f37165d7684c : SUCCESS in 1h 15m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d0232c363130476daba66a57261b0611 : SUCCESS in 20m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b88e8208c1ac4247ba017dd6a15d1617 : SUCCESS in 18m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b19d340e3eb745538e1e179a93d0ea2e : SUCCESS in 1h 53m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c697f4242eb4edbaa672a9a4320a738 : SUCCESS in 1h 18m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/118d164871e94ca7b90eaf4d17609b72 : SUCCESS in 37m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0e2bbad21474b17b883ad6f20fbde11 : SUCCESS in 30m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/929868deb025442481efc3dbdbbc26bc : SUCCESS in 37m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ee76620991744a39abca9805738d904 : SUCCESS in 23m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7620600c14304ae2bfca134c8c865d00 : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a90da5079c6497dab4fb7feb68550db : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":14},{"id":"7bd83142984b0dab4e45a70f5fe70f50af2c9580","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-02 06:59:15.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"d596641bb097555e7cd91e18a6369fe526fc0ec7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 07:58:37.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"a8a1b746d5e86f6e2cac7bc8856526c3e9460898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 07:59:19.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"66ed04fffef2b2630dac8f1c9c5dd3619aa086d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 09:45:35.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9d1a9bcf7604e998ca90066a1edf962 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe7269ba2e954ced8df059c6e183b5f6 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f76325f127d443a861eb188b0c2e9e1 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c1062f843fb4a65bd96bc9ff965cda5 : SUCCESS in 9m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f65ae5e831454374a7b6f6374261234f : SUCCESS in 7m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/47cea70eba5b4a299b8bdfea364e3df7 : SUCCESS in 12m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f36f0612986e448c824c8a4146a52967 : SUCCESS in 29m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/18db1862fea24fb98cf821486f967a7b : SUCCESS in 1h 03m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ee48e796975455daa0699e12e36706b : SUCCESS in 48m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/385adb1271f44638bf06f14074945d52 : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a36cc3a70126410fa5e56d8588f37fa2 : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b61ee0982554d91aaacdf40e4b5cf5c : SUCCESS in 23m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d7b86068efb746fe90bba1462c879fc8 : SUCCESS in 1h 46m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/14711f7087d3401f9398bc978d03236e : SUCCESS in 37m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c33a9f921be348f99f2d2a3c223fea35 : SUCCESS in 46m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4d2b263d38a349a0b228e92e0bacbd9b : SUCCESS in 34m 09s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0c45d0b8372348eb91defb4f5371eb69 : SUCCESS in 13m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73511c6c491b4e51aa581da58cd32fc5 : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":14},{"id":"5bedff5e4c87501d197aadd42192000342f9dcc2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 09:45:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"31915bc181b96ea2f878047f4a539fc61ea581f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 09:51:32.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe1eb73dbb74ac09625da6d981cf09b : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":14},{"id":"2d966d159653fc74eba349a421b779b3972170de","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-09 20:20:32.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"17b4ec521feeea0adde8773d82a7d4d55ffdd059","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-10 09:18:11.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"3b901641025bc9dd2f9bcf1c53798958539bd3eb","revisions":{"a92eba77c4544598ece8ae471b6f80589f542207":{"kind":"REWORK","_number":1,"created":"2021-06-23 12:12:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/797626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/1"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a92eba77c4544598ece8ae471b6f80589f542207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a92eba77c4544598ece8ae471b6f80589f542207"}]},"branch":"refs/heads/master"},"7fe178ba61564627c2ef0582cf4de04a94dca4b7":{"kind":"REWORK","_number":2,"created":"2021-06-23 12:55:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/797626/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/2"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:55:33.000000000","tz":60},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fe178ba61564627c2ef0582cf4de04a94dca4b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fe178ba61564627c2ef0582cf4de04a94dca4b7"}]},"branch":"refs/heads/master"},"37431585048f5a401926c230f0f14e59c9e7ff40":{"kind":"REWORK","_number":3,"created":"2021-06-23 16:01:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/797626/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/3"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 15:57:10.000000000","tz":60},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37431585048f5a401926c230f0f14e59c9e7ff40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37431585048f5a401926c230f0f14e59c9e7ff40"}]},"branch":"refs/heads/master"},"fe601aaf9d0dff059f70374dc3fb60b681566224":{"kind":"REWORK","_number":4,"created":"2021-07-26 16:49:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/797626/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/4"}}},"commit":{"parents":[{"commit":"cf60ba0d49b54abdbace9f878d85fc77cec315f7","subject":"Use stream8 images with molecule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf60ba0d49b54abdbace9f878d85fc77cec315f7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-26 16:49:18.000000000","tz":60},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe601aaf9d0dff059f70374dc3fb60b681566224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe601aaf9d0dff059f70374dc3fb60b681566224"}]},"branch":"refs/heads/master"},"9a07e04318dc5846a76166b05c8a653bb5d18d24":{"kind":"REWORK","_number":5,"created":"2021-07-27 08:05:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/797626/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/5"}}},"commit":{"parents":[{"commit":"cf60ba0d49b54abdbace9f878d85fc77cec315f7","subject":"Use stream8 images with molecule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf60ba0d49b54abdbace9f878d85fc77cec315f7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-27 08:05:27.000000000","tz":0},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a07e04318dc5846a76166b05c8a653bb5d18d24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a07e04318dc5846a76166b05c8a653bb5d18d24"}]},"branch":"refs/heads/master"},"ffecdf7217f6d36f2e096c9b20b619c4a3a3b0ca":{"kind":"REWORK","_number":6,"created":"2021-07-27 10:00:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/797626/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/6"}}},"commit":{"parents":[{"commit":"cf60ba0d49b54abdbace9f878d85fc77cec315f7","subject":"Use stream8 images with molecule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf60ba0d49b54abdbace9f878d85fc77cec315f7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-27 10:00:38.000000000","tz":0},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffecdf7217f6d36f2e096c9b20b619c4a3a3b0ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffecdf7217f6d36f2e096c9b20b619c4a3a3b0ca"}]},"branch":"refs/heads/master"},"6508358f3ea957bc164c45e1e6c8e1fd068438a6":{"kind":"REWORK","_number":7,"created":"2021-07-27 10:01:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/797626/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/7"}}},"commit":{"parents":[{"commit":"cf60ba0d49b54abdbace9f878d85fc77cec315f7","subject":"Use stream8 images with molecule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf60ba0d49b54abdbace9f878d85fc77cec315f7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-27 10:01:16.000000000","tz":0},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6508358f3ea957bc164c45e1e6c8e1fd068438a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6508358f3ea957bc164c45e1e6c8e1fd068438a6"}]},"branch":"refs/heads/master"},"a5b1c1fab39eb3f36362322061eff84613493b48":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-09-17 13:06:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/797626/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/8"}}},"commit":{"parents":[{"commit":"f1212d0f07cd127c4a618126178a9edfb44160e6","subject":"Merge \"Support setting docker registry credentials to the empty string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1212d0f07cd127c4a618126178a9edfb44160e6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-17 13:06:33.000000000","tz":0},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5b1c1fab39eb3f36362322061eff84613493b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5b1c1fab39eb3f36362322061eff84613493b48"}]},"branch":"refs/heads/master","description":"Rebase"},"193650cb4d78be69b8c0441f5c619e01b71be3e4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-09-20 08:18:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/797626/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/9"}}},"commit":{"parents":[{"commit":"81d6fa9a07c4f20d6bc244799d149101d0f62dd3","subject":"CI: Avoid molecule 3.5.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81d6fa9a07c4f20d6bc244799d149101d0f62dd3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-20 08:18:06.000000000","tz":0},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/193650cb4d78be69b8c0441f5c619e01b71be3e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/193650cb4d78be69b8c0441f5c619e01b71be3e4"}]},"branch":"refs/heads/master","description":"Rebase"},"4d49b246f4b73dad67d3fe5c58fbd8e168ae82da":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-09-28 14:14:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/797626/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/10"}}},"commit":{"parents":[{"commit":"f9e31f0b540474e7257ab10fa9642388f46305ee","subject":"Merge \"CI: Fix bashate ignores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9e31f0b540474e7257ab10fa9642388f46305ee"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-28 14:14:21.000000000","tz":0},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d49b246f4b73dad67d3fe5c58fbd8e168ae82da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d49b246f4b73dad67d3fe5c58fbd8e168ae82da"}]},"branch":"refs/heads/master","description":"Rebase"},"1ec37a82d75faf26a6e73c7009919ec9839c4f12":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-09-29 08:03:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/797626/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/11"}}},"commit":{"parents":[{"commit":"f09faa43d1a1d38b30967cce948af6b60c34be70","subject":"CI: Disable libvirt debug logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09faa43d1a1d38b30967cce948af6b60c34be70"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 08:03:54.000000000","tz":0},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ec37a82d75faf26a6e73c7009919ec9839c4f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ec37a82d75faf26a6e73c7009919ec9839c4f12"}]},"branch":"refs/heads/master","description":"Rebase"},"26568609685d7e77fac27d9f286eee93536228d8":{"kind":"REWORK","_number":12,"created":"2021-10-01 10:57:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/797626/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/12"}}},"commit":{"parents":[{"commit":"b8abf706d4d8a40f09e7e8c343710bbf4d5404d9","subject":"Build and deploy source type images by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8abf706d4d8a40f09e7e8c343710bbf4d5404d9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-01 10:57:28.000000000","tz":120},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26568609685d7e77fac27d9f286eee93536228d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26568609685d7e77fac27d9f286eee93536228d8"}]},"branch":"refs/heads/master"},"fa79464d918ba1f77258af313c88e3c1b715df4d":{"kind":"REWORK","_number":13,"created":"2021-10-01 13:49:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/797626/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/13"}}},"commit":{"parents":[{"commit":"b8abf706d4d8a40f09e7e8c343710bbf4d5404d9","subject":"Build and deploy source type images by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8abf706d4d8a40f09e7e8c343710bbf4d5404d9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-01 13:49:14.000000000","tz":120},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa79464d918ba1f77258af313c88e3c1b715df4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa79464d918ba1f77258af313c88e3c1b715df4d"}]},"branch":"refs/heads/master"},"3b901641025bc9dd2f9bcf1c53798958539bd3eb":{"kind":"REWORK","_number":14,"created":"2021-10-01 14:06:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/797626/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/797626/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/797626/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/797626/14"}}},"commit":{"parents":[{"commit":"b8abf706d4d8a40f09e7e8c343710bbf4d5404d9","subject":"Build and deploy source type images by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8abf706d4d8a40f09e7e8c343710bbf4d5404d9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-23 12:11:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-01 14:05:33.000000000","tz":120},"subject":"Ubuntu: add upgrade jobs in CI","message":"Ubuntu: add upgrade jobs in CI\n\nNow that the previous release is Wallaby, we can add overcloud and seed\nupgrades on Ubuntu.\n\nChange-Id: I69868dca6e1f3e4a1bf123c9e7f1f3412162b264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b901641025bc9dd2f9bcf1c53798958539bd3eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b901641025bc9dd2f9bcf1c53798958539bd3eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
