)]}'
{"id":"openstack%2Fkayobe~799245","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I76d658b93b9195c760c3a91fe5aee96bad8d6b42","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"I76d658b93b9195c760c3a91fe5aee96bad8d6b42","subject":"Ubuntu: CI: rely on os_distribution, remove interfaces pause","status":"MERGED","created":"2021-07-02 16:02:24.000000000","updated":"2021-07-27 09:28:09.000000000","submitted":"2021-07-27 09:25:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799245","cherry_pick_of_change":788478,"cherry_pick_of_patch_set":3,"meta_rev_id":"fed8a1598293ebe2c9c3f635dd5ef748f709204e","_number":799245,"virtual_id_number":799245,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-27 09:25:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-07-27 07:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-07-05 09:43:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-07-27 07:33:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-02 18:01:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-27 07:33:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"7b2242bba236f7bfe73a04435a333a0e6d79488b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-02 16:02:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d90026f5928275e2b44ba0235286fa482ba8332c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 18:01:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3518df1d66c433ea872fc344ad26fa9 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c362354436d4a198918dd35c27b074f : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bceb8fb3b5b49a18a662d24cf0d037d : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a168bebdc1914cbb96ead2e103d78590 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d6204c267bc483fa328b5b9bcbf5e8f : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78d398189ff4edab567adb089c19f7d : SUCCESS in 9m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/adf66ca4360e44c28bf566b62e8d549e : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d15d0d2f8326474ea71de86f7c67de1a : SUCCESS in 10m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a2ec7c4368d8401b9314c7c9d7c28c3e : SUCCESS in 22m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/21311be4ab0f40c39a1b0674adbcc5ee : SUCCESS in 1h 07m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86815f70eca34475bb8d57721f2d93bf : SUCCESS in 49m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a521219ef2794fddb84c1a7514aebb9d : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1924cefadba84d8c995b6260dee75049 : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c60aee930ea43408e73cc430924c544 : SUCCESS in 15m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1307970482fe42e5bc65bc3e453ad398 : SUCCESS in 1h 55m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/94b8080167dd4060bf89c2ab1cfd7d86 : SUCCESS in 34m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e82e2da044f747e4b585daae1e6ccfdd : SUCCESS in 28m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e6955a8376de4348b2185766afb1866a : SUCCESS in 26m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2c4ec6a19c74430fb22d66b44c3af472 : SUCCESS in 16m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c6336805cac488e829dc90533d683c3 : SUCCESS in 13m 06s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"9f51cb710c7bfe2170b2a4d3e805a5c2183cdcca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-05 09:43:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0147ee14d764669e364dd688f32db831c98a08e4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-27 07:33:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64afada2df5e0b00f743546423331ed5a84743bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 07:34:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"034129e6926bbd4d7a25238523b108556cc8d81b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:25:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d344112347645d4bf40e02910bb5ac2 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea55dc1d70b747e18c3882c48bd2cb75 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dec9c2508b64701b58f8a8b0826daa7 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce68ccb0cd024202aa20da005f285488 : SUCCESS in 9m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c763492886b4945b507caad0419fe8d : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7bc03461a96448c99ad5cddd847f964f : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d56238230d28487f90a28d978fcbb312 : SUCCESS in 20m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ac241b6de9454f099d5e9b72e375150c : SUCCESS in 1h 06m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b6869c9251045489cd6728549a225f1 : SUCCESS in 51m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f8d01fb6a808468e89d8d9f80757d217 : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/121a9609b7d941c29ed794446f42f0e9 : SUCCESS in 21m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae9518b4f4914a61afe283f82fd09096 : SUCCESS in 14m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/49f7f5a793f947bdb4859c80449937b5 : SUCCESS in 1h 50m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/16d41e47e14c4e42b811aa10e587d984 : SUCCESS in 35m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28b334dfee2844d0877e9528eedbaf19 : SUCCESS in 30m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/56e7850fa7ea4d6191ea5c2b3644ea61 : SUCCESS in 27m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a2a2babc02454166adb1317b2a08307b : SUCCESS in 13m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0c5c3c738fd49c79a33226c09c9a318 : SUCCESS in 10m 40s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"dce5d0edeb76a85a8c1001065c2263e2515ddc71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:25:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fed8a1598293ebe2c9c3f635dd5ef748f709204e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:28:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf2afbed26f46e08405a1faa7c633b0 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ac76bfa1c22b981122df4c8164eab46f3e81ac2","revisions":{"9ac76bfa1c22b981122df4c8164eab46f3e81ac2":{"kind":"REWORK","_number":1,"created":"2021-07-02 16:02:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/45/799245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/799245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/799245/1"}}},"commit":{"parents":[{"commit":"feeb4353b8ae437833e15eff540d0458cee35c4f","subject":"Remove outdated upgrade step","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/feeb4353b8ae437833e15eff540d0458cee35c4f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-28 11:51:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-02 16:02:24.000000000","tz":0},"subject":"Ubuntu: CI: rely on os_distribution, remove interfaces pause","message":"Ubuntu: CI: rely on os_distribution, remove interfaces pause\n\nWe can now rely on the os_distribution variable to set sensible defaults\nfor Ubuntu. It is set automatically in kayobe-config-dev.\n\nThe interfaces pause workaround is no longer required since switching to\nsystemd-networkd.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/791944\nChange-Id: I76d658b93b9195c760c3a91fe5aee96bad8d6b42\n(cherry picked from commit 111375fb1ab335a02109dde34558d3f46aeaa124)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ac76bfa1c22b981122df4c8164eab46f3e81ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ac76bfa1c22b981122df4c8164eab46f3e81ac2"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
