)]}'
{"id":"openstack%2Fkayobe~799845","triplet_id":"openstack%2Fkayobe~master~Ie52ec41c3cd9cd02501f2c8f2006aa44ac87096e","project":"openstack/kayobe","branch":"master","topic":"ip-allocation-segments","attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-10-26 12:59:06.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-10-26 12:59:06.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-09-12 23:16:20.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-26 12:58:51.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Ie52ec41c3cd9cd02501f2c8f2006aa44ac87096e","subject":"Add support multi-segments for ip-allocation","status":"NEW","created":"2021-07-07 14:52:32.000000000","updated":"2023-12-01 01:09:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"720d58ee6ec969aa3806afd9d66705177f982fed","_number":799845,"virtual_id_number":799845,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-01 01:09:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2023-11-30 22:45:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2021-07-07 15:54:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-08 09:22:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-08-16 06:19:26.000000000","updated_by":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"reviewer":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"state":"REVIEWER"},{"updated":"2021-08-16 06:24:54.000000000","updated_by":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2021-08-16 11:32:29.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2021-08-16 11:56:08.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2021-09-16 10:31:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-10-01 05:36:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-01-13 11:07:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-06 09:58:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REMOVED"},{"updated":"2022-04-06 09:58:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"state":"REMOVED"},{"updated":"2022-04-06 09:58:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REMOVED"},{"updated":"2022-04-06 09:58:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REMOVED"},{"updated":"2022-04-06 14:44:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"a57190a957c83d87b893aeea40534b8b8edaeb68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-07-07 14:52:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e888fa32a39f2207d9443b94bc4534839d888e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 15:54:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d8d85f89987425790d98b54d268096a : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a13a6d74d8bd499ca6055e375cb6d543 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c573d8b586aa41eeab85286ec18f1f58 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17e7de7e143547e79109354197368407 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e21a9bc6cc85429594fa1d0a8107581c : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b87f9fba9f742d79dce585defdc6e52 : SUCCESS in 9m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5acaa25a6fde418c813ba27a3619e09c : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ab4d4706a2e24f418a2b546145b0fe4f : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a92fe537abcd46fd9ec4701213bb16a9 : SUCCESS in 21m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4f347dfdd85044e8bb9f0b66580759de : FAILURE in 49m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e41dcf8509a5478d8c4d9815bdb0c51b : FAILURE in 41m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/28bf99262c164c6f88832606625a8266 : FAILURE in 50m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8989236b80ba438d8d07f56340ec7256 : SUCCESS in 18m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e080623f4004ce8b25902ed32fafa96 : SUCCESS in 13m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/140160d4691a4e6885cd5ef1e6c6bab2 : FAILURE in 48m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4f15e2f7e79f41ff923da98bbbb22d5b : SUCCESS in 31m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e695155efe84e01b21bb722979498b9 : SUCCESS in 27m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b6dfe03aeba5463490a1b2620bd69e68 : SUCCESS in 24m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7b0ba19cd2884c91a50351246526880f : SUCCESS in 14m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1881acd3cce04b4a96796a95471450f3 : SUCCESS in 9m 34s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"42d01b44bef44fec92ff6619d5dacceb63a70c17","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-08 09:22:54.000000000","message":"Patch Set 1:\n\nThanks. I know the approach is different, but does this doc describe what you are trying to achieve?\n\nhttps://docs.openstack.org/kayobe/latest/configuration/reference/routed-control-plane-networks.html","accounts_in_message":[],"_revision_number":1},{"id":"f8028ac768a852563be0bc55494ad3cf4b1ee956","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-07-08 23:34:32.000000000","message":"Patch Set 1:\n\n\u003e Thanks. I know the approach is different, but does this doc describe what you are trying to achieve?\n\u003e https://docs.openstack.org/kayobe/latest/configuration/reference/routed-control-plane-networks.html\n\nyep. already achieved) now the error \"Found invalid existing allocations in network\" happens only when we have the real mistake. we use alternate approach with routed networks when named networks are the same for each rack of the servers.","accounts_in_message":[],"_revision_number":1},{"id":"982f3d896e3c98cb88ae43a3235b904f8f6852c0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-07-09 05:48:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dfad12b5c8be948d2402d1b810af55e29ea22f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 06:46:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a729cf13cef4649b8dbd54b5f4f8f44 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/632cebf234264c3fb9f85a055c3ebae8 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/522c23076434420c8c437ccee082c7fc : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9478ee973679473f82cb31ad5dac3548 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77c8b775d491405dac8e402faaba2cf3 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e310c9448443408dcd5c80acef0958 : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dfc5f9f1b04d40d69104b462c817dbb0 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/59102046e7454317a040b416c080a504 : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/15105741c1a4461e80c6c9e9964d2d76 : SUCCESS in 20m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4752128ea6e14c64ad5264b156a5ce0e : FAILURE in 53m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d61e25f3e9384bc3bd0a96c1cbdc36ba : FAILURE in 43m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6fee442ae93943c6947999a82dc36839 : FAILURE in 55m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/49e0a251cf81440f843ad8f25d67b9f0 : SUCCESS in 21m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72906096e5df4b7f868a0cf995cdc68c : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f7604802f95c47378477c0feac9bfb9f : FAILURE in 55m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0e18ea54677d4cfcb11d2134fcdd440e : SUCCESS in 36m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24e1954b650342e69e0783409c16615c : SUCCESS in 26m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c7015a16144044989bbb942127cd1a8d : SUCCESS in 28m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0480aefa894f45c7baf5847e6ac00d54 : SUCCESS in 15m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbdc8d35d53343679bdba4a2459abc2b : SUCCESS in 10m 39s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"e12c339dbaf9217096eb4681f05a087e0fa3f9ca","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-07-09 14:27:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fcb5491815361da8f8df6ae8f83ad24d09737b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 16:13:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daed46b29f544b7096248b755596f9e1 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7a568583c4474790f2e2939ad0e4f9 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab166a233ba940fc98b3fde15bbf159f : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed40621b70b0475e90745f35a882406c : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b299099dab5f4b74a1ae68989b94b514 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b345e967894f2cb0058be46f3dd4ca : SUCCESS in 9m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d97957aa19a427685d60a9ac4a19998 : SUCCESS in 6m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d8bf93ed5b341b192551da59cd52aec : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e87deb8eb57047d5a9b7470fa2fcfb03 : SUCCESS in 20m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8ee7b8798947485497f68a6355870214 : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/651bf1ce0e9c4193a98d7719f39ece8e : FAILURE in 42m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2a994299df704f81888ad447e90c2877 : SUCCESS in 59m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4de6f7f2e5d141b191e32492ad37f3cc : SUCCESS in 19m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6b2fd61d4fd42b18de7d3daedf4aaa8 : SUCCESS in 14m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5112febe7ce44dfe841b5d3aa260304a : SUCCESS in 1h 41m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/aeb5f9106fe747bc8029bde4342eae76 : SUCCESS in 34m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/416002a9875f4e96a5eb0a2ce31d860d : SUCCESS in 26m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cfa9b20f057d4ed484a6f234aae79d0c : SUCCESS in 25m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f0942b46031849f7bec1f627f030a3b0 : SUCCESS in 14m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28e15057043846c1bc14e03ac9c0ecf3 : SUCCESS in 10m 30s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"cbd55dcfd71643f958a6ea84129dbcaf349c5491","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-07-12 19:06:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0eea66249f50b76c98c1655bca41d5facf184239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 20:50:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0696f619acdf44efb8cd77a64360549e : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0587665eb7c84ce3a00eeb4b2245c022 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7efefa87d7b640e1b9ecc3091a08d747 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1b9c47209204a3699a4a64540d2e366 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89344c3f9e1d433db3a8e83949b818f6 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/421c71d5f36f444095ef904c625c2767 : SUCCESS in 8m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/642e4e84bf9049f8b2a4492c70933623 : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c2a8f6bed1747fc93400f131cf6bf29 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7d994c8542624731ae5484a4d58e6e89 : SUCCESS in 18m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/991d1b1ecd6c4770b04dac8450c37411 : SUCCESS in 1h 08m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99796daa232542488a4a665d368e0f3d : FAILURE in 40m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2ede59bb25db4cd79524df72fe4023f5 : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a8620e3737014d2586fb328480470ef7 : SUCCESS in 17m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2122327263af4f008d6d03198186831c : SUCCESS in 13m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8c89955580484e00b0b68b4e80920d21 : SUCCESS in 1h 40m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/30e67c43de9d45599d80f29d4d622efd : SUCCESS in 34m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/059c154dff7743068f88d4c9a84e524c : SUCCESS in 25m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/102f2bc07cff4792ba9ff04789cb4a0f : SUCCESS in 23m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4a9ada7a13884e12956b143a10a7c8bf : SUCCESS in 12m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac474559f3df4ce6898f894112b3b3eb : SUCCESS in 10m 09s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"58c20725c0056c1089322c5d7681b9fc6ea5cc7e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-07-12 21:23:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9e97c5676755e7bb47e76ae3a4633e30c88040af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 23:15:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e907fda658b94713b658f8ca79555cad : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7147b02e134d98b4dc8d8e018f16a1 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99609486bd684824823ba53259906a44 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3948ffdd73f14718b44468800995c3dc : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/707bef4e53de41b5a0443dd1898032a0 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e2137a0ed54f2aa74dfd9bdd0eb96f : SUCCESS in 9m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/433140416ad24d7988f5923b9df53d72 : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d918d19bab3e4abfb0eaf526613584e0 : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/60152bca982c45029dd00d02c6e6eeae : SUCCESS in 26m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cdef72c2874a4d25a812468706da4cdf : SUCCESS in 1h 09m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71bc59b5a0ed481e8c41d7bc90e2fe92 : FAILURE in 43m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/87506b10d97e4ac181c932946cadd919 : SUCCESS in 1h 12m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/365095c9c661468c812ef16da8f440ed : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b92a717fefd0477189a09140c1c37277 : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/df7c81e157554967b89d69b4dad95f60 : SUCCESS in 1h 51m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3a81a388147a4d57a5c8aad1db029465 : SUCCESS in 36m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b30d45234d4b496fbd9472e45599bece : SUCCESS in 29m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5b7092bed483427db1d73f1c991abb08 : SUCCESS in 27m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7905316c87c846d1adaa055012c481e4 : SUCCESS in 13m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af31efd885e64835893c5b0ba7254a8e : SUCCESS in 10m 42s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"426e13f3deb950786177180ea5aa03814ea94987","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-07-14 14:58:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd1969a02bdc1bf3377b1411b6fb73264348fdbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 17:17:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/776af2bd1b92428cb8b36cf4f2fc4eec : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18c04382da3b4a7690652d44f9acb0a8 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/240c159d56e94df7833dc382e14cdf03 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7aafd882a20541fe8d96e5220e3ddc57 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e72ea18f4b2d4d44857a07265bb3a894 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4314937a1fa24d9c8e987e2306981f83 : SUCCESS in 7m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36349d1779184ee9833d6512905fe7c9 : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a194342743ca48c6a625a3685fa66904 : SUCCESS in 13m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7bc0988571d2405f9027e90e5e81336e : SUCCESS in 20m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5c03d37bd8cc4c58a74df623b94788e6 : SUCCESS in 59m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd3f8cb5ca1c4f8ab9866cd1d62860d6 : FAILURE in 57m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/224dc560a6414cf9a10e4f02e9923810 : SUCCESS in 1h 00m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d373c0e24da24077b91df9c5c5282b2d : SUCCESS in 18m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f16b8290f8949d5b2055809a6068583 : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7ea97f94b72f473d82de9e40db361022 : SUCCESS in 1h 55m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3ed3b7141b714935a097ffe417885239 : SUCCESS in 54m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95f0fd0d320c46f5af98f737cf04c85c : SUCCESS in 25m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/93e6165e36d84f278361279df8fb5dfe : SUCCESS in 22m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ee2e1ce83d3c4d169a9dd5dd37990ac0 : SUCCESS in 12m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e29cbd2152d42169b09e6db113f1c35 : SUCCESS in 10m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"6e686f5e086f03bd975791e6cd89b06a646bef87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-07-29 17:09:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"09ba9e1f819b3a5d37993e2ba0a6b6d99a5f1a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 18:57:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef7b27dfb4d94a629c1014ce667ef751 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b60c3998074254a110ec55a6bab704 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5c9e3c9090a476a9e6a69bad11c38bd : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26754e5f161a48ac86f137092210a6b2 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bde407bbd1044a9e856ecbf8ac4279b5 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89cfe45da2554f4882e7ab0f94995950 : SUCCESS in 7m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17143fab98d14e67b0633882174be43a : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/110400bce8b0431bb9637464c279e24a : SUCCESS in 10m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ccd672d8b4ba461994dc301171fda623 : SUCCESS in 21m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ece696ba309b43ca94518877d54d1262 : SUCCESS in 1h 14m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccce1b4b1e3c4496948debe12b0162e3 : SUCCESS in 58m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/928d6b221f5d48b99b17952dee3ac343 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b9ce6a7abeee4d65846513003055ccd8 : SUCCESS in 17m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b73e7c04888e4a2e9cfb0807b32b6b4b : SUCCESS in 19m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fe5fefaa44734da0b80eec19aa79450d : SUCCESS in 1h 45m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3c5bd8f96a5d48c7b5d87cc5eb4c9b51 : SUCCESS in 35m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6eea2b94a0f4bcd818912ab359c4a26 : SUCCESS in 28m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a08046f5099d48dcaf181e012a11d061 : SUCCESS in 31m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c097a6340624492e9878b7631f8e4486 : SUCCESS in 12m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a78a4b8708e5433e80f97af3e7453af4 : SUCCESS in 10m 49s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"49f98dd4f59f55ccd75bb3eaf1213e0ac6364be7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-15 20:21:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8f336b576495a9bef90f35368ea50221e3ac358f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-15 22:03:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d099c367ee564aeda57f3b6ad201b85a : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b699be8651d4aa1bfdc9c6da1e8b339 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd1ad55f4bda4dca89e59114167cf6e4 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e8c857ca02b44b4a715d6588cfee48d : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0eaff72953340d1934e75a0b9b3e7e1 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3dd3615a13435ba2f5df554985676b : SUCCESS in 8m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5bfd855c16294f84a6f84ff61f6fc551 : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d4fdbd25baa44da81b0500264903325 : SUCCESS in 10m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12ef87e31cea4793afc51bdb7c983553 : SUCCESS in 20m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/738f95a6f73344288ec420158999fba1 : SUCCESS in 57m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/909d7447ea8a4adfb6a18c874f9dda44 : SUCCESS in 46m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cfddaca4dc22486bab8b0e453ccbb1cd : SUCCESS in 1h 01m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c4e0dfec8f70473382fde1b6bdef9e21 : SUCCESS in 19m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c91c4d690d145d3aa792643acc335da : SUCCESS in 13m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0ffab3ede32646e99e022f5839e753f5 : SUCCESS in 1h 39m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/338fe5ad22a2479fabe86da1d34a9c30 : SUCCESS in 30m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f0f886e91d74e74a267146fb890d670 : SUCCESS in 26m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/974f6d6e5993477297471dd2dd93b40b : SUCCESS in 27m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/955ad3df76e145a985957ca70d71f2ad : SUCCESS in 12m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9aec9652c574ff6a1f4bbf9e0b3c47b : SUCCESS in 10m 42s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":3},{"id":"3e4341bb70f924cc2c40e9fced18016582a2cfc0","author":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"date":"2021-08-16 06:19:26.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2358016bfd1a7217ce0c60beef2fd278efde5d13","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-08-16 06:24:54.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c0823585b4e847d5c8c08497d7611a91772233bb","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-08-16 11:32:29.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4ae475c74b23c0eefb8d40489fb9922879bd8a84","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-08-16 11:56:08.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1210bef2eaa4be15d1905f6c08c18c7c820ab086","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-20 08:48:57.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"497a03d2f8de155e805d7fbb5d713926f3ff9965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 10:47:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e90a41aeb8d24e3db5889bbb7865ef3a : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e01d84873c54f40b4880a1f746df471 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4081afb1dac84442868186491bfeec58 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9e2f622aebe4a15baf01b440f3c836b : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bdfc847218c4ba19d3ad9bc771926bf : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c02092ab81c43a7b627c7f47c898c1b : SUCCESS in 8m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/39130b227b6d4166a2fb9f3fe58af213 : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/80b53efafa7148b5b92bb018d3bd5054 : SUCCESS in 12m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/126ff625fc6b4d4583fa4305614a76d7 : SUCCESS in 19m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/18c63732e0484cd2a7474155cb060474 : SUCCESS in 1h 09m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3f96b776f6941c08d5dfa5b4438d91f : SUCCESS in 46m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d18d1e1fc0d24105a78b339496fd33fd : SUCCESS in 1h 11m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/adaedab0d7694262b54ad51990a74063 : SUCCESS in 28m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7ef780c09b547549a56e99f0a7ff1a1 : SUCCESS in 15m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/03b2692e17424ec995986116bd9f91d4 : SUCCESS in 1h 55m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5fb480e493a14e918a05e9e751930e5e : SUCCESS in 38m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a822f033110f412999f184b0dec4f8c8 : SUCCESS in 34m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba90d1d84afb488f8aa7c5033233941c : SUCCESS in 30m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/36a203bc5db34fe3a206e7d3ce1e8f0f : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0546b86a49d34f69a70c3380398c01bc : SUCCESS in 11m 27s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"a29cdda71b5f894fcc72cdab17df160883a0b3ed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-16 10:31:15.000000000","message":"Patch Set 4:\n\nmissing reno, docs for starters - a better motivation in commit message would be also nice.","accounts_in_message":[],"_revision_number":4},{"id":"f5024ccb941f67c249c5715f5feb0468e58b3cfa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-01 05:36:03.000000000","message":"Patch Set 4: Review-Priority-1","accounts_in_message":[],"_revision_number":4},{"id":"0c403b545f350579fba838612a9f8ad4550025eb","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-09 13:51:31.000000000","message":"Removed Review-Priority-1 by Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"a976c83201f9b9e135503a6e347a06390b32564c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 10:48:25.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"55276d1549d4baecb2a397a58253b99e66359414","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:08.000000000","message":"Removed Backport-Candidate+1 by MargaritaShakhova \u003cshakhova.margarita@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"4b35b661edeedc86dfe5a652d9cd68f766573835","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:09.000000000","message":"Removed Backport-Candidate+1 by Andrei Yachmenev \u003cweanew@yandex.ru\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"f1a122caf844648a9d69373e8460d06955e07eaa","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:09.000000000","message":"Removed Backport-Candidate+1 by Sergei Raiskii \u003cconnoreika@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"c25264cba5080f3953f86d68355d7c455967df0a","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:13.000000000","message":"Removed Backport-Candidate+1 by Ilya Popov \u003chebulrih@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"89efb13eb90cd3f2be0d11b020b8df198e0cecfd","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:19.000000000","message":"Removed Code-Review+1 by MargaritaShakhova \u003cshakhova.margarita@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"e38ee8771aebd886c9ce285cd6b1c331f8184f6b","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:20.000000000","message":"Removed Code-Review+1 by Andrei Yachmenev \u003cweanew@yandex.ru\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"7bb21e20babac760b9ff18f543e29d7db7edd254","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:23.000000000","message":"Removed Code-Review+1 by Sergei Raiskii \u003cconnoreika@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"b2aa48ac5ac10480ba978a5aef959ac38a0e0328","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:25.000000000","message":"Removed Code-Review+1 by Ilya Popov \u003chebulrih@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"7f45f1369de1a9435e617f0c5d3104f8fc3eb23b","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:34.000000000","message":"Removed reviewer MargaritaShakhova.","accounts_in_message":[],"_revision_number":4},{"id":"c847840e1c49aae5dfc52aed0a17c03a815ea699","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:39.000000000","message":"Removed reviewer Andrei Yachmenev.","accounts_in_message":[],"_revision_number":4},{"id":"3a38d0a1a5e6341e73fe18e621069262a6f16188","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:41.000000000","message":"Removed reviewer Sergei Raiskii.","accounts_in_message":[],"_revision_number":4},{"id":"588cd3f17992055e4d5f63b0258856a1a8f88d1b","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:58:44.000000000","message":"Removed reviewer Ilya Popov.","accounts_in_message":[],"_revision_number":4},{"id":"e2ed906d6426996b5a168612604329adfead4e51","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-06 14:44:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2d57e670eb875614e003230b19ecc3d81a5aa514","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-07 20:50:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"834a81032d33f3d2248bcd24a76879d844f6d7fc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-07 21:01:57.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"906505623dcc8af888dc0c12929a52905153121e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-08 12:35:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c4e8374974fe1fd4e968cf33ddc46e135174830","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-09-12 23:16:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"273f5e8967eb1357d7e992f3f2efa5911c78de51","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-09-12 23:17:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"699e1d53d545776a2bcdee304670657942c09208","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-26 12:58:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"452b7b9c58f40ed081a47b1a38cf2d6f3c6b9a6f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-26 12:59:06.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"4a2cedd21cffe407de7c5a803c485c0c3ea99e1b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-30 22:45:54.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"720d58ee6ec969aa3806afd9d66705177f982fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 01:09:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56296ca39f7641178370727b89b368f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e12a0415fc204d81befc56d52fe33f95 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/081460ee3c234c8d82ab54583c429897 : SUCCESS in 9m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/176836707e914d90992c26624059e95a : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3df6d74c6d93494093d8130a7575c72e : SUCCESS in 13m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7fc5687e6cf345879fd7301908b1130e : SUCCESS in 10m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bb0d127b3f8b4032930d3a433e7d377c : SUCCESS in 1h 04m 35s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4fa14ee345304315b80421d461598162 : SUCCESS in 53m 40s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b651e0a558c442e4a6767be28204971a : SUCCESS in 46m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/afaa4c1d55e849cda149c881ebd67dcc : SUCCESS in 1h 07m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7731939510184b92a113f0b09005386d : SUCCESS in 16m 13s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ef66d2cc4c1a4721a66e504a70a11bce : SUCCESS in 14m 31s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4881b11191e54882bbb7575fb4c1951a : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/972ef237b53f4b5f9b49ff06c89c6738 : SUCCESS in 2h 19m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbdd7dfba5ed4a2d9173e20201dd09e2 : SUCCESS in 1h 58m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8fdfc68c060c496c82bb31c797519bda : SUCCESS in 17m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f1d4154400f45258e8ac25f463ebc3b : SUCCESS in 11m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/eb4e814a252a400e908de6465921fd88 : FAILURE in 31m 08s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/10021e15ee134da8a1eef64c1840bbfb : SUCCESS in 29m 01s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4d15b4ec7db40aa94162fbd158ec3ed : SUCCESS in 19m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eb203853c3cf40a4b47b3dd591220209 : SUCCESS in 12m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d5d7a0e69774d7db34c4f69d9d1ef9c : SUCCESS in 11m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7c1bf66fda8548a0b04974cd64861877 : SUCCESS in 12m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b35aa9ed57d456bb874d05356d8d8fd : SUCCESS in 14m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22cf354d4a894acfae9eb5344ab12046 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/870795c7a28c458ca7a427e8249e1d08 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f14b6c2f4e0049688f13b53a8feb3043 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b3ef454210c0a9fb3dc869a156181337e1809858","revisions":{"91e848193897b85d4d19111a15dfcb2c4c0a6e12":{"kind":"REWORK","_number":1,"created":"2021-07-07 14:52:32.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/45/799845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/799845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/799845/1"}}},"commit":{"parents":[{"commit":"e9a3733e24c5280dda241aec839097d005014c99","subject":"Merge \"Prevent matching unrelated images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9a3733e24c5280dda241aec839097d005014c99"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-07-07 14:48:17.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-07-07 14:48:17.000000000","tz":180},"subject":"Add support multi-segments for ip-allocation","message":"Add support multi-segments for ip-allocation\n\nThis change dont\u0027t allow to filter out invalid allocations when used\nmulti-segment networks.\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: Ie52ec41c3cd9cd02501f2c8f2006aa44ac87096e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91e848193897b85d4d19111a15dfcb2c4c0a6e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91e848193897b85d4d19111a15dfcb2c4c0a6e12"}]},"branch":"refs/heads/master"},"21aeea1f59c44e45765188c4ac9bfed4f887754e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-29 17:09:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/45/799845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/799845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/799845/2"}}},"commit":{"parents":[{"commit":"da304da58eb184546f10083990b08005bafcd392","subject":"Merge \"Fix capitalisation of Ansible task name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da304da58eb184546f10083990b08005bafcd392"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-07-07 14:48:17.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-07-29 17:09:36.000000000","tz":0},"subject":"Add support multi-segments for ip-allocation","message":"Add support multi-segments for ip-allocation\n\nThis change dont\u0027t allow to filter out invalid allocations when used\nmulti-segment networks.\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: Ie52ec41c3cd9cd02501f2c8f2006aa44ac87096e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21aeea1f59c44e45765188c4ac9bfed4f887754e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21aeea1f59c44e45765188c4ac9bfed4f887754e"}]},"branch":"refs/heads/master","description":"Rebase"},"9b2e4d3fcaa95141b8bd80a740d3b0188cf96322":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-08-15 20:21:25.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/45/799845/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/799845/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/799845/3"}}},"commit":{"parents":[{"commit":"0a55c83465cd0b3d91bc357bf9eb93bf149f1527","subject":"Add Arista switch support for the Neutron ML2 genericswitch driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a55c83465cd0b3d91bc357bf9eb93bf149f1527"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-07-07 14:48:17.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-08-15 20:21:25.000000000","tz":0},"subject":"Add support multi-segments for ip-allocation","message":"Add support multi-segments for ip-allocation\n\nThis change dont\u0027t allow to filter out invalid allocations when used\nmulti-segment networks.\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: Ie52ec41c3cd9cd02501f2c8f2006aa44ac87096e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b2e4d3fcaa95141b8bd80a740d3b0188cf96322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b2e4d3fcaa95141b8bd80a740d3b0188cf96322"}]},"branch":"refs/heads/master","description":"Rebase"},"c60bc3aa8a31d15ac19d12556071be59072a5ef2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-08-20 08:48:57.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/45/799845/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/799845/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/799845/4"}}},"commit":{"parents":[{"commit":"a48cc2498859f5949d304266d32ce3e620ad7f36","subject":"dnf: add metalink argument to custom dnf repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a48cc2498859f5949d304266d32ce3e620ad7f36"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-07-07 14:48:17.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-08-20 08:48:21.000000000","tz":180},"subject":"Add support multi-segments for ip-allocation","message":"Add support multi-segments for ip-allocation\n\nThis change dont\u0027t allow to filter out invalid allocations when used\nmulti-segment networks.\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: Ie52ec41c3cd9cd02501f2c8f2006aa44ac87096e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c60bc3aa8a31d15ac19d12556071be59072a5ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c60bc3aa8a31d15ac19d12556071be59072a5ef2"}]},"branch":"refs/heads/master"},"b3ef454210c0a9fb3dc869a156181337e1809858":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-11-30 22:45:54.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/45/799845/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/799845/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/799845/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/799845/5"}}},"commit":{"parents":[{"commit":"323912d769c41bf7f99c7eabe1eec00574108541","subject":"Merge \"kolla-bifrost: Add precheck for inspection allocation pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323912d769c41bf7f99c7eabe1eec00574108541"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-07-07 14:48:17.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-30 22:44:15.000000000","tz":180},"subject":"Add support multi-segments for ip-allocation","message":"Add support multi-segments for ip-allocation\n\nThis change dont\u0027t allow to filter out invalid allocations when used\nmulti-segment networks.\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: Ie52ec41c3cd9cd02501f2c8f2006aa44ac87096e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3ef454210c0a9fb3dc869a156181337e1809858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3ef454210c0a9fb3dc869a156181337e1809858"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
