)]}'
{"id":"openstack%2Fkayobe~801447","triplet_id":"openstack%2Fkayobe~master~Ie5c21597266c8e470994eba16f3c2ed2224d0dcb","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ie5c21597266c8e470994eba16f3c2ed2224d0dcb","subject":"Support setting docker registry credentials to the empty string","status":"MERGED","created":"2021-07-20 11:10:38.000000000","updated":"2021-09-17 10:48:29.000000000","submitted":"2021-09-17 10:46:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801447","meta_rev_id":"488fe36feb041930ce8bfd5c591f3119bd79ea5e","_number":801447,"virtual_id_number":801447,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-17 10:46:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-17 08:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-02 12:39:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-09-17 08:55:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-20 12:51:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 12:39:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-16 10:36:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-09-17 08:55:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2f5b1a220e9a76821393979fc404beb1b111c617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-07-20 11:10:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6af6caa0a3c8d3ec5215c5fe5c44417bd4f0aa70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 12:51:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5467c0f6c5e1461a922c81109f92845e : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6f4b3f9ba8474597767d2966bb0c67 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1efba0f28e6641158760a09507d638bd : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77a906c18d794b44906109d2d0a76dca : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cb97663b8084fb88a18657c827211d9 : SUCCESS in 16m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc346d3e75c45a1b5f6ab531bfc9d05 : SUCCESS in 8m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca66a2d567d248ba9d09cd34f8dfe4ad : SUCCESS in 7m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6e0d6b3002dd4f57815131e363760d30 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d3b23776cd3417e879f8547aef364ba : SUCCESS in 24m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/877cb8877cea4e548bb84714dc614e8c : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e659dcc486a14cafb4c0308dc1e45890 : FAILURE in 45m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9d74d42fd13841bbb9a375f33798448d : SUCCESS in 1h 12m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ed52efe2bb974a20ba257050df4c9a6a : SUCCESS in 18m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b2d71f536394941b088d8aa5477ec1a : SUCCESS in 14m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/481db556f21c4aee99f69361971f6e38 : SUCCESS in 1h 39m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/70c9a31cc33b482a9567852bdbf6365f : SUCCESS in 34m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62c83adc7a474cfbad7c0654cbb96597 : SUCCESS in 27m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b808aeec27434275bdeee184812f032c : SUCCESS in 26m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/16da163f2c444df1b649a50ca2465cdc : SUCCESS in 11m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62ff849262c64f67b91f0184b329581d : SUCCESS in 9m 36s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"92c89dbab24e3a68f011cd148a98b6b0be79108e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-02 12:39:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"baac109685ad00bd42c4934f1d41f6f58d184010","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-16 10:36:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b060538e30fce24ee12eca8d08c0ac906bcc0dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 12:45:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db780e73590b4af4aa79cd7e19ab59bd : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc73e1ddca9645d780e16f14507b56a8 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3ea90b4b1994fc084ec0db7ce547e57 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc3bbbc5c6f844da87e59d29a4bdd880 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e93e75912554fb88b63c2ea3cf1368c : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc9affd978e74d7abf098e6c292f1204 : SUCCESS in 8m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca0c17114ed6451eaa938c69a9661944 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e043327fd4bc4099a7de738764a87601 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/57de6d0b13474b228925a27104a99c8a : SUCCESS in 31m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0361743b1afd4618abbd7abf641b7ecc : SUCCESS in 1h 00m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ec08fb69909433ba46041369e640e61 : SUCCESS in 43m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bd97d7316c784474bdafe69aaf3a7ddc : SUCCESS in 1h 09m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c9b7534546204ec9aeba20c17e6a5bc1 : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/392f90e0e5ee4ff79623ba5254f68dff : SUCCESS in 13m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ada2004214ea436284aa98253f2e82d2 : SUCCESS in 1h 55m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/205a141bb6b64501a76ad422b132d416 : SUCCESS in 36m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8933c47779c5434fb03c53042f517f92 : SUCCESS in 29m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8936d509aeca46f2b33c930760d617d1 : SUCCESS in 27m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cf4c3042eca64f0aa0742f0139ec8c12 : SUCCESS in 14m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a6e0dee88934b71869f8f1c21ac68e9 : SUCCESS in 11m 04s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"0863ed06946ef8a2da909dbf8ab487e924b15746","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-17 08:55:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19916a402fdc1be98a35fb541e685930a0af0a68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 08:55:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9bbe63417cc5c8cccb3783d4e5f59c61720b2c18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 10:46:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af2d50c1af9b4577aa11f86c0db69d31 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62bd85326114404fa330ff30b541c922 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d45a9c7cf39647499782389caadf4ed5 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5112c883f5714cfb931942b99a3feaae : SUCCESS in 11m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0d97b96f7a64107a1f089014feb2a22 : SUCCESS in 7m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2f27eabaf485481ca1ede31cccc03f37 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b4c66e9168341f6b49411d65f8aa207 : SUCCESS in 34m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8b09e2cd02204072a573a6dc6087d39d : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c9b340deca64b2f8fcf20aedd65258a : SUCCESS in 50m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4bd41c6572f5497e99a968964b7cd0a2 : SUCCESS in 1h 12m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fc254a29c3a64f9a94a37869f130c6d6 : SUCCESS in 21m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/012ec4787f3546ec99993b27c11ecde6 : SUCCESS in 15m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/578529e15d164b6c98169cffa6d816a9 : SUCCESS in 1h 48m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/36572f3673f641dfb6a9840af8383a9b : SUCCESS in 37m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87e7de7121784467957312fc32ae17cf : SUCCESS in 29m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6ca26e862e7541888967b78409dfe03a : SUCCESS in 24m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b0ea0e8a6ad649a4b86fc621fb001e3e : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd73530a7f854fc3a8393d5074d3fc3f : SUCCESS in 12m 28s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"b9f8dcd6d24921f36114afbf918ee6afc3abacea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 10:46:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"488fe36feb041930ce8bfd5c591f3119bd79ea5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 10:48:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f464dd3087b4a0a98fc91f921be5784 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0263e65bcefb9224a673a5ef5ad72caac14b587","revisions":{"d0263e65bcefb9224a673a5ef5ad72caac14b587":{"kind":"REWORK","_number":1,"created":"2021-07-20 11:10:38.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/47/801447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/801447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/801447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/801447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/801447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/801447/1"}}},"commit":{"parents":[{"commit":"e9a3733e24c5280dda241aec839097d005014c99","subject":"Merge \"Prevent matching unrelated images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9a3733e24c5280dda241aec839097d005014c99"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-20 10:56:33.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-20 11:10:27.000000000","tz":0},"subject":"Support setting docker registry credentials to the empty string","message":"Support setting docker registry credentials to the empty string\n\nThe use case is that I want to set these credentials via extra vars\nlike this:\n\n  kayobe overcloud container image build test -e kolla_docker_registry_username\u003d$HARBOR_REGISTRY_USER -e kolla_docker_registry_password\u003d$HARBOR_REGISTRY_PASSWORD\n\nSupporting the empty string means that I don\u0027t need to use any\nconditional logic.\n\nChange-Id: Ie5c21597266c8e470994eba16f3c2ed2224d0dcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0263e65bcefb9224a673a5ef5ad72caac14b587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0263e65bcefb9224a673a5ef5ad72caac14b587"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
