)]}'
{"id":"openstack%2Fkayobe~802218","triplet_id":"openstack%2Fkayobe~master~I082e1b807497d17ae40ddf86e1473fbfcdf625e2","project":"openstack/kayobe","branch":"master","topic":"bug/2009072","hashtags":[],"change_id":"I082e1b807497d17ae40ddf86e1473fbfcdf625e2","subject":"Ubuntu: Set MTU for veth in the network files","status":"MERGED","created":"2021-07-25 19:59:48.000000000","updated":"2021-08-03 11:31:03.000000000","submitted":"2021-08-03 11:29:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802218-bug/2009072","meta_rev_id":"62444ac5c99d077efc143f789a9e5b7642ea9ec9","_number":802218,"virtual_id_number":802218,"owner":{"_account_id":33466,"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","username":"lzalewski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-03 11:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-07-30 10:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-03 09:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-03 09:40:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-25 21:46:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-26 09:01:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-07-27 07:35:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-07-27 08:21:11.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2021-07-30 08:43:00.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2021-08-03 09:40:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"c6755c81e237456ed02e7e3bad3707587952c684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33466,"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","username":"lzalewski"},"date":"2021-07-25 19:59:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ff37ac813afec6167ff9a9a2c5419c875d85c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 21:46:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/578f3fc18772466a93bb88f22c1c8b2f : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e947beb8cf4a7e9442680f71b64ff7 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90fe059bff014356890de4b3e958002b : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adc4cb9aa9e748bdbe23d77e5087f0fe : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8768ea2f8e24513b36038f046d93296 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00281c22b4734c239d8bca2de0ed4ee9 : SUCCESS in 8m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f58fc09f82d492fb1df833012fb0c89 : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/768ed036904747e8b62a04ba208f2861 : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/284dfee48baf43e185d67763bd2a1604 : SUCCESS in 19m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e5f6e432d0d045b3801ea5661cda1712 : SUCCESS in 1h 03m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eb03181a2144e3f9c039ad93adb9d5e : SUCCESS in 45m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/54e989e7813842899eee30c779fa9b5a : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/040f15d6ecde4ab2a256e4e1aaef2fa3 : SUCCESS in 19m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d00bb2deec7f4ba7b5950460ec5503bb : SUCCESS in 13m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e8a4ff39ebe4985b17b8b5828aaa376 : SUCCESS in 1h 42m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/44b5b9707d364ca89ade0571fa13041c : SUCCESS in 32m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b1d897c36ee4cd5b9a604d666c3b0e2 : SUCCESS in 27m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/54c8e60a6b7446a8a575e8b9ae3d4fb0 : SUCCESS in 24m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/660d322b9fda4c729faf3027484790ee : SUCCESS in 13m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b64f7f77ff5464db6f804fd56fe2efa : SUCCESS in 10m 18s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"e7483942527cb34a21ae27e89ed6c8b9c8475dfb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 09:01:18.000000000","message":"Patch Set 1: Code-Review+2\n\nGreat, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"b7eb536f03f186125bbdbe45ac9e0bac3d05b2bb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-27 07:35:56.000000000","message":"Patch Set 1: Code-Review-1\n\nmissing release note 😞 https://docs.openstack.org/kayobe/latest/contributor/release-notes.html","accounts_in_message":[],"_revision_number":1},{"id":"b2b6981a5aafbca6dcaac239503096782fab81f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33466,"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","username":"lzalewski"},"date":"2021-07-30 08:39:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3c3630236740d4492484aa45bd59940ea75b4b95","author":{"_account_id":33466,"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","username":"lzalewski"},"date":"2021-07-30 08:41:34.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e missing release note 😞 https://docs.openstack.org/kayobe/latest/contributor/release-notes.html\n\nApologies, release note has now been added.","accounts_in_message":[],"_revision_number":2},{"id":"04d47109be6e249571bb88d8ffeb37a461543904","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-30 10:00:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e4d74927ff9a686a3132ef7230707b13161b7918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 10:26:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da1ec8c46b0b4b51bae1a170f31dee65 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a65a03de89041c79d968a466cff969b : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2386951c5b76421fb23e80fa06d4939f : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59f5122ead704eabba135fe2f1f48038 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ca190b12a77473facf9b273e8d6ec22 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38790f5d37a84dc0a22cbe0bc783215a : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6b03de2272c464a816b3526d182e247 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a1c6c2f7ddc495b81503c933e8f223e : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0bc0c31d8d75485f9f9ead5c5c414845 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f6aae09b98a4124a84368c2501673dd : SUCCESS in 19m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/30145b49f2bb4f41882f47f2ed67f091 : SUCCESS in 1h 09m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b82139446ec34a7cb6c04257af072fa1 : SUCCESS in 42m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e2d10d4c8c064b69acddff16167ce30a : SUCCESS in 1h 11m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/86a4718c0b884793b237f1abd1133a77 : SUCCESS in 18m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20e2b13cdf6f4689a41ba62acade7141 : SUCCESS in 13m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/baa41d20bcb546769ae466748766465a : SUCCESS in 1h 42m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7b389d9a9a0a4288b5bdc6d9f419ef89 : SUCCESS in 44m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f26304ac59ab44e98d0fb472b2fb0d9d : SUCCESS in 26m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2c40f723f0ca448783d308c631040043 : SUCCESS in 26m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3ee68ec208444686a943cc47a6f7b504 : SUCCESS in 12m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20de10aba53f4cb5a50a27602143235c : SUCCESS in 10m 09s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"18f2620a8a1d020df53ec160a0964782bbac3508","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-03 09:40:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nMany thanks for your contribution!","accounts_in_message":[],"_revision_number":2},{"id":"867fbfc73b5de93574aa47bf9317af70836a9b1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 09:40:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2aebf75dbb2ff79b4d8733f2b2749828cc0f2ca5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 11:29:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43c6c2992e7a47aa83a17f09bf29a7e8 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/467f227935584c85a88666c9fc2feea7 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c3736038cf640699aab48599554eda1 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d8d8ecd0dc0443eb25bbc2eef961d03 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed968ba88b314df9b0ec9e3c3603b201 : SUCCESS in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36112998aaa042e591425758afa6adad : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/165e989e1ed641fb9181620bd922f142 : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd48f395aa874442b3cad33330a3da1f : SUCCESS in 19m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c37abe45f03942d58abdc414b1dca9c2 : SUCCESS in 1h 03m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbfbc9a4de674d11802ef307632aebd8 : SUCCESS in 48m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6c8eaab7ffb74cd584a7d34f90e87a97 : SUCCESS in 1h 11m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5146de14aee04139bb833e4e78301f82 : SUCCESS in 19m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/404ee088f4af401f8d4905eb545dcbdd : SUCCESS in 14m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/131eb9dcc82a468b8e162a65328c97a6 : SUCCESS in 1h 47m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ca7552dd06344204959f6ff6cb7413ec : SUCCESS in 36m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24dac28f5a994292a2aad643c4a6b94d : SUCCESS in 32m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ec9830802d924158ada688da067d71e8 : SUCCESS in 28m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cd68c79cde8e4a1983164d15ad7241ec : SUCCESS in 13m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5a25052fe0f4d8d90ee7d730c08b6fd : SUCCESS in 10m 24s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"f2af4012cba6666d18088aba101716ded674d64d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 11:29:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"62444ac5c99d077efc143f789a9e5b7642ea9ec9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 11:31:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6080b68f85e74d098d79ec23a8c61ef6 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd8dcb4728b34e7f8b71b7a9b12f2696 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5a0206aef1f78d54845593556c13acbde99a7419","revisions":{"6fbc6daf917215c88d73336396168f98c4082fda":{"kind":"REWORK","_number":1,"created":"2021-07-25 19:59:48.000000000","uploader":{"_account_id":33466,"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","username":"lzalewski"},"ref":"refs/changes/18/802218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/802218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/802218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/802218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/802218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/802218/1"}}},"commit":{"parents":[{"commit":"f0aae6b8bc468ebb136f80374cfd3fce3c65b2d7","subject":"Remove iscsi deploy interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0aae6b8bc468ebb136f80374cfd3fce3c65b2d7"}]}],"author":{"name":"Lukasz Zalewski","email":"lukasz.zalewski@kcl.ac.uk","date":"2021-07-25 13:10:17.000000000","tz":60},"committer":{"name":"Lukasz Zalewski","email":"lukasz.zalewski@kcl.ac.uk","date":"2021-07-25 19:35:33.000000000","tz":60},"subject":"Ubuntu: Set MTU for veth in the network files","message":"Ubuntu: Set MTU for veth in the network files\n\nConfiguring veth pair MTU in the systemd-networkd .netdev files\ndoes not set the correct MTU for the peer and there isn\u0027t\nconfiguration option in the [Peer] section to do so [1].\nThis results in an MTU mismatch with peer having the default\n(1500) value.\n\nThis change moves the MTU configuration from the .netdev file\nto the [Link] section of the corresponding veth pair .network files.\nSame approach is already recommended in [1] for the tun and\ntap devices.\n\n[1] https://www.freedesktop.org/software/systemd/man/systemd.netdev.html\n\nStory: 2009072\nTask: 42882\nChange-Id: I082e1b807497d17ae40ddf86e1473fbfcdf625e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fbc6daf917215c88d73336396168f98c4082fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6fbc6daf917215c88d73336396168f98c4082fda"}]},"branch":"refs/heads/master"},"5a0206aef1f78d54845593556c13acbde99a7419":{"kind":"REWORK","_number":2,"created":"2021-07-30 08:39:58.000000000","uploader":{"_account_id":33466,"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","username":"lzalewski"},"ref":"refs/changes/18/802218/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/802218/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/802218/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/802218/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/802218/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/802218/2"}}},"commit":{"parents":[{"commit":"da304da58eb184546f10083990b08005bafcd392","subject":"Merge \"Fix capitalisation of Ansible task name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da304da58eb184546f10083990b08005bafcd392"}]}],"author":{"name":"Lukasz Zalewski","email":"lukasz.zalewski@kcl.ac.uk","date":"2021-07-25 13:10:17.000000000","tz":60},"committer":{"name":"Lukasz Zalewski","email":"lukasz.zalewski@kcl.ac.uk","date":"2021-07-30 08:35:22.000000000","tz":60},"subject":"Ubuntu: Set MTU for veth in the network files","message":"Ubuntu: Set MTU for veth in the network files\n\nConfiguring veth pair MTU in the systemd-networkd .netdev files\ndoes not set the correct MTU for the peer and there isn\u0027t\nconfiguration option in the [Peer] section to do so [1].\nThis results in an MTU mismatch with peer having the default\n(1500) value.\n\nThis change moves the MTU configuration from the .netdev file\nto the [Link] section of the corresponding veth pair .network files.\nSame approach is already recommended in [1] for the tun and\ntap devices.\n\n[1] https://www.freedesktop.org/software/systemd/man/systemd.netdev.html\n\nStory: 2009072\nTask: 42882\nChange-Id: I082e1b807497d17ae40ddf86e1473fbfcdf625e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a0206aef1f78d54845593556c13acbde99a7419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a0206aef1f78d54845593556c13acbde99a7419"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
