)]}'
{"id":"openstack%2Fkayobe~802242","triplet_id":"openstack%2Fkayobe~master~I202e1d2ec54e8ebcddefab668f8e2486f1bb1aa6","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I202e1d2ec54e8ebcddefab668f8e2486f1bb1aa6","subject":"Fix capitalisation of Ansible task name","status":"MERGED","created":"2021-07-26 09:08:32.000000000","updated":"2021-07-28 14:43:24.000000000","submitted":"2021-07-28 14:38:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802242","meta_rev_id":"2f7f5000d3a4585ebb9de5406816fa98dcf4f7c8","_number":802242,"virtual_id_number":802242,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-28 14:38:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2021-07-28 11:54:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-07-28 14:43:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2021-07-28 12:05:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-26 09:12:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-07-26 10:54:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 11:54:47.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"}],"messages":[{"id":"2907117a11bf6a7fe55642c801cb588f3b718c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-26 09:08:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c14d892480318fb710943653a21a1c5f47e81aba","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 09:12:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56ca066623101e7497585c6b852e9ae0a65f9e3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 10:54:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3a5d92e87c04748a351034bbeb5ba02 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb3c47aa83884d0d89dcab51e83223bb : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a449b95db11448a8db372ca5813bcc4 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d08d7ac023345f2adb4e0ddeba8eeb1 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/752c457c640447b7916576861e8871b9 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7853dd0a5ff74d28b937824dd9ee367b : SUCCESS in 8m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e575ca2e50a040a7b68b0a69475fcdb1 : SUCCESS in 8m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/60bfbf46f3f74ad6afea307e46869cf9 : SUCCESS in 12m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9dff10654c7d4070a68300a63e5c2f73 : SUCCESS in 22m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b223b1b1337f40c5842f7747be7d8222 : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d84d6bdb5c448d9abb86b3115b0a582 : SUCCESS in 41m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f03fc8dcd4354a48aaa92ca792b1900e : SUCCESS in 1h 12m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/00a385de3a6744d4bd24f37061d0df8e : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7f909f9076c48469a3e9b39f51f3e50 : SUCCESS in 14m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1130699fde764f73af6bf3ec1ae784ff : SUCCESS in 1h 42m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b1c979999cc546b999fb3eeadd654d61 : SUCCESS in 34m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67cc44666fa94eb0bab83718b13b8785 : SUCCESS in 29m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/53f7110852cf4e1c84ff9f076f5f757a : SUCCESS in 28m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/895a7d4222844060a6e0e1d15095ae53 : SUCCESS in 16m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eef644d8dc974846a61eb70fa8dbcd3c : SUCCESS in 11m 03s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"20a9323f01d1a6ec936707f75482132c72622359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 10:55:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7ad547fd47632c62104a9b7c1ddb01bfec17a23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 12:53:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3f5f40db572460f9d4889127f0cb59e : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43ccb50916db409eb2964b698934a7fd : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/652cdde8d4a04a4185df9271e4204e8a : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f2dd473d0e14c45a95f90b2a3d9ee15 : SUCCESS in 9m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9d207b0548c0424aa8b4d36b1e6d1409 : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a1f4b6521cfa4f0baf617a869d035700 : SUCCESS in 12m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d92ab1184708417e97d08f3bb7f38abe : SUCCESS in 20m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fabebd67e3084c3293cf43d73ee46499 : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/382d8be554834f09a67b608c3bf7698b : SUCCESS in 48m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f77780685f284a868ef484d50a916af9 : FAILURE in 22m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/532ebe86352c4d50bd8f36f930dd6983 : SUCCESS in 19m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/883ed01be68241c9b392fc4f503d9ac6 : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b06574cd9faf4cb48a26e77e6d7aa953 : SUCCESS in 1h 53m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eb0ab9679eb547b2b8c3e2d924ed27d1 : SUCCESS in 33m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0eda846976647dea0b850114a0974df : SUCCESS in 28m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/46b56b24466d4b9d847f54d85291dbbb : SUCCESS in 28m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/31a7f755dff5485089ebf33d1e8a86cd : SUCCESS in 13m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb9fe5a781b9470b9ee005602dced5da : SUCCESS in 10m 38s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"6cf4ed8d2c5e9178f6018f123e801acf567037ae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-26 13:39:13.000000000","message":"Patch Set 1:\n\nrecheck\n\nNetwork issue:\n\n    \"msg\": \"Request failed: \u003curlopen error [Errno 101] Network is unreachable\u003e\",\n    \"url\": \"https://tarballs.openstack.org/ironic-python-agent/tinyipa/files/tinyipa-stable-wallaby.vmlinuz\"","accounts_in_message":[],"_revision_number":1},{"id":"506c4bc8e51d623b5e7b6bc251f01169b9af7604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 15:44:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4c9fc833a254ac38021f3e99d10815f : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9a2b5b61a4948e3bc67fb14c018a395 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f3a9716d89f4527bb6b3e92a667e994 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0bcfbc2b4884bb3b3363700d3fc3601 : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52bea3bdb67442d3911215016fbe4b52 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e028bd4b9049a490adbe50c152b5cc : FAILURE in 7m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/54308014ce1c4001b65a0805a09eee55 : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b14a7483395d47eaabce9a0f9090e1d5 : SUCCESS in 9m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/671998b6c488494eab18dc4d519e88f1 : SUCCESS in 20m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/84eb17f67ac64538b7a1addc6f8988b8 : FAILURE in 1h 00m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f07c9723f4614f00a899692cabb01ab2 : FAILURE in 36m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/75bb14ba98394dd5b6c88d3def07365c : SUCCESS in 1h 01m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2bcc5828dc764e85b95ae46b7058c58b : SUCCESS in 19m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9840168adb50463488cf5b3aac9e3da6 : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cdc80d84e032489790b62f8cf6fa76fa : SUCCESS in 2h 00m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c1d9bc911c9e47d3a410805cfd1f9d38 : SUCCESS in 33m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e0b8f9c72bd4fc38f4234e798629362 : SUCCESS in 30m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b4a9b3d95d324ace93e0bfd072876e51 : SUCCESS in 26m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bbd1b3a020d643d0b4237ae649ea9f0f : SUCCESS in 13m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25804a10ac50458ca4b62e6a50268d5d : SUCCESS in 10m 51s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"353bd449457d825a17ad952e19680b7c4c49aa0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-28 10:44:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"60e86d9dda3da8e01897fff359a261cdf3c638fc","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-07-28 11:54:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2f157efa70ca0c8f18228afad6a2e8067da1205c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-28 12:05:07.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"72405ae10bcb41e9c5c7688ee6bbb09e5d127b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 12:48:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4592eea0d2fe4c3db24d9881fd7edd97 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abf55e7f03c94b1691020e26364f8ed0 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0feee78483d4c39a1ec84f0f519cb6d : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08406c426aff48debf27d807afd74634 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20bfdeb4a6d648f89a6401084113c36e : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4fe4831871433e9d00cd9901225613 : SUCCESS in 8m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46464eef6d394310a56c013bc0ff0f3a : SUCCESS in 6m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1bc959e3db8145ce9166c928d1d66587 : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4997481c9fd448b18e5a4a1a392c4e14 : SUCCESS in 18m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e701ab7646024ddba4bd1246a5a1d00f : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94798f7303694770974cbc2aa5906abf : SUCCESS in 42m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1895ec0f370d4c8e824d98c374f7855e : SUCCESS in 1h 06m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7cdf69de70f346d19e9f681e21b2325f : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1111c5c7b3d044c8a0fd4227286933c2 : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c1d259cc36124bbcb85646dcf4b9f4d1 : SUCCESS in 2h 01m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b1b7bfd352e741bba7a992fdfed0aabb : SUCCESS in 31m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18a634ab39824047871129e903901e1a : SUCCESS in 28m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/772bc3c37c0d4855a9b0c7975feddcd0 : SUCCESS in 23m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6b93688ca0e94fcca9b226288443c903 : SUCCESS in 14m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1640b873273469da05129136b6c1164 : SUCCESS in 10m 02s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"f489ebd532e928d374bbb19b3cdaef21de5af893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 12:48:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5456c84201aa81e2dd1770672c9ab145a8e03561","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 14:38:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27a1e14eb82a43e989f78aa8d7561ac4 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/705eceb378d94478a6d74bb2eff1c2c9 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61307cd249fa4b92890ee925af535e19 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06ed6d4dfe9942fc9491827d37943970 : SUCCESS in 9m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7542e93228cd40b5956d8b2a0ce77628 : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/83bc2771d9e6431c97ccd7065fd96832 : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fdc5401689614d4b8af553045b5409a2 : SUCCESS in 20m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/eb30c5685be94548aa6cd4c1025e52f3 : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72e260f17cca48e683647cec364d1a4f : SUCCESS in 47m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4def8aa5fbdd4004b7f1e1c977898187 : SUCCESS in 1h 04m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/da505d4e946d43f68297e2c4c90a7760 : SUCCESS in 19m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76fd2cc2b3624c32b001e39070cffbb1 : SUCCESS in 14m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9e76e74ef3904767b91eccdf8ee50593 : SUCCESS in 1h 44m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6af049a460eb49b9acf5c2a28be2598a : SUCCESS in 33m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6b3d25d84b745dd8d1d9c2f372071bb : SUCCESS in 26m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f5a0be087e3344b2a9202e14796ee9cc : SUCCESS in 26m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5e7a5140083f442a81d7ac70a343fc34 : SUCCESS in 14m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d502893d718c4397bac4aae931ee2275 : SUCCESS in 10m 13s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"ef91bbe9640f7a93af596be10c4efb3f7c43c7b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 14:38:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ccf17dd66cfaa79268a930909de2446c409e7d79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 14:43:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75fa9a5f4a814465b952fe06759d153f : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5815f0ba3a92d8971056b01e7e3a42c66e42ead9","revisions":{"5767e05409131334ee7af495397a1835d3bdd023":{"kind":"REWORK","_number":1,"created":"2021-07-26 09:08:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/42/802242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/802242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/802242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/802242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/802242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/802242/1"}}},"commit":{"parents":[{"commit":"f0aae6b8bc468ebb136f80374cfd3fce3c65b2d7","subject":"Remove iscsi deploy interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0aae6b8bc468ebb136f80374cfd3fce3c65b2d7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-23 14:38:19.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-23 14:38:19.000000000","tz":120},"subject":"Fix capitalisation of Ansible task name","message":"Fix capitalisation of Ansible task name\n\nChange-Id: I202e1d2ec54e8ebcddefab668f8e2486f1bb1aa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5767e05409131334ee7af495397a1835d3bdd023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5767e05409131334ee7af495397a1835d3bdd023"}]},"branch":"refs/heads/master"},"5815f0ba3a92d8971056b01e7e3a42c66e42ead9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-28 10:44:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/42/802242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/802242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/802242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/802242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/802242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/802242/2"}}},"commit":{"parents":[{"commit":"b8ff8fa811735d4a0610cd313e2583aabd92b34d","subject":"Add missing font for PDF generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8ff8fa811735d4a0610cd313e2583aabd92b34d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-23 14:38:19.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-28 10:44:53.000000000","tz":0},"subject":"Fix capitalisation of Ansible task name","message":"Fix capitalisation of Ansible task name\n\nChange-Id: I202e1d2ec54e8ebcddefab668f8e2486f1bb1aa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5815f0ba3a92d8971056b01e7e3a42c66e42ead9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5815f0ba3a92d8971056b01e7e3a42c66e42ead9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
