)]}'
{"id":"openstack%2Fkayobe~802684","triplet_id":"openstack%2Fkayobe~master~I0fde948a306ef9b8cf9525f1e93650d36d5f14f6","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I0fde948a306ef9b8cf9525f1e93650d36d5f14f6","subject":"seed: Disable SNAT configuration by default","status":"MERGED","created":"2021-07-28 10:51:11.000000000","updated":"2021-09-30 10:36:50.000000000","submitted":"2021-09-30 10:35:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"802684","meta_rev_id":"48a61e84e485dc86671c9a8d3250db40f38e07f3","_number":802684,"virtual_id_number":802684,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 10:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-30 08:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-29 19:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-30 08:17:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 12:55:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-04 08:37:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-09-28 13:43:48.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2021-09-29 07:24:53.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-09-29 08:51:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-29 16:46:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"37eed6a249267eb9668bb05a38d99fa79d5ce38c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-28 10:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bad1d19e668af13f15d87862a3e4bac2be4b4e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 12:55:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/087e2af0e2d54c1a8d67866349220654 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3689e58088094085a9ab904904b6e5a1 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7bfea351e4c477099d3e6307bbef853 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deb16719f24d4af58782c30dcabc8fd3 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7112d4033ca34c8dac38610f949c448f : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12d293b12cf2470e89deb5e7ea8424cc : SUCCESS in 9m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b715c0fbadb43059bdefa8411e0ca0a : FAILURE in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d2135a9ae08a4a98abe77b11cd8473a0 : SUCCESS in 12m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8502563f6ef64721afacdd931b8eecef : SUCCESS in 20m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c9a3bdd2c5f94205945d619477fec4f4 : SUCCESS in 1h 10m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e85543a3d68b45a0ae9fd56843dba58d : SUCCESS in 50m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/57da8b4dea6b4c6ab7bd2888f7f32a66 : SUCCESS in 1h 11m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d7e7c485718341f7a5b4adbc9d8afded : SUCCESS in 18m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5c87b98ba44448cb847b2e4ffdb19e7 : SUCCESS in 13m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/63db2aba4ba64b1aae7af4125c441a38 : SUCCESS in 1h 58m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/10d15653beb54a11a9052dbfda9bae55 : FAILURE in 9m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa9b122cd155484793dc14c440f24b5d : FAILURE in 8m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0392c28b01c9488f98016af836a49984 : SUCCESS in 29m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9830d6436e4c406ca9bce40052eb9ba8 : SUCCESS in 13m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6f3bc597e0349f28a33398827c716f4 : SUCCESS in 10m 38s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"aa3b4468ffd5ad9fffac4b8b88439a26ca3fba46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-04 04:38:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7cb8392efbe6434807b63319467a75f90aeb9633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 06:24:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79e8c14a1f124a5f88ed0395ce6913d7 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a599fdcd0481448290f98ce1cc73e95c : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae7fe24394bb4b21bc87ef0da64e83de : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea53e9feb6fa4c27aa71f330db7e7bb5 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30bcbc5a465a48ddbe9b334633ed7725 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55d4c527cbac4e33be5ce634475be026 : SUCCESS in 8m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/16ac578fc50d4883ade88ff4d9d0db89 : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d82d3f99a1174dc6b641404a90601306 : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac78a39c58c140979d7628fa2ee00314 : SUCCESS in 18m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2bdeaab5f72043a48218b0c6d5d5716d : SUCCESS in 1h 03m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf78443ade9146ea9e5ad08f75895974 : SUCCESS in 43m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f5c0056a4a5648068d9411e86877a5ab : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/12d9b55c17084782bc86e686aa59905d : SUCCESS in 19m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bcc0e5d174642518afbd47d33061da4 : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/54d7884af1674a76bc14be2c9a75570a : SUCCESS in 1h 43m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0561dedfb9144ccbabedaffb98e1906d : SUCCESS in 33m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cea2eab861ae4363ac43fa2ee8edcbce : SUCCESS in 28m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/af5d70bb639b434c90ead4bf104e0a07 : SUCCESS in 26m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5eaa5bde58d94c55bd4799e7ee7365f2 : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/521bced8025e4aed91b2e2be7fc433d5 : SUCCESS in 9m 34s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"00c96a97f6c6329275ed4a1fc6e2dea17863aa53","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-04 08:37:42.000000000","message":"Patch Set 2:\n\nreno pls.\n\nAlso, do you need the IP routing on the seed?","accounts_in_message":[],"_revision_number":2},{"id":"4e7f005561f90545f70fdb7f56107ec73ff971db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-28 13:03:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3c09a6019d1db690b2dc1c6138c6be39df1f3b1c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-28 13:43:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9b42cca93630bc2ed84d7c0adb1c8c259dbfe29a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-28 13:53:15.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"584a7999225c07aa43de211a2eac37ef40015b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 15:53:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3454cc6da59b4567ad611a252f6e249a : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/133a80e71b494009aabed98f2c1b05d7 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b83189e1dfa495c9206ba0a21ab2916 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a421f67c79b54a2dabd1391a711808a2 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37b0412276004050b3a491f309b76d9e : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec6a6db3388649fb8e50bd6c3a0e1a30 : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae744e5d5e9d40b0bbd6b77faf190d10 : SUCCESS in 5m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ecee48c78cf446f29be170fbd2f27e67 : SUCCESS in 7m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd5a99955ee24c26a115b454aa20ff0b : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e3f9aa4a9d14f309d1c2b191d34e193 : SUCCESS in 29m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/be7885d96e41495da7d7ddd8d8ee5794 : SUCCESS in 1h 04m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/974c3d1d54814976801666eb626bafb4 : SUCCESS in 50m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b2e5b691b28f4067bc502971e2af5db9 : SUCCESS in 1h 06m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cfbd8551d02a4228b43ca8613012c62f : SUCCESS in 20m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bc2dca847314c96b9fc57b17500fb67 : SUCCESS in 16m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4c45eb6718a6453186226e0011339c14 : SUCCESS in 1h 54m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/885dbba63e344c04b675cc52cc7dbd3e : SUCCESS in 34m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2557112cdfc44373b44f33e644c9ff99 : SUCCESS in 28m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b28c015c3b794baf8eac6d07b7f9cbe7 : SUCCESS in 34m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1609e822ca394f6bb6c93ee1969f1cf2 : SUCCESS in 14m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0b0c815fbff41258758aee602445a62 : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":4},{"id":"0bd9ac3d22b302825c8875577221a7884ba8d422","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 07:24:53.000000000","message":"Patch Set 4:\n\n(4 comments)\n\nLooks OK, some minor comments.","accounts_in_message":[],"_revision_number":4},{"id":"441a05d6dda94c492b9f0746ab223c467cda8e89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-29 08:32:40.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"35f2d34ddfe3053e368a7ac3f8323b7fd1a8c1c9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-29 08:51:16.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"34c2d657f37eb5144c66887184df733582813edc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-29 09:00:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ac5885f29fddcc3ffdc298230d41f0494391eb5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-29 09:01:10.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"a7e8eb04ce30312595e8daf0d94838f3ef609b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 11:04:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf3ad1608f494585836f0d4fafaeb180 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/629625d855da4654b0a8f27de78c186f : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24a31d141a684a7780499de373e267c3 : SUCCESS in 8m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/506c2f5110214ae091fee991503ded3a : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41e28a2b18164c5582c9b59c0a606030 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4be64e88c194f1f902535fc6dd6405d : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/055b03ee7705497f9b149f81d6f49bc2 : SUCCESS in 4m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d27d294e3ca4299b4b1aedc0b541e7d : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d47d8ac90faa4409a3ccf32bed96c7ec : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/47c015cd0136472ea6c204fcee53f96d : SUCCESS in 31m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ac95cc0310c94b26bbe70a8b8538bbb2 : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/791e8f352e2743b88dfc6fa23ced487a : SUCCESS in 49m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ad1953f01b6b4ec19c184d19f77a141f : SUCCESS in 1h 05m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/30623952108442278fda71bf575f0fd1 : SUCCESS in 20m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7cf74c7ce4444d48ef17cebff822344 : SUCCESS in 20m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a099a4feb87243b2b175352db3bf39cd : SUCCESS in 2h 00m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8017fd2ee51546019a7b19f22bf3ac8b : SUCCESS in 38m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4742b8e5bc064595a120b0ec73900c6f : SUCCESS in 29m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/aeb958900eb04657b188a46d194e6d58 : SUCCESS in 29m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3ac120853bf84b53a0d7068fc338d4f0 : SUCCESS in 15m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45fe831bdebf4e1183abbabc632ec33c : SUCCESS in 14m 59s","accounts_in_message":[],"_revision_number":6},{"id":"080f0f5fddb4746b273625cb994ad551ad12ef7a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-29 11:54:48.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"cefdb2876c6ba68c58d079bd01c989e90646b6f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 16:46:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6614d32b9f085a4dfd60250a5f8e3fc97c9f8f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 18:45:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3ede3bbc14e47fcb0a34f6b6a413fa4 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69b71f0051ab4a14ba6156e3689f7433 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/876326e40d474f119a797ea2ef097ce1 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59ef2955ab3945e4a7f47a06b34d139b : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0fc03d0c80844928942664475702f90 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34fb25fdc4c649648e83bcf628ccb430 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e5f656c2f9f490dadc36478bfb58b8f : SUCCESS in 5m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7bbfd01db68c4229ba61d130edffd48f : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1780408eca484c6e9a09ffa0487fbc74 : SUCCESS in 12m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c94d8f4b791401a88331dde980b4064 : SUCCESS in 34m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6839df671fd54479b08aed6410d014c1 : SUCCESS in 1h 11m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/851ed5a5087348ba9fdafd5330fd6e33 : SUCCESS in 50m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/63cd85aacced4d959a9ece3b2de46e99 : SUCCESS in 1h 13m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fd1999f7c9d44ca8b3d0cd82bc433ac5 : SUCCESS in 22m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34ba025ef7a7452a87320c681934be9c : SUCCESS in 17m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2db558ea31f4c1db6aa741be98c96ce : SUCCESS in 1h 56m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0ea44db5b5c643b0b61b8390747c030c : SUCCESS in 44m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/372067b55690473db65c8764379b7ed3 : SUCCESS in 35m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c1b5fa64463942ffbcbdc7822bf7d1bc : SUCCESS in 29m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bca1140e05f44b11a87103bd1df67594 : SUCCESS in 17m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/375897c557594ab7a36d288d3a0a94ef : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":7},{"id":"ef31b5c92e9cb743c1d637b7958657b3df4b5f87","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 19:53:20.000000000","message":"Patch Set 7: Code-Review+2\n\nI updated the reno to avoid a misunderstanding.","accounts_in_message":[],"_revision_number":7},{"id":"c62d28e8a56eb0a49f58ba34952c94fcbf8b1865","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 08:17:19.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"72b31e5418ef171ec5ad66c597af8cd17e8756a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 08:17:58.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0007dcb96ca76970830f37cb844ff718e01135b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:35:16.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7492f9420f114228b531b5d7eaf1e980 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52fee1faff88469fa6a62059f301defb : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40f41098230a41548a11f0fdde9fd1cb : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8843c0507a1a4ab19b772b02065492ae : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edd72892d6174a4391f16426bb8f6986 : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1b08a217f2d942168083f4c990f1a8a8 : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/755364b7fde645048ff07586ac6c61c5 : SUCCESS in 12m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba9f11dfc7c14bd3905e6f13de1c32d4 : SUCCESS in 33m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a62e51bc789c423f81e2199588f6de65 : SUCCESS in 1h 21m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be64a99b92464dfd91b884be29910492 : SUCCESS in 53m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/800c6eca508e47b990a594163ecc7368 : SUCCESS in 1h 27m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0994abf487854b79b3dbbec77d71a3df : SUCCESS in 20m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53e6fed8d4c54eb6b2341bbbb40f197d : SUCCESS in 17m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cf8081fc223746c7abbbc7c5ea61189d : SUCCESS in 2h 11m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2871e09dec8347dba37ce7bf79156d5e : SUCCESS in 44m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f055be1dfed45c9a2d4e6256781d679 : SUCCESS in 40m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c2b9b328679843408aa9a57b6cf6784a : SUCCESS in 30m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ce7bc361c98147459b5145d4f6716b98 : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/823904f0a68f49e686ff74bae8163a0f : SUCCESS in 14m 59s","accounts_in_message":[],"_revision_number":7},{"id":"2fdbf333c8ae59c9175639c1618c8d136562e0a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:35:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"48a61e84e485dc86671c9a8d3250db40f38e07f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:36:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b7ebbe6cbf44dab87facbe3b5a9e0c7 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b73e48c41004903ae191cf5897aa7d9 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"53f37da19815592799d63eb7880088bb197430b3","revisions":{"e75efdaef84edd1040e65e576b3c3e5f13e3a545":{"kind":"REWORK","_number":1,"created":"2021-07-28 10:51:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/802684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/802684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/802684/1"}}},"commit":{"parents":[{"commit":"b8ff8fa811735d4a0610cd313e2583aabd92b34d","subject":"Add missing font for PDF generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8ff8fa811735d4a0610cd313e2583aabd92b34d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-28 10:50:06.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-28 10:51:07.000000000","tz":120},"subject":"seed: Allow disabling of SNAT configuration","message":"seed: Allow disabling of SNAT configuration\n\nIntroduce seed_enable_snat with default of true\n\nChange-Id: I0fde948a306ef9b8cf9525f1e93650d36d5f14f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e75efdaef84edd1040e65e576b3c3e5f13e3a545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e75efdaef84edd1040e65e576b3c3e5f13e3a545"}]},"branch":"refs/heads/master"},"784f1c2a82c4a07d35f2bde42de753ed62c909cd":{"kind":"REWORK","_number":2,"created":"2021-08-04 04:38:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/802684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/802684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/802684/2"}}},"commit":{"parents":[{"commit":"b8ff8fa811735d4a0610cd313e2583aabd92b34d","subject":"Add missing font for PDF generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8ff8fa811735d4a0610cd313e2583aabd92b34d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-28 10:50:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-08-04 04:38:40.000000000","tz":0},"subject":"seed: Allow disabling of SNAT configuration","message":"seed: Allow disabling of SNAT configuration\n\nIntroduce seed_enable_snat with default of true\n\nChange-Id: I0fde948a306ef9b8cf9525f1e93650d36d5f14f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/784f1c2a82c4a07d35f2bde42de753ed62c909cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/784f1c2a82c4a07d35f2bde42de753ed62c909cd"}]},"branch":"refs/heads/master"},"ffdd4d528fc9a6ed33756e2ea707d180479b1a25":{"kind":"REWORK","_number":3,"created":"2021-09-28 13:03:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/802684/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/802684/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/802684/3"}}},"commit":{"parents":[{"commit":"b8ff8fa811735d4a0610cd313e2583aabd92b34d","subject":"Add missing font for PDF generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8ff8fa811735d4a0610cd313e2583aabd92b34d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-28 10:50:06.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-28 13:02:45.000000000","tz":120},"subject":"seed: Disable SNAT configuration by default","message":"seed: Disable SNAT configuration by default\n\nIntroduce seed_enable_snat and change default to false\n\nChange-Id: I0fde948a306ef9b8cf9525f1e93650d36d5f14f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffdd4d528fc9a6ed33756e2ea707d180479b1a25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffdd4d528fc9a6ed33756e2ea707d180479b1a25"}]},"branch":"refs/heads/master"},"dde5fcecf3a3f1ec9f4cd11a5567217782fd6d8b":{"kind":"REWORK","_number":4,"created":"2021-09-28 13:53:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/802684/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/802684/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/802684/4"}}},"commit":{"parents":[{"commit":"b8ff8fa811735d4a0610cd313e2583aabd92b34d","subject":"Add missing font for PDF generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8ff8fa811735d4a0610cd313e2583aabd92b34d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-28 10:50:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-28 13:53:06.000000000","tz":0},"subject":"seed: Disable SNAT configuration by default","message":"seed: Disable SNAT configuration by default\n\nIntroduce seed_enable_snat and change default to false\n\nChange-Id: I0fde948a306ef9b8cf9525f1e93650d36d5f14f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dde5fcecf3a3f1ec9f4cd11a5567217782fd6d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dde5fcecf3a3f1ec9f4cd11a5567217782fd6d8b"}]},"branch":"refs/heads/master"},"5dff48dfb4373546a2d4d85c1c01bf11a6473a74":{"kind":"REWORK","_number":5,"created":"2021-09-29 08:32:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/802684/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/802684/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/802684/5"}}},"commit":{"parents":[{"commit":"b8ff8fa811735d4a0610cd313e2583aabd92b34d","subject":"Add missing font for PDF generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8ff8fa811735d4a0610cd313e2583aabd92b34d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-28 10:50:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-29 08:32:36.000000000","tz":0},"subject":"seed: Disable SNAT configuration by default","message":"seed: Disable SNAT configuration by default\n\nIntroduce seed_enable_snat and change default to false.\n\nChange-Id: I0fde948a306ef9b8cf9525f1e93650d36d5f14f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5dff48dfb4373546a2d4d85c1c01bf11a6473a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5dff48dfb4373546a2d4d85c1c01bf11a6473a74"}]},"branch":"refs/heads/master"},"dd43c20b2751899b95a5d5278be5491d9d156de5":{"kind":"REWORK","_number":6,"created":"2021-09-29 09:01:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/802684/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/802684/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/802684/6"}}},"commit":{"parents":[{"commit":"b8ff8fa811735d4a0610cd313e2583aabd92b34d","subject":"Add missing font for PDF generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8ff8fa811735d4a0610cd313e2583aabd92b34d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-28 10:50:06.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-29 09:01:07.000000000","tz":0},"subject":"seed: Disable SNAT configuration by default","message":"seed: Disable SNAT configuration by default\n\nIntroduce seed_enable_snat and change default to false.\n\nChange-Id: I0fde948a306ef9b8cf9525f1e93650d36d5f14f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd43c20b2751899b95a5d5278be5491d9d156de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd43c20b2751899b95a5d5278be5491d9d156de5"}]},"branch":"refs/heads/master"},"53f37da19815592799d63eb7880088bb197430b3":{"kind":"REWORK","_number":7,"created":"2021-09-29 16:46:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/802684/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/802684/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/802684/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/802684/7"}}},"commit":{"parents":[{"commit":"f09faa43d1a1d38b30967cce948af6b60c34be70","subject":"CI: Disable libvirt debug logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09faa43d1a1d38b30967cce948af6b60c34be70"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-28 10:50:06.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 16:46:25.000000000","tz":120},"subject":"seed: Disable SNAT configuration by default","message":"seed: Disable SNAT configuration by default\n\nIntroduce seed_enable_snat and change default to false.\n\nChange-Id: I0fde948a306ef9b8cf9525f1e93650d36d5f14f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53f37da19815592799d63eb7880088bb197430b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53f37da19815592799d63eb7880088bb197430b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
