)]}'
{"id":"openstack%2Fkayobe~802709","triplet_id":"openstack%2Fkayobe~master~I0c41d59123aba3351ac3b0a5a76cff517cca93e4","project":"openstack/kayobe","branch":"master","topic":"inspector-port-desc","hashtags":[],"change_id":"I0c41d59123aba3351ac3b0a5a76cff517cca93e4","subject":"Add twentyFiveGigE to port descriptions in inspector rules","status":"MERGED","created":"2021-07-28 13:16:52.000000000","updated":"2021-08-23 10:42:34.000000000","submitted":"2021-08-23 10:41:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802709-inspector-port-desc","meta_rev_id":"48003a5dd538283fa34a41d5408902010c9cdca0","_number":802709,"virtual_id_number":802709,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-23 10:41:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-08-19 08:53:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-08-23 08:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-08-23 08:50:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 14:59:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 09:07:51.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-08-18 21:20:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"b5e64f5849c83b9a4844a4b04eeeab86c1dd9b5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-28 13:16:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31a1d3578f392c06c7e93a046f84d6d103610bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 14:59:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fa605bd890348d5ba31bdbc6d02f439 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6fefc3813a140d69a09c9ca93ddeb8b : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b62d07cafb749ea9b968809468826d3 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b16fe6300ae4978817a119c6f9ff6f7 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef9691b6079147d0b96c6afc77fafad0 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/024882e0002941f8b7847c2b64cf9cc0 : SUCCESS in 10m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef918d51017049c69eb7989087e2309b : SUCCESS in 6m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aeec340ea4ec45328fd7ef0ff3ab8e64 : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/acd74381937c438c917842ef143755b3 : SUCCESS in 19m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/522829b1f2274baeaa81b541c01b6de0 : SUCCESS in 58m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e971f15061b14688983626915b03d11a : SUCCESS in 42m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3b53971da91d4257ae2a59bdaa6437fb : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c316e695ac4644f1a2330be7898f6ba5 : SUCCESS in 19m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c2a8bf5d68c4f01857eacd040bc3082 : SUCCESS in 14m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/79ab131631b44774b9bf2c0e42f637da : SUCCESS in 1h 38m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/924d87aeaf104a68adfed022d87df9fd : SUCCESS in 34m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a9daa8626b042bab92b4ce019880825 : SUCCESS in 27m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d8104d247e57478a9b7954a9d4e23bc6 : SUCCESS in 36m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9046f4b43f97497183ec13aceacf5d64 : SUCCESS in 14m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/899d9f68e2a642e4ad592279b8a2e2e1 : SUCCESS in 10m 46s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"3fd4ec28c71afe844c54626237fa04da1c9e7256","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-29 09:07:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc7b5fa4abfac7b0110a79282dcbea9d078f8ff7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-08-19 08:13:11.000000000","message":"Patch Set 1:\n\nLooks good, but do we want a release note?","accounts_in_message":[],"_revision_number":1},{"id":"f5aaae6405d2ce93c77deb5061b350c632c7a287","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-19 08:33:31.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Looks good, but do we want a release note?\n\nIt didn\u0027t seem important enough for a reno, but I can add one.","accounts_in_message":[],"_revision_number":1},{"id":"1e11e5b2c576671b540be5112258dd17cbdf8d18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-19 08:39:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b4cb4a50bf9999edb79896b9bbcbb1234e6f9b4b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-08-19 08:53:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cb25e85198c147d0a27c4c935e08999cad592a60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 10:40:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60aabba7cd9340aab35c16aea2a9d497 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d311e3cdb9c4195baa19d1030c64cf0 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/300a96970ef54403966d32f5932a571b : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fc84ce20a9348f090322ccad5b2998d : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04b99a6257914e0aac7ab22705d4ea2b : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf95a9d90f44db496584b7e251d2545 : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ed7b96a1b62437fb76ed4f30fd80af4 : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/07dfc965b3794e569788c88d4fae536e : SUCCESS in 8m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9fbb0619f28b4d0aaac23affd8bcf7ae : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/df312248ca3e434f926846ac794c683a : SUCCESS in 19m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2d846b31a8a640a2a9d318b92c884517 : FAILURE in 37m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44711b25d7264a7b85cc1512f0e13955 : SUCCESS in 47m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ceedb9005349479a81c11877018dffaa : SUCCESS in 1h 10m 37s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/183dc0eac0c247cabdd1d4c579c64454 : SUCCESS in 20m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/547eb6bb4f924bceb2e03f4ac5412b0e : SUCCESS in 14m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f0ed5b3aa7ce45398f2e529b71463381 : SUCCESS in 1h 54m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c46972c10f10403a8ac1c8e5e358afe4 : SUCCESS in 37m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a07dc34fdc584c31b921ae84863698ee : SUCCESS in 28m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2d83388c3a8f4df59089b566a1a017dc : SUCCESS in 25m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cd1efc088ba742fd88076ffcd212d7d5 : SUCCESS in 12m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52636122a764427aae2871d80b57e94c : SUCCESS in 11m 19s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"cc5bd710c979dc3b94f6362469a26ced687459be","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-19 13:21:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1bcf934d27d897a05463daf0996a5522d4477bc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:18:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23984f582a2b4c3a8c93252b42221dc6 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16de7917dc914bbe9287315edf170ab7 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/042954a64e684b5284c3b2d908d5d384 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f011dd1e939343888e99b0f5adc08f2e : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e9e2807a7d54ebc925d6f582b0e7b94 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4376e75063bd49ad9b9bba95d8d3e35f : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e32fedec3a04f9491021afb2068aca3 : SUCCESS in 5m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/efce1c4c18ce4faaa79cc827ff70bfc4 : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7603e3e8d3fb4dacb462f3cf506edfc9 : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/679cc4b021c049f2b0dc1b12515b50b9 : SUCCESS in 20m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d7f536021789442186eb4d543c4b12e2 : SUCCESS in 1h 00m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c412c331ef9c4594a28c9ac8dcf48379 : SUCCESS in 52m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/431431a10ab9494089e24bdd3ada8f72 : SUCCESS in 1h 03m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d8c1234d7f9743bea244ef4545f5343a : SUCCESS in 22m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dfee5c058284b9a9662d34e7af9d7bd : SUCCESS in 14m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3f68da89bea34a3496d16de80b2de54b : SUCCESS in 1h 55m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/539a5701176c4506a00fb81b1395b4d5 : SUCCESS in 37m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/429a440b3ede4d2784c047d6005cc9fc : SUCCESS in 29m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bb8f22e554954edfb09c1e0e310dadca : SUCCESS in 26m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6184fa55c1a04526b61b0b20841ff91c : SUCCESS in 13m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ada15ddc270446183ef40cb942d7159 : SUCCESS in 11m 45s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"b50f557edd1d69bb123e8ded5e75471a9aea115f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-23 08:50:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"49828141537c63e402c1ae6c7e138d6035d5cdd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 08:50:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c0f8e2498f55efff27b29a64aaba660690c169fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:41:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b34a3859fd0a4e9695fe737a1f3cb49c : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45bae3332c73477d8e0b5f8c06ba19a5 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06fb507d8a0345e0a100a82227b31f0e : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ee22cc985948eb95c0cceed56f19c0 : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74f98078fdaa4cc6b03473c69a4f4542 : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e948c91b831429ca9ce1ec6bca3c200 : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/36f11381c6d242cdace253027406e56d : SUCCESS in 11m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0cb3eb3a45b542a788bc46ac90a3327f : SUCCESS in 22m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2bc65491b67c458a9a256801a656ad5a : SUCCESS in 1h 04m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a50d987a1e2437088c551c03a6e530b : SUCCESS in 45m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/77be6c1686e54685ab9cf56a6fe5bbba : SUCCESS in 1h 10m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/463e43e62817468995eafe1b33397624 : SUCCESS in 19m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01e44ca123c94315b662fd423f744961 : SUCCESS in 16m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2f2aad5c11ac40839572f1c4711f6184 : SUCCESS in 1h 47m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5fb331b43a65463484e7533b7d4c653e : SUCCESS in 36m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f36cd961fb1349aaad7a566712386bbb : SUCCESS in 27m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3ee9f29515694b90bd5f3fa54d291e8e : SUCCESS in 30m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d2943e1f743a4939aad62e9542e18ef5 : SUCCESS in 14m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8562aa8526684f5a82efc04f2baf805a : SUCCESS in 11m 49s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"6dc267c3d1fce7cfd6f28a413dd8f8acedc5dc88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:41:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"48003a5dd538283fa34a41d5408902010c9cdca0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:42:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d4a24ffb5d24c56aa2b9b05f87e9e10 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a113c24132b24918b6982a54108e3eb9 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"91b2dddc415e1a5ca6aa66fdd65d7f97f5781cc3","revisions":{"4d7c3b99582feb5ed4535ce1d8ddc5d2c5f81e5d":{"kind":"REWORK","_number":1,"created":"2021-07-28 13:16:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/802709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/802709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/802709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/802709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/802709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/802709/1"}}},"commit":{"parents":[{"commit":"b8ff8fa811735d4a0610cd313e2583aabd92b34d","subject":"Add missing font for PDF generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8ff8fa811735d4a0610cd313e2583aabd92b34d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-28 13:15:52.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-28 13:15:52.000000000","tz":120},"subject":"Add twentyFiveGigE to port descriptions in inspector rules","message":"Add twentyFiveGigE to port descriptions in inspector rules\n\nChange-Id: I0c41d59123aba3351ac3b0a5a76cff517cca93e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d7c3b99582feb5ed4535ce1d8ddc5d2c5f81e5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d7c3b99582feb5ed4535ce1d8ddc5d2c5f81e5d"}]},"branch":"refs/heads/master"},"91b2dddc415e1a5ca6aa66fdd65d7f97f5781cc3":{"kind":"REWORK","_number":2,"created":"2021-08-19 08:39:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/802709/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/802709/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/802709/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/802709/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/802709/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/802709/2"}}},"commit":{"parents":[{"commit":"a48cc2498859f5949d304266d32ce3e620ad7f36","subject":"dnf: add metalink argument to custom dnf repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a48cc2498859f5949d304266d32ce3e620ad7f36"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-19 08:39:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-19 08:39:35.000000000","tz":120},"subject":"Add twentyFiveGigE to port descriptions in inspector rules","message":"Add twentyFiveGigE to port descriptions in inspector rules\n\nChange-Id: I0c41d59123aba3351ac3b0a5a76cff517cca93e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91b2dddc415e1a5ca6aa66fdd65d7f97f5781cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91b2dddc415e1a5ca6aa66fdd65d7f97f5781cc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
