)]}'
{"id":"openstack%2Fkayobe~802786","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Iddb699aa98ddee8446d532ce4bd23ba9174fb38e","project":"openstack/kayobe","branch":"stable/wallaby","topic":"dib-disable-kernel-cleanup-stable/wallaby","hashtags":[],"change_id":"Iddb699aa98ddee8446d532ce4bd23ba9174fb38e","subject":"Remove stale DIB_DISABLE_KERNEL_CLEANUP references","status":"MERGED","created":"2021-07-29 08:52:51.000000000","updated":"2021-07-30 17:11:25.000000000","submitted":"2021-07-30 17:10:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802786-dib-disable-kernel-cleanup-stable/wallaby","cherry_pick_of_change":799853,"cherry_pick_of_patch_set":2,"meta_rev_id":"78d73f47fbb1f568aafa23f5c638af3d06f946e6","_number":802786,"virtual_id_number":802786,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-30 17:10:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-07-30 10:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-30 14:59:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-30 14:59:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 10:52:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-30 10:12:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"82ce18a94a37972ebd9ab0baebb20d87d817942a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-29 08:52:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"296d5b8a814c7ce85a06931fc778dd7aa6c783bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:52:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8daaea61face4898b88f75d3b1370dca : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11b91a733b3e4190b144577d94e20b68 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1514a8e104345ec82b82dc0717fb076 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faae3fff3495458dbb078df7018b5f80 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ec2ff4bdf764224bf817effcf79ffd9 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ada3e57da90424d90fe5c0b94b04b84 : SUCCESS in 8m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7407596831fe472aa3cd25264fdee595 : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9d1845b1a5084c47bfdb1aca5b761b8b : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/462133588e744f1e8594a6d7f4f841dd : SUCCESS in 21m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b740bbca2d364a4790cdd32a9309b960 : SUCCESS in 1h 07m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c7147f201f1431f9e1164d970fcd7b6 : SUCCESS in 46m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cf652fc99bf24010a7699f5a9b343487 : SUCCESS in 1h 14m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6934b76644d847d0a25e431ce61689e0 : SUCCESS in 20m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9be3195dec21459a839e04772a3c645e : SUCCESS in 13m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9e593774e4a9430a980a33ef6d4ab52e : SUCCESS in 1h 53m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/67ec456c60fe41dca6aafc67f1d28bbf : SUCCESS in 33m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93077b9aaa7d4a0da2ee934692e06738 : FAILURE in 17m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba517c2d08df49829cc0d137d641419f : SUCCESS in 25m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8910794a5f4e4498a1f6d5d67901268b : SUCCESS in 16m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/584e2ddf436841d7b9384c5b11285b9e : SUCCESS in 9m 55s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"a26c1ec0396d5481102d996bc13f829d08a5a6c0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-29 11:00:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d509ec8fdc28dd91a7e2daf641e92e16624523a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:53:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/589dbb3daad347bb83838aa329193366 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7940f7e6181041298e14b79382fc03f1 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94692e9ccc944b2e9027960a4272bd40 : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea45b949b85f4911b46d3853aba6d80a : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab5ee191325447f3ad7ea95f35712675 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a06dc1d10f40f087c83899df8d9869 : SUCCESS in 8m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19fc140e4b994889b9651634c10c38a4 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/09e974ad829d4d8496bef646492f44bd : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b9833c14f1b4fdbb2078de2a262eecf : SUCCESS in 19m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ef7b6e85e880421784d8a8fa8c4d59dc : SUCCESS in 1h 05m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb51d2687d5d4050b892ae2618ad2942 : SUCCESS in 47m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/84a6d063dc55476bb8aa604f2ca59aba : SUCCESS in 1h 11m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d5b118b917194129807e807d5607dacb : SUCCESS in 21m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b4ed4fd17bf41d6a6511e96e1c8e83d : SUCCESS in 16m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c51fb852d84f4e06923ffcdb70f85c70 : SUCCESS in 1h 47m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5c5d890f159b414f905954c10cb1e4dd : SUCCESS in 35m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6ceb924eedc46239b375edc1b1813fa : SUCCESS in 27m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a1875c486aa848e1bd4ef2b1c67ef594 : SUCCESS in 28m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8641047f23e24685839abd4590d15610 : SUCCESS in 14m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91ce224433094c2b8bab2373ba5d7143 : SUCCESS in 10m 43s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"d7896f16551e0bc177ec31777287ce907f8dc13b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-30 10:12:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e27c705effe9f0020824cab02e26e97756b9edd3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-30 14:59:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a987fdf1d448cc024e5db2066e0acf1c58ed952","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 15:00:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7bdc6ae98b7561cae4ba1e48540f40ea3fbeee5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 17:10:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f910054306420b8bccda7889b0d460 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/602c879fc2424a7bb9007d24dfdba574 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03ce8a73479d4fe3a7bde51b5c5e27c9 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d77a4648f31f4dd2899d9bf3f9208dcf : SUCCESS in 8m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e896bd2e73ac4e2d9ba047bbddebef2f : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/647bc4398ea14f5a8bbd5a2c9edf98ad : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3dafca14f1874c92b4b9d295adde8bd5 : SUCCESS in 19m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5b4489a2be464ca6a71b21f18a669656 : SUCCESS in 1h 03m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecee9c45776e4c16b537576b278251cd : SUCCESS in 45m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0ad33b6dbc844adc99b259c327a38ab0 : SUCCESS in 1h 11m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f7fa173ba44046c9896a39d03fdf8849 : SUCCESS in 19m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca457dd42f0d4e1a8f6a5cd6ffcd3dd7 : SUCCESS in 12m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/57a16e281e48489d8a272b75b1633685 : SUCCESS in 1h 57m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/40835b1153e846e2a902eb7fd4748df7 : SUCCESS in 32m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6236e98a631491991775ec0fe4aed96 : SUCCESS in 32m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eb62c3a5c9344fe5aece19118e77eb17 : SUCCESS in 23m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ea9abd3529264fa3a407c386149394d3 : SUCCESS in 14m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdd3168099f34ffa8dcd3f042142a0ad : SUCCESS in 10m 02s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"b4411b368da7ddb114f20b07c4a7b87afbc07466","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 17:10:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"78d73f47fbb1f568aafa23f5c638af3d06f946e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 17:11:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f014d5872104ee29323c044de8b63db : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69d2b648f7c0b23adcaecdff1ecbb201a994d8e7","revisions":{"69d2b648f7c0b23adcaecdff1ecbb201a994d8e7":{"kind":"REWORK","_number":1,"created":"2021-07-29 08:52:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/802786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/802786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/802786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/802786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/802786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/802786/1"}}},"commit":{"parents":[{"commit":"b51df61880ff43443dfe250aab4367a64d3448ec","subject":"Merge \"Ubuntu: CI: rely on os_distribution, remove interfaces pause\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b51df61880ff43443dfe250aab4367a64d3448ec"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-07 15:28:26.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-29 08:52:51.000000000","tz":0},"subject":"Remove stale DIB_DISABLE_KERNEL_CLEANUP references","message":"Remove stale DIB_DISABLE_KERNEL_CLEANUP references\n\nDocumentation about DIB_DISABLE_KERNEL_CLEANUP was left in place after\nthe variable was removed in Ie25e88bb96f186399ae69837e93185127628ce86.\n\nBackport: wallaby, victoria, ussuri\n\nChange-Id: Iddb699aa98ddee8446d532ce4bd23ba9174fb38e\n(cherry picked from commit d4bbdada8209d3e1872c2beca9aee6fe62850ece)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69d2b648f7c0b23adcaecdff1ecbb201a994d8e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69d2b648f7c0b23adcaecdff1ecbb201a994d8e7"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
