)]}'
{"id":"openstack%2Fkayobe~802787","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Iddb699aa98ddee8446d532ce4bd23ba9174fb38e","project":"openstack/kayobe","branch":"stable/victoria","topic":"dib-disable-kernel-cleanup-stable/victoria","hashtags":[],"change_id":"Iddb699aa98ddee8446d532ce4bd23ba9174fb38e","subject":"Remove stale DIB_DISABLE_KERNEL_CLEANUP references","status":"MERGED","created":"2021-07-29 08:53:04.000000000","updated":"2021-07-30 17:11:34.000000000","submitted":"2021-07-30 17:10:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802787-dib-disable-kernel-cleanup-stable/victoria","cherry_pick_of_change":799853,"cherry_pick_of_patch_set":2,"meta_rev_id":"3c16ceb8c05ef90013ac8f3c2bc796a5b9977dbf","_number":802787,"virtual_id_number":802787,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-30 17:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-07-30 10:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-30 14:59:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-30 14:59:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 11:14:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-30 10:12:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"030a7deb5108dfbca51eae1bd58459d782b8cfdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-29 08:53:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8f8da8b8292a38a9ed7871f47993df51681fdcf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:14:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54edc01a53464e148d7b42ca90d87b99 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dfec12427c8422cb09eb60daf2a3e7c : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0e91f82e77241648533940c07f8e4ac : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bd44a59109347c88d670f770ce9b264 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c061aa8ba9094620821585408576ebef : SUCCESS in 7m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d346f89b04c147a783d539213c531192 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/509b620816ea4658a0b24b955fe1c661 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/62ce75609c8e4546acbcbf5343bc9471 : SUCCESS in 19m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1b10dec624e7411caf94172b843f6e34 : SUCCESS in 1h 07m 39s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2c85324b2c2a463cabf0cf7533164343 : SUCCESS in 1h 05m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bbdaf239dbc5422e8aab02dada4876cb : SUCCESS in 15m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2c061451f2db4f9c8c11f993e33ea4a7 : SUCCESS in 2h 16m 45s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4f6f2deac29f4977a2f062af71ef919b : SUCCESS in 26m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f791e9bbbbe04abca8701b8b03e8b97a : SUCCESS in 34m 10s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c3a9d46c52844a54a62b578f723cda59 : SUCCESS in 10m 52s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"e30f0662294742393533a6555c13d2515b1d23e5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-30 10:12:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c583fe932fffe6a35412feb07f10cbd430b15526","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-30 14:59:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41751ce96c60093d2c3ec4ec9cf30dcc6acf7ce5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 15:00:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e14a25b009c10f7f165b49dd37dcdc716dead80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 17:10:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e8e3e045574d0eaba71e8612575f59 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2699a4d5fc01468f8894c6ed4e15fb60 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe7258d719934817a0d04354f1fb72c3 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2131bc3bfc904fe7a319f8fcd8e02e99 : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/551301480c854cae95bb380313e5ea1a : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/61883118a4f147e980bb1c69ff4d5b76 : SUCCESS in 9m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16ff703d07ec4d43a3801d6ac9a405e3 : SUCCESS in 16m 33s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/50ee7cd964b540c18c3f3ab186353811 : SUCCESS in 57m 25s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3ecf6350d43341d0a9e5708f5f8b398a : SUCCESS in 1h 01m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/918c8c67842f4451a1eedcfb58784f34 : SUCCESS in 17m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f29d9adf2e99439d975428e7cc64621c : SUCCESS in 2h 04m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/70eaff0e7f5c403d930dafe85dee123d : SUCCESS in 26m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a2ede7f137024e038a328a1f5713c50d : SUCCESS in 30m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9883ad5b0827424ba31a37182d6c70ad : SUCCESS in 9m 49s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"fdcad347db8b22ca1469b7c8e997e4211008ac76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 17:10:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3c16ceb8c05ef90013ac8f3c2bc796a5b9977dbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 17:11:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de5d0aed2d654e12867bd0d9c8282df4 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aacb270935bccd39dd6b1ad749771b8841bbf4b2","revisions":{"aacb270935bccd39dd6b1ad749771b8841bbf4b2":{"kind":"REWORK","_number":1,"created":"2021-07-29 08:53:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/802787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/802787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/802787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/802787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/802787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/802787/1"}}},"commit":{"parents":[{"commit":"67c2e1220ca536ca549728821e8c2258dae71a49","subject":"Prevent matching unrelated images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67c2e1220ca536ca549728821e8c2258dae71a49"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-07 15:28:26.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-29 08:53:04.000000000","tz":0},"subject":"Remove stale DIB_DISABLE_KERNEL_CLEANUP references","message":"Remove stale DIB_DISABLE_KERNEL_CLEANUP references\n\nDocumentation about DIB_DISABLE_KERNEL_CLEANUP was left in place after\nthe variable was removed in Ie25e88bb96f186399ae69837e93185127628ce86.\n\nBackport: wallaby, victoria, ussuri\n\nChange-Id: Iddb699aa98ddee8446d532ce4bd23ba9174fb38e\n(cherry picked from commit d4bbdada8209d3e1872c2beca9aee6fe62850ece)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aacb270935bccd39dd6b1ad749771b8841bbf4b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aacb270935bccd39dd6b1ad749771b8841bbf4b2"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
