)]}'
{"id":"openstack%2Fkayobe~802788","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I424c1e3e6c488913d316e6b792d429599ea4de7c","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"I424c1e3e6c488913d316e6b792d429599ea4de7c","subject":"Use stream8 images with molecule","status":"MERGED","created":"2021-07-29 08:55:28.000000000","updated":"2021-07-30 16:44:40.000000000","submitted":"2021-07-30 16:42:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802788","cherry_pick_of_change":787763,"cherry_pick_of_patch_set":4,"meta_rev_id":"fa4757d0fefbcd63b8efd3e919d6fb547bef7934","_number":802788,"virtual_id_number":802788,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-30 16:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"all":[{"value":1,"date":"2021-07-29 16:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":2,"date":"2021-07-30 10:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-30 14:59:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-30 14:59:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 08:55:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2021-07-29 10:44:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-30 10:04:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"441e53d6df5eaa84002eaaa8964ff3cf9cb20e82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-29 08:55:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6fda46df02e445ad31a5bf9302a950e740396cca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:44:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f89dab67ab764d7d8e379537b61443f6 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5b932e2b60459c8882e2c8a123b63a : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f893a4b43b84a71be23ca733352f914 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bae69b6b6c041f380f00402f663a430 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24be8cab74ad4eb8bdfee76e9869390d : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3fb49b9f7e4a8e82aeb11bb0193891 : SUCCESS in 8m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d2512d6f9d8741f5a470a7144b9bc642 : SUCCESS in 7m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d31988a136c342338298a73098292f6c : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/56298af579e648bf893142f4c9eeab92 : SUCCESS in 19m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0f0c7fdaea4f49ae8bed8c5851250c8f : SUCCESS in 1h 00m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/748b1feebdbb4e0db807edf724223bac : SUCCESS in 49m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/94f945379ce542fa98cdae34bdff4a67 : FAILURE in 1h 02m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1ad5cd2db01b40baaec9aa06971aaf5d : SUCCESS in 19m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2770e1bb4ad940a78096829fe1a76e35 : SUCCESS in 13m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/592fccd4a8e9456f8dd3894feb0253fa : SUCCESS in 1h 39m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2c55ef0a462c47cdb9650d52a9cf5963 : SUCCESS in 33m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e8b8ae1135c4249a45d261a91c02d61 : SUCCESS in 31m 17s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c91924d31eb14396b973fd8f74262d67 : SUCCESS in 31m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b5cf78172d8342e3b5d1a89631c20c50 : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b282e6bbabc447d393ff22355aa3cf4c : SUCCESS in 11m 13s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"7171371bc39e8dc1de41227d380dac3371bcaec6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-29 10:59:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f1688fb01afdbe64817708e61cf9421483152eec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:54:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dca6f04792be48a4be43b054e1d68294 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a16928b606e84ec59821a7e5f86b8f8c : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f08bbee0cd845188243d5a7200235f1 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b108ce0d2f245c197dbaf8dbc9a9054 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2a44b3d054d49d181669b1d2bfbb682 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47f8664ef3664ca29cced82968946c73 : SUCCESS in 9m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e81e0114780c44f5a409861e181980a5 : SUCCESS in 7m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c07cc7abaf8478da45aba1741857923 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75b646129a554ae9b3164b9613869ee2 : SUCCESS in 20m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9a07b9fde5374ccf861bddc803c80625 : SUCCESS in 1h 01m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bb06eff2e8a496ab486aa5dd42bc495 : SUCCESS in 46m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4652657673744cd2a2ccf2cb20d6df4f : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7d63c7213c3640e388cd5da2a42edf64 : SUCCESS in 19m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e31cdc0410e64ca9acf84a7bbef746f4 : SUCCESS in 14m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1687b413be1042c38d597dfa81daa4a7 : SUCCESS in 1h 53m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a4f16b4fab524d4aa2d2eecf12f43538 : SUCCESS in 33m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a35f49b26884e029f1d224ffd16f4cd : SUCCESS in 28m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4aae31df918a4294903a6dde7757c187 : SUCCESS in 25m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/81a2cf29326640b09f791a1c379ea21c : SUCCESS in 13m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4da7b119b6bc4a6d85aab64f887e5803 : SUCCESS in 10m 04s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"b5bd49a8ac9c29dd8d99c0f0d225be5715d9c12f","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-07-29 16:29:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b0825b700c67d61d0861e5a9d363995a232da2ff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-30 10:04:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd88353568a7bf729a8d1caaa996432b394fd945","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-30 14:59:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c46eb56829cb20b0281b39a2dba0d77c590c987b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 15:00:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7fe5b0e8a957dc58f9b6437e50817952246c0e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 16:42:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48bccd077da44103963a7120e193add7 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a02ec04d63f4e9b96db6a57a5420458 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/394af9401cbe4b378a2ef7181545b7a8 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4333f7c94ad4cb4b6e54c9e6b5597e7 : SUCCESS in 9m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f1534141ffe848c1af5298af45083b44 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f9ab47808bf4060b289dc7f9ee77138 : SUCCESS in 12m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/141b3cf102b4494f8ede2d487e6eb4cd : SUCCESS in 19m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cb2d0982d0394eecbcce10dac192a8d1 : SUCCESS in 57m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37676f6e37e14542bd517006324fbeb6 : SUCCESS in 48m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/02705c5278914e549a901ec5fe59663a : SUCCESS in 1h 40m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f828733edae543b0b39b34819a6f0b18 : SUCCESS in 18m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6da6ecc9e28485ca935bbd2816cc8f2 : SUCCESS in 20m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/30cea67c9e504216b0231b3f72b7ae16 : SUCCESS in 1h 40m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/de865a06f5854452a8b1c23f81184591 : SUCCESS in 35m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f39752f791934e3d960a1bc432c4729a : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d6c2ce7b642c4649b498b2b5449f2852 : SUCCESS in 25m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/25ae488b4e96492c944949184fe00783 : SUCCESS in 14m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45fc9be9c366429e92912c6418730aa0 : SUCCESS in 9m 41s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"92ffb87b1087f0c66be59868ec3c789503bfbe89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 16:42:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa4757d0fefbcd63b8efd3e919d6fb547bef7934","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 16:44:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74cabe22470e41828dbb35f9505ec86e : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ca07b851ac9ca5d43a353a572e4cc553999ce98","revisions":{"9ca07b851ac9ca5d43a353a572e4cc553999ce98":{"kind":"REWORK","_number":1,"created":"2021-07-29 08:55:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/88/802788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/802788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/802788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/802788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/802788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/802788/1"}}},"commit":{"parents":[{"commit":"b51df61880ff43443dfe250aab4367a64d3448ec","subject":"Merge \"Ubuntu: CI: rely on os_distribution, remove interfaces pause\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b51df61880ff43443dfe250aab4367a64d3448ec"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-04-23 14:43:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-07-29 08:55:28.000000000","tz":0},"subject":"Use stream8 images with molecule","message":"Use stream8 images with molecule\n\nChange-Id: I424c1e3e6c488913d316e6b792d429599ea4de7c\n(cherry picked from commit cf60ba0d49b54abdbace9f878d85fc77cec315f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ca07b851ac9ca5d43a353a572e4cc553999ce98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ca07b851ac9ca5d43a353a572e4cc553999ce98"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
