)]}'
{"id":"openstack%2Fkayobe~802995","triplet_id":"openstack%2Fkayobe~master~Iecae3bf7d54ba0bfe8fbf68a3deae5f74f0b831b","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Iecae3bf7d54ba0bfe8fbf68a3deae5f74f0b831b","subject":"Add Arista switch support for the Neutron ML2 genericswitch driver","status":"MERGED","created":"2021-07-30 08:18:45.000000000","updated":"2021-08-11 02:23:57.000000000","submitted":"2021-08-11 02:22:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802995","meta_rev_id":"fa356ccfce1063ce8e18f33d3b977ddcbd611a02","_number":802995,"virtual_id_number":802995,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-11 02:22:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-10 20:45:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-10 20:45:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-30 08:19:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-07-30 08:19:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-07-30 08:19:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-07-30 08:21:57.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2021-07-30 10:12:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ac714f5d94c47bc163e566717ec808cbf05476c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2021-07-30 08:18:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d887d20de33596b5697254262ad42a35a7d9fb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 10:12:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce677e108e1d468fae6debbbe1506c65 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cae7c68c50ef45f7a478c6c4ec4989df : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/effa42f2766e4520ba29202b9dd5b903 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c037d7b512574dfd961683787f1ba35e : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14bcee4cdbd14e258777bf8c9734182e : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71050fb80f2048dd963b422d0772a621 : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae85c014732049458cb4088e25365c7c : SUCCESS in 5m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dcd42fe228da44c38c5b6f6ee3191c54 : SUCCESS in 7m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e4236612c1fe46c292f504ae15cbd52c : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fd2d964ea2834ca29f8bdb7f448fed1a : SUCCESS in 21m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/31815bdcb34a4b5c8de2fa91b88459ff : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b9ae20537a345fc923184cc5facbc51 : SUCCESS in 48m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/da6de095dfc946688cff12c2c807cf6a : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/60ed4af949a74d4393e4851bf95d7bde : SUCCESS in 19m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f82e6234c74440019a4c149583e62c13 : SUCCESS in 15m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2e423f8560f94f568dadf60e1338bb10 : SUCCESS in 1h 52m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d8520c94f9d140a7bfc249cd74d303a3 : SUCCESS in 38m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6d504bf1a1b44679565f54c39cbdf9b : SUCCESS in 27m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/457ebfd183534640b1c7c56fea3d2496 : SUCCESS in 27m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f4de9b2c22d14cbeb1000797f076af4c : SUCCESS in 12m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5e31b8513224991835bde61ba1a4115 : SUCCESS in 10m 50s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"8034f956470d87832948c4f6ab2914e1ed473e66","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-10 15:13:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"583fdf191b5cf1d996a74953a5a2965c39c411a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-10 20:44:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5adf3ce24721268b06fb1bc90695a1cf1acabf6f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-10 20:45:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"196411bacaba7a02e3a63b4e39ed5409f0cb74f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 00:10:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d154b9545224612b051f52c90d42f70 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6ae59b35a414d18a786814d4f158da5 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/298aa4e16f544adea54b08b6e48e7039 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6be1b613bfc8448ca9dd2e52f35d7e1e : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1738f14704074b9492d6a875cfe36b4f : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b8c63106c14d65bc8c6e3a4ebe2030 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a023865ab4144cf49199176e491db6b7 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7de7368b090b435ca970f2df59fc199d : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0063c2d12f6842ac9cb39b89346386ba : SUCCESS in 14m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5867d6a8c49d440ba3e186df3338e398 : SUCCESS in 18m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d104bea052fa45038574e6ddd48fac9f : SUCCESS in 1h 07m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e62462ffeb4d416f8c879632baf75d37 : SUCCESS in 47m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f5f7bdb1da2042a0b96b6fd7cb4158f7 : SUCCESS in 1h 11m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/69932c7cd6534462babcce728a78acdc : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ca3dcfdcd2c46e89932a5a5e3f1c471 : SUCCESS in 13m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/59860877415d439faa369912cc701aba : SUCCESS in 1h 49m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b9088fedf2dd4452891681ce007ed9d8 : SUCCESS in 31m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/946e5997810e4005b2d787f6c9c85561 : SUCCESS in 28m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0ec2091d69ce4f338b5f25bb3a7b630e : SUCCESS in 26m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5126b7680b6e4be39eecff3784caf80e : SUCCESS in 12m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c354cd205b004930869adf15d1a38e65 : SUCCESS in 9m 34s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"6adcf44f82dbe0625517612450b1a48e41a4a6f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 00:11:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb6441407e7ca0e2c2aa4e346c934ecf3e6cf217","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 02:22:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7964469307ed4fd6aff64ee5a66e4fdd : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edc41b50baf3463ca317036959ffe212 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20b167c7df2346558745d01e499cfca5 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c6d7ba63ce3411e96f78852ee3c0c98 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf81951ad7984d1eb4185178edeb33cc : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/adf0e0e336274d2fb791001682b90ecd : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bf0fbd944e8a41a4a5950607549d9b8d : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a03431d92ea947578c5d4b424ec0e552 : SUCCESS in 19m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4ded66632c814d2286a246a6e64d35ae : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7eeace75f3f4f308bf79d6f57384443 : SUCCESS in 46m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ad00fac8618645e9a62a9faf75abdec5 : SUCCESS in 1h 04m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/12e17f34cd7744de83557008ecc9adfe : SUCCESS in 17m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dad776fa14134624b67d363fa3235273 : SUCCESS in 12m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b40af027bc1240e7ba810f27a70bef60 : SUCCESS in 2h 06m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c2b9fba3c0564ae18b9067d0d6393d78 : SUCCESS in 35m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29a14cb180d14193995ce73aad176d42 : SUCCESS in 27m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/51373f41ae9e4bde807b956ad59a5e0f : SUCCESS in 27m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/387d8bf572e54c3cb4f60dd3518d6791 : SUCCESS in 12m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ef06871a885458aa1e98dcd70c3ed57 : SUCCESS in 10m 54s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"866e1af1f1b568c9486a392d3979ba124e21c0f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 02:22:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fa356ccfce1063ce8e18f33d3b977ddcbd611a02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 02:23:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f234675e8f41529461b06f35814925 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2c77e949a7846dbaf7e560df51fba73 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a55c83465cd0b3d91bc357bf9eb93bf149f1527","revisions":{"de712c6471a85f3d6fc8a72fb1b3ee5822cdb62c":{"kind":"REWORK","_number":1,"created":"2021-07-30 08:18:45.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/95/802995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/802995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/802995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/802995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/802995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/802995/1"}}},"commit":{"parents":[{"commit":"da304da58eb184546f10083990b08005bafcd392","subject":"Merge \"Fix capitalisation of Ansible task name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da304da58eb184546f10083990b08005bafcd392"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2021-07-30 08:18:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2021-07-30 08:18:33.000000000","tz":120},"subject":"Add arista support for Neutron ML2 genericswitch driver","message":"Add arista support for Neutron ML2 genericswitch driver\n\nChange-Id: Iecae3bf7d54ba0bfe8fbf68a3deae5f74f0b831b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de712c6471a85f3d6fc8a72fb1b3ee5822cdb62c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de712c6471a85f3d6fc8a72fb1b3ee5822cdb62c"}]},"branch":"refs/heads/master"},"0a55c83465cd0b3d91bc357bf9eb93bf149f1527":{"kind":"REWORK","_number":2,"created":"2021-08-10 20:44:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/95/802995/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/802995/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/802995/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/802995/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/802995/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/802995/2"}}},"commit":{"parents":[{"commit":"769ec55de55808f4e378de40bc5423fd4d33520f","subject":"Remove references to resolv.conf workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/769ec55de55808f4e378de40bc5423fd4d33520f"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2021-07-30 08:18:33.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-10 20:43:49.000000000","tz":120},"subject":"Add Arista switch support for the Neutron ML2 genericswitch driver","message":"Add Arista switch support for the Neutron ML2 genericswitch driver\n\nChange-Id: Iecae3bf7d54ba0bfe8fbf68a3deae5f74f0b831b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a55c83465cd0b3d91bc357bf9eb93bf149f1527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a55c83465cd0b3d91bc357bf9eb93bf149f1527"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
