)]}'
{"id":"openstack%2Fkayobe~803064","triplet_id":"openstack%2Fkayobe~master~I0748f7bd3e6f7f91e35b2178acb6973e99710f88","project":"openstack/kayobe","branch":"master","topic":"merge-configs","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-09-01 18:12:23.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-01 18:12:23.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-18 08:32:32.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-05-31 08:09:15.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I0748f7bd3e6f7f91e35b2178acb6973e99710f88","subject":"Add cached plugin","status":"MERGED","created":"2021-07-30 18:06:49.000000000","updated":"2023-09-01 18:12:23.000000000","submitted":"2023-09-01 18:12:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"803064-merge-configs","meta_rev_id":"776136c56fc1e070888185931612a0b93e1129f5","_number":803064,"virtual_id_number":803064,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 18:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-30 19:18:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-30 19:18:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-05-30 19:18:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-05-30 19:18:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-30 20:02:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 09:31:30.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-12-12 16:18:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-18 08:15:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-19 12:58:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"658ae575e1daa0f81dc374e0e67db2d8cd7db6e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-07-30 18:06:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72dbd5868d666591f72ff1e578fda95825ca2251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 20:02:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cdbdd9efcfc4801a822b1bd9523f875 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e31d03765a4cd09c7a23a16412c751 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a88b67d602bb410180fd16177f506a23 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a4218c308684003b41c0f91493fc529 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59974736dc82450c9e2d4e045d5fd0c2 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d439652c9c3a40c49e60adcea3f8988b : SUCCESS in 10m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/abb20eba3f0849759334bfed32d44d18 : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d41620e927104c479e8d55569fa62cd3 : SUCCESS in 13m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6a0d239dc3064376a8d36ea4a786c46e : SUCCESS in 12m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fcebca0841c24d23bf07e9a10c9316ad : SUCCESS in 1h 03m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f19aad378aaa47239be4881ff5b7a07b : SUCCESS in 46m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/94234850bdf94b8784b10b3d70934dd8 : SUCCESS in 1h 09m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/15a74e6d701b4db48b58e7bc2ff99365 : SUCCESS in 20m 00s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/063137bdc1c5487d80474f00bdc1166c : SUCCESS in 14m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5eabfae1b0804aebbcbf59abe54870ff : SUCCESS in 1h 50m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cb9705c699604ea0ace32ac9bc86532c : SUCCESS in 36m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3cfe8b221354fdcb0d1f4919307e215 : SUCCESS in 28m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/617e03fa6f9a42eea5f1bb7361f2711f : SUCCESS in 26m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bebb43fc3f16419199254e96ef118c59 : SUCCESS in 13m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6de423a449b2468381e557df32f298c0 : SUCCESS in 11m 30s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-molecule.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"d125cf72591bb8e26a4e73fcb60a5168ca11070b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-11 09:31:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf3f05e662db01e8db6c60bb04462bfe91862908","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-10-07 15:16:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"73746072385d1321caaaacd3d14b28cf6ad641b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 16:46:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e00718f5e9d2433b8a17957ebe056522\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec37ec8a831e4ec3b27e73a0baac92a3 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/776e32f432924e4cbb107b37d5f09a17 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27ab2db34dd5407091e20629403d7153 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15f2f3d237d34f5f86f29dbd101c73e2 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7be21b411e674487af751599084ef95f : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e220244f9096432dbbd9d955d142e70f : SUCCESS in 9m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cd5243493a8d4aeca2270e8a498ac1c7 : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dfe1b0a6a50544ec94564532805e3aa0 : SUCCESS in 14m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e54adc8671d94ee39fac030ccf31e9e7 : SUCCESS in 13m 49s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39527e5f98ca4252a9bec58965350b71 : SUCCESS in 59m 18s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/165d2bbe0fb54b07adcfd8355db1bc8b : SUCCESS in 17m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3dc5b1fd4914376bae86a0b8ce40d1b : SUCCESS in 1h 12m 25s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4011b33f7cfc4dbc8fc6ce745b594fe7 : SUCCESS in 18m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51d38246b3d84fb39cbacdb29e17d1ee : SUCCESS in 32m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5dabce3a6cf40cba00899579c18c291 : SUCCESS in 12m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bb84f15de3e44a883a41f0f9c704612 : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8636f354b259cbd503a9697b5d761ef3cd429875","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-03-07 10:22:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"23dd4fae0a91b2285ddea53df8c6504f9333cd55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-03-07 10:29:27.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"284d1d2e25d7621c4f91d3d61050ac2bf6ccce3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 12:04:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fa8da5464c4407a9f310c0ab4484e3a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a188edd3487e461baf4ee3da2861a879 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee07579dd5684cb7bbc3f3fc7e482664 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4a4c51aadfb4258b2661ee7089c0ef5 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e7cdac24ea7424ba9a423479cce060c : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/322a7cc739ff413c850b5b8568c5bf19 : SUCCESS in 9m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ddd699b1f0b54a04a68f1d42e09cefcf : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e57ea0dc09e48fbac64296aa8e50735 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ffa7f7344273426ab04b8e346d28910d : SUCCESS in 10m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3a253fbc5eaa43c5a67bdc6261109d81 : SUCCESS in 56m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c175322a89004b85ae4492bca508a3c0 : SUCCESS in 59m 58s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd08794365ef433099582c9657b840eb : SUCCESS in 50m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ab6472484c4049b68428c4b92dd323f6 : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/451c6cd90933425a9b41b1e433560124 : SUCCESS in 16m 59s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/872df1b7c8504647962e473e770f174c : SUCCESS in 18m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b96e378bfc2b482fa8f52d3c1d3c329a : SUCCESS in 15m 57s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0088ba76240542abb093399006d3ca09 : SUCCESS in 1h 29m 10s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29fa167fdf5a4e7e9d5ab2381ae4b2bd : SUCCESS in 1h 15m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/cb20a25cc79047edbe31f4d93078c2d0 : SUCCESS in 15m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3638dbb09b754021b73543d8658a63f0 : SUCCESS in 14m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b7fa80c906f34cea93da49e79dc7f214 : SUCCESS in 28m 18s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a2d93eda1874e1aabb7035098f8b372 : SUCCESS in 39m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea80d32b85b648dbbbdb3be5761eddeb : SUCCESS in 23m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/99302fd8504a447e9238f1b869be24bd : SUCCESS in 11m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0617f83072794d3e9b7cc80896d341dd : SUCCESS in 10m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2ed006b3eb794671b7a3f1ce0d172edb : SUCCESS in 18m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/643fec5855e14d10a2d80f46f972481a : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":4},{"id":"3c5a8fa6be2f49783a00c930783231eb29dda952","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-18 08:15:31.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cf90ffcd7b1c0861b582754260635f69c0d793a0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-18 08:32:32.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"126c90988731c61bab5b76209ce5737c0815f795","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-19 12:58:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9a685e57e00ff23280f49ea510c1ac887220512a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 14:07:43.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ea84cba6e48969ac91e20c47755ba2bd5a4b9620","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 14:09:33.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"81b3aca13f39726a4792029608fa8727da74cee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 15:40:15.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"94da5060c6b6136fd37dbec8e5ee34fb2edf2510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 17:38:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d94c017dd74b6a8c0c7c30cbc2facd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37c60ebbb1a34684a1728b4af423e15b : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5130262559e40bdab3d2751130e5bc1 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f279dbe168f43f1a22da4b5f16ffadd : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a015daab0ef49268fa7f3f89ba6e214 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f82ba1a41eba443c86f00d19e3d32f5b : SUCCESS in 5m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/418610a29c2b46c8a37a29c85b1f39e9 : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb954a0a90524e188e0f1fcfc5c7c83b : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/64c175c77da44e95a972b0fdaa02a53e : SUCCESS in 11m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5837564b34734169a643ad9a9918f567 : SUCCESS in 52m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3bc7ba86b4044f11bbd170ef19e1f3dc : SUCCESS in 55m 09s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b61adb8270264b09a2aa7e2c05161a55 : SUCCESS in 43m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2789542d56014cfebecdfc2619299a14 : SUCCESS in 50m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/076dabdcb9fe472abe919cc48e10194b : SUCCESS in 15m 39s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f79f49353076447989f975868f0adcf6 : SUCCESS in 19m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4d91034d2294c9a8fd6d51b889c31bf : SUCCESS in 12m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/70a397d3e21845f49a7e9f7a009491c6 : SUCCESS in 1h 52m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4aad71b4fff841f293166d9d385630a9 : SUCCESS in 1h 06m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3dd0bcc448c04f73ac39e477ffec7fd5 : SUCCESS in 15m 34s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd44866d89c949509217caf66871042b : SUCCESS in 13m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7032ac8525da45a59be609718900bd6f : SUCCESS in 29m 17s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/80603ba0d4f84855bce4e39fe3784de5 : SUCCESS in 30m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b27422cd2a844479856dce91f9cec86a : SUCCESS in 25m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d1fda29409564d05a410b34c1a8a292c : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b126abba8e34c7e9606fa9d83f6d7b0 : SUCCESS in 10m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2e5336b34d284abcaedc8e7d66fba05d : SUCCESS in 14m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75eab1c381844a728ab7409488b7d031 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":7},{"id":"ebbaa129dd4678841cf5179b79f1fd6a620eb065","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-20 09:19:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d221b67173dc2c2e9fcd3193829d659140487c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 10:55:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79ca5288feae4bca954ee011d0f8c9a2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80b7a5fe89084073a39aac624626ee77 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f671dac21b884249b051c403f5c57c68 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53d2e68f70ec4db194bc0ed40929f979 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d87bc5769c945b68783329a5b89332b : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60dc405bdb184d81afea781b1dd9780d : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/65490ad5180d4682a822b0ff73b2e4fc : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4134161a27ef4b26888779f1c70aa838 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4fbf9e7ca22c44e784d6d845ed346289 : SUCCESS in 11m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b61d403d552342139b6e8b2501ebc781 : SUCCESS in 1h 16m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/30505c5372c84e09910a3bdd400fcaa2 : SUCCESS in 50m 39s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d40a7b2331941b88b30aa99b55359dd : SUCCESS in 44m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c7b8347b02e44e0391d2c959acc5c6d2 : SUCCESS in 1h 02m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9596f22db2d245c2ab9b5a87bb3e718a : SUCCESS in 16m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/14bf12d06bea417096eb9442aa8f1cb5 : SUCCESS in 16m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1109163021948018782e3e7df8c975f : SUCCESS in 11m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/403e02dbe08345f4b61f8e1945f0e0ae : SUCCESS in 1h 24m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0d74adb62c74f2f8cbe7c9f159c40d6 : SUCCESS in 1h 07m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ecd5af4e33224476b3b49981e8513879 : SUCCESS in 16m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2ef5550365e47ce811524c999c2f172 : SUCCESS in 13m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0556cbe468114c2cbb553ad85f6a1fee : FAILURE in 30m 17s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d2a10e31e45d4f66b40662160cd5e6e1 : SUCCESS in 26m 34s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55ffb45bb09945a2a8cf89752f796b24 : SUCCESS in 23m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bad82d9c615a4e70b13ac8598db05ce1 : SUCCESS in 15m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/923a003077294871ba464e6a3ee444ac : SUCCESS in 10m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/79050a889b0b44919bf6439ee5c1d5ab : SUCCESS in 26m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c9e5031657d41308bfac0a5923a2408 : SUCCESS in 13m 08s","accounts_in_message":[],"_revision_number":8},{"id":"a2890379a154a11c5c1cf4d034d76aa62a19af88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-30 19:18:26.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"35b1db169d1e437c1b5e0d3687aabbb3cdcc1dab","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-30 19:18:32.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"1ee048e3d11c568c4feca89da2d0fe39a95095dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 20:49:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b30808b85fe242209a64484a5d79960e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1220f55376b4bc4a51aee588ab6552a : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd3ca8defa3d403da45113826df0978f : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7f51b3875ec428499e0da543a7fba39 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccc9e1ac8bac4b6696fbea607428074f : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e9585080a514caa9c79e9bfd33bcc84 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf58508d98634dd8b50188dfa682e47a : SUCCESS in 7m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae2e79c329d643ee8fed2a21a722aac8 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/75c320b7883a40e48d002ae6f3b58d34 : SUCCESS in 9m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/20c8f1c937ef414c8373e93ed365a066 : SUCCESS in 10m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/72af2cc3c50f48239bb9128bd01a6629 : SUCCESS in 1h 05m 34s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/487b366f7ba14b82862c3078ffbb7afe : SUCCESS in 53m 55s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae4d34c4d5d845dc8c69b59760793a00 : SUCCESS in 52m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/31b82b427d9547c1a607e5b35f40c97d : SUCCESS in 55m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/295c431d7825425dbed86ff6c04e8dfb : SUCCESS in 16m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4af972efe58e461f898aaacec2461140 : SUCCESS in 17m 19s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23562aa75d204e0d84b776a006536a3b : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7b962f9a990349ebb1555e4084b76c5d : FAILURE in 21m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5495a2f0a4b64df3bf90ba7ccb745a65 : SUCCESS in 1h 27m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e243dbae2c1843979bdab42bf6bb61e5 : SUCCESS in 18m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09a3e8f53c7d4981a51abb297da9dc97 : SUCCESS in 15m 47s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1858396f24a64ce792b672e2e16c85ef : FAILURE in 35m 54s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1d98612f68bc44d790fd33dd5012647b : SUCCESS in 30m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e0520c705114b76a8c784bd816aa469 : SUCCESS in 25m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bb282b3927af461c93d745c53856de60 : SUCCESS in 13m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/550b1bc0794b4151a202db8a353e02b0 : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/52a83016fb3943838ea063b5228a45a7 : SUCCESS in 21m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/411bf174b7314c51aa59827f9b3500fe : SUCCESS in 20m 00s","accounts_in_message":[],"_revision_number":9},{"id":"4a722d4314d8e0cc28500e91e7c906a3dfe68b7e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-31 08:09:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"589b36b6f33124ac0a4e089b8944a3db7c015b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 10:01:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ff3dca095f848a497bf6a9b9a564c8e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96e79a3b1bb74faca632b9b9df08d985 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fecb26e69ef4fd0a480e3f5b79ae119 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e43f101d5e6458a99a8cbc62a7909c5 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76fce8c308a24737afa2aafdf265d812 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2548e00b0083493e82e7df5bf5ff7dfb : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a716889f7c419cb02f1903f45b2790 : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dd41022714314681b195fed03e24767b : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/349d099e9b274bef99e88b6728880b60 : SUCCESS in 14m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c5c57248cc847628cf536049de5ab40 : SUCCESS in 13m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7a87e0b583264f7486ffeae15e22a6a3 : SUCCESS in 59m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a179246f1a534119b1bbe2a98d36a1c6 : SUCCESS in 1h 02m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59ec160aec3d4c8daeb3f9080b8fd635 : SUCCESS in 57m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/863e8584667f4f32813d73902707829e : SUCCESS in 1h 11m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4d46014780404232b633cdc28c684f7e : SUCCESS in 20m 37s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3101905dcd8649cdaf5a3ba1c84a3cb3 : SUCCESS in 20m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/803b8a35a3944d5ca9e36ff3a52bf2e4 : SUCCESS in 16m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ba29bbd80e0844909b8258be3f30a627 : SUCCESS in 1h 38m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c9361352c9a4c15baf998f96e12054b : SUCCESS in 1h 43m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/81b4b430c0b041bbac1b6a071ebb7c92 : SUCCESS in 21m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e86fc5c279eb4cc8a8153b6451d819e9 : SUCCESS in 16m 18s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/187fbaf6287b422682a2edafaa8b0015 : FAILURE in 34m 48s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/18a96927ace74d3ba8848221d4435e18 : SUCCESS in 28m 01s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ac60839395c4c0fb826a93198aabd24 : SUCCESS in 25m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e8e030a22dad4c4e97942bdb1635bf5c : SUCCESS in 15m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbd10ea0ebde4222851029e166c7d32e : SUCCESS in 12m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2f93e2425dc34639b4061685318e6662 : SUCCESS in 34m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2476f6e0d014c1686cebafee786b436 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":9},{"id":"f493017c185c6db73f6ca459ba4e2da8a685ed5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 16:21:11.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"418804ad23aa5aca9585e80aa15e432eb6644101","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 18:12:23.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3456d6af86da4219b000ef885cb4f08b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8336c2af3c064c0fa761a2be91313be8 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4507b6910a2942a0836369c669e63407 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ffd24588ec841a1a0b6b08010d8d96e : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73da61f2d9924fbe9803a8aa38c62530 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836e8e178ca24e1488158695ec7fb6df : SUCCESS in 8m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f985860a3a6749cb86743e25bfcd49a9 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/af8b4a13aad1479f96eb8c4fb6996517 : SUCCESS in 12m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/30803538e0164240a77cb443ed99e097 : SUCCESS in 10m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bef491bf83b34a7792ff7565fb9f3c5b : SUCCESS in 46m 50s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64e19886a8b5446aa89260264bf6b159 : SUCCESS in 52m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9fb18ecfbff44f09814a0ebea36c229d : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/60a9808c8e494c02ac9515332c1519f3 : SUCCESS in 17m 08s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/119ffbbe76fd42f9b2ce9d0b79901a7f : SUCCESS in 15m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6f252062f86c4ac4ac178fb5d3556a46 : SUCCESS in 1h 26m 11s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f18e671a30940bc923e2d1ef40a8c8d : SUCCESS in 1h 37m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9f2a4bd5bf4e4082aa8cbb7e21f6c220 : SUCCESS in 17m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12c27e563f1148ac97676f0d27ce3b0a : SUCCESS in 13m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d380ebc398a5431abba6791cb4184e07 : SUCCESS in 31m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69533bf5554b4f2192e6dad2c740f637 : SUCCESS in 28m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ce66aa6a66d2403cb6ec690a1b8cadfc : SUCCESS in 9m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a00e990714e4cbebdb9761d2535433a : SUCCESS in 11m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2e877167d3f740c183ab79656187dae6 : SUCCESS in 31m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ce0d4a3cb2d4c5c9b11ffc2913c1b2a : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":9},{"id":"776136c56fc1e070888185931612a0b93e1129f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 18:12:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"a46b3ebc8e5ffbc4f175fa23cd103d595ea1ffed","revisions":{"6405bb3468176460f01cbf910c603eb0bb893b20":{"kind":"REWORK","_number":1,"created":"2021-07-30 18:06:49.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/64/803064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/803064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/803064/1"}}},"commit":{"parents":[{"commit":"aa516b159658650bc4bca76341b63123e8cfcc45","subject":"Support merging kolla group_vars when using multiple inventories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa516b159658650bc4bca76341b63123e8cfcc45"}]}],"author":{"name":"Kayobe Automation","email":"kayobe-automation@example.org","date":"2021-07-05 16:47:52.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-30 18:06:00.000000000","tz":0},"subject":"Add cached plugin","message":"Add cached plugin\n\nChange-Id: I0748f7bd3e6f7f91e35b2178acb6973e99710f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6405bb3468176460f01cbf910c603eb0bb893b20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6405bb3468176460f01cbf910c603eb0bb893b20"}]},"branch":"refs/heads/master"},"0cecd10197978b2525ca3dd69272331cc886f32e":{"kind":"REWORK","_number":2,"created":"2022-10-07 15:16:42.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/64/803064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/803064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/803064/2"}}},"commit":{"parents":[{"commit":"17bb7086f8c1e2e5960bcebd7fb0070ca901a433","subject":"Pass through kolla inventories as is","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17bb7086f8c1e2e5960bcebd7fb0070ca901a433"}]}],"author":{"name":"Kayobe Automation","email":"kayobe-automation@example.org","date":"2021-07-05 16:47:52.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-07 09:40:37.000000000","tz":60},"subject":"Add cached plugin","message":"Add cached plugin\n\nThis is useful to cache expensive lookups.\n\nChange-Id: I0748f7bd3e6f7f91e35b2178acb6973e99710f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cecd10197978b2525ca3dd69272331cc886f32e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cecd10197978b2525ca3dd69272331cc886f32e"}]},"branch":"refs/heads/master"},"3889e13bf14542aa2806f67ce2e13e72dfff2cfe":{"kind":"REWORK","_number":3,"created":"2023-03-07 10:22:48.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/64/803064/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/803064/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/803064/3"}}},"commit":{"parents":[{"commit":"09ecd1370f3df86a3b1559f2567b5c9bdd0359ee","subject":"Pass through kolla inventories as is","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09ecd1370f3df86a3b1559f2567b5c9bdd0359ee"}]}],"author":{"name":"Kayobe Automation","email":"kayobe-automation@example.org","date":"2021-07-05 16:47:52.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-03-07 10:03:53.000000000","tz":0},"subject":"Add cached plugin","message":"Add cached plugin\n\nThis is useful to cache expensive lookups.\n\nChange-Id: I0748f7bd3e6f7f91e35b2178acb6973e99710f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3889e13bf14542aa2806f67ce2e13e72dfff2cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3889e13bf14542aa2806f67ce2e13e72dfff2cfe"}]},"branch":"refs/heads/master"},"6de6ca59b62a9457d71f86b88c8648c157abced0":{"kind":"NO_CHANGE","_number":4,"created":"2023-03-07 10:29:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/64/803064/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/803064/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/803064/4"}}},"commit":{"parents":[{"commit":"3d7d5d2551496862564161670bc841e19890e867","subject":"Use merge_configs and merge_yaml to generate Kolla custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d7d5d2551496862564161670bc841e19890e867"}]}],"author":{"name":"Kayobe Automation","email":"kayobe-automation@example.org","date":"2021-07-05 16:47:52.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-03-07 10:29:18.000000000","tz":0},"subject":"Add cached plugin","message":"Add cached plugin\n\nThis is useful to cache expensive lookups.\n\nChange-Id: I0748f7bd3e6f7f91e35b2178acb6973e99710f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6de6ca59b62a9457d71f86b88c8648c157abced0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6de6ca59b62a9457d71f86b88c8648c157abced0"}]},"branch":"refs/heads/master"},"b8cb10e17970fb4d1a54bd4fb332210efb7990e7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-04-19 14:07:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/803064/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/803064/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/803064/5"}}},"commit":{"parents":[{"commit":"9c1564232eebc6bdf72a2f6778076393165a2a60","subject":"Use merge_configs and merge_yaml to generate Kolla custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c1564232eebc6bdf72a2f6778076393165a2a60"}]}],"author":{"name":"Kayobe Automation","email":"kayobe-automation@example.org","date":"2021-07-05 16:47:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-19 14:06:46.000000000","tz":60},"subject":"Add cached plugin","message":"Add cached plugin\n\nThis is useful to cache expensive lookups.\n\nChange-Id: I0748f7bd3e6f7f91e35b2178acb6973e99710f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8cb10e17970fb4d1a54bd4fb332210efb7990e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8cb10e17970fb4d1a54bd4fb332210efb7990e7"}]},"branch":"refs/heads/master"},"f94c5901a61beee85cc9d5408e165107164487cd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-04-19 14:09:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/803064/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/803064/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/803064/6"}}},"commit":{"parents":[{"commit":"373edbecb1d2603f6e25176fc0f9b47c34cd511e","subject":"Use merge_configs and merge_yaml to generate Kolla custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/373edbecb1d2603f6e25176fc0f9b47c34cd511e"}]}],"author":{"name":"Kayobe Automation","email":"kayobe-automation@example.org","date":"2021-07-05 16:47:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-19 14:09:11.000000000","tz":60},"subject":"Add cached plugin","message":"Add cached plugin\n\nThis is useful to cache expensive lookups.\n\nChange-Id: I0748f7bd3e6f7f91e35b2178acb6973e99710f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f94c5901a61beee85cc9d5408e165107164487cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f94c5901a61beee85cc9d5408e165107164487cd"}]},"branch":"refs/heads/master"},"406b40869b582e4e87e314511b4ffc11ed6b6af0":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-04-19 15:40:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/803064/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/803064/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/803064/7"}}},"commit":{"parents":[{"commit":"2ea190f2b52fbab998149d26082102fc3f99aa32","subject":"Use merge_configs and merge_yaml to generate Kolla custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ea190f2b52fbab998149d26082102fc3f99aa32"}]}],"author":{"name":"Kayobe Automation","email":"kayobe-automation@example.org","date":"2021-07-05 16:47:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-19 15:09:26.000000000","tz":60},"subject":"Add cached plugin","message":"Add cached plugin\n\nThis is useful to cache expensive lookups.\n\nChange-Id: I0748f7bd3e6f7f91e35b2178acb6973e99710f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/406b40869b582e4e87e314511b4ffc11ed6b6af0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/406b40869b582e4e87e314511b4ffc11ed6b6af0"}]},"branch":"refs/heads/master"},"3c49d0a413e5ff711bbfdf0f651190101dcedc5e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-04-20 09:19:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/803064/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/803064/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/803064/8"}}},"commit":{"parents":[{"commit":"9bfd19293b2686cfd94448ee72d37ee34e348592","subject":"Use merge_configs and merge_yaml to generate Kolla custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9bfd19293b2686cfd94448ee72d37ee34e348592"}]}],"author":{"name":"Kayobe Automation","email":"kayobe-automation@example.org","date":"2021-07-05 16:47:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-20 09:19:35.000000000","tz":60},"subject":"Add cached plugin","message":"Add cached plugin\n\nThis is useful to cache expensive lookups.\n\nChange-Id: I0748f7bd3e6f7f91e35b2178acb6973e99710f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c49d0a413e5ff711bbfdf0f651190101dcedc5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c49d0a413e5ff711bbfdf0f651190101dcedc5e"}]},"branch":"refs/heads/master"},"a46b3ebc8e5ffbc4f175fa23cd103d595ea1ffed":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-05-30 19:18:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/803064/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/803064/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/803064/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/803064/9"}}},"commit":{"parents":[{"commit":"428ef10fa4d4a281f0c690876fc3cbfff41c2bbb","subject":"Use merge_configs and merge_yaml to generate Kolla custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/428ef10fa4d4a281f0c690876fc3cbfff41c2bbb"}]}],"author":{"name":"Kayobe Automation","email":"kayobe-automation@example.org","date":"2021-07-05 16:47:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-30 19:18:26.000000000","tz":0},"subject":"Add cached plugin","message":"Add cached plugin\n\nThis is useful to cache expensive lookups.\n\nChange-Id: I0748f7bd3e6f7f91e35b2178acb6973e99710f88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a46b3ebc8e5ffbc4f175fa23cd103d595ea1ffed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a46b3ebc8e5ffbc4f175fa23cd103d595ea1ffed"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
