)]}'
{"id":"openstack%2Fkayobe~803283","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I082e1b807497d17ae40ddf86e1473fbfcdf625e2","project":"openstack/kayobe","branch":"stable/wallaby","topic":"bug/2009072-stable/wallaby","hashtags":[],"change_id":"I082e1b807497d17ae40ddf86e1473fbfcdf625e2","subject":"Ubuntu: Set MTU for veth in the network files","status":"MERGED","created":"2021-08-03 12:47:32.000000000","updated":"2021-08-03 18:08:35.000000000","submitted":"2021-08-03 18:07:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803283-bug/2009072-stable/wallaby","cherry_pick_of_change":802218,"cherry_pick_of_patch_set":2,"meta_rev_id":"ba03c9c235000bff46167cf037f64c8e90829fa1","_number":803283,"virtual_id_number":803283,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-03 18:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-08-03 15:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-03 16:01:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-03 16:01:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33466,"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","username":"lzalewski"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33466,"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","username":"lzalewski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-03 12:47:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":33466,"name":"Lukasz Zalewski","email":"lzal3wski@gmail.com","username":"lzalewski"},"state":"REVIEWER"},{"updated":"2021-08-03 14:32:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-03 15:56:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"2d298708800cc295c63ddfb27ebcbd59f9ec5474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-03 12:47:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"88284fa6e040c957c7c1405b88a073b4f79833a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 14:32:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db0be1d7d42e4b11b418de71cff966fa : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b372719de2d447e38ae6c689758d1320 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a04af1a32ddf4790bb763f3819bbba7a : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0595750b1ca4ff99d12383bac508189 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b096be79a7a437eb551b2928b8b50c5 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57fe5ed6b46d4229b2cd3ee09eb7cbcc : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6af8ad17e814b3aa2a2a1737ea6010d : SUCCESS in 5m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ed5ab153e897401694f2a3b225cfedd8 : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d871ed91c2714e5bbaa67190d5839172 : SUCCESS in 12m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a645df58ab24e16906ba56c60ec19a0 : SUCCESS in 20m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3f968a11d43846e1becf9c9430717bd9 : SUCCESS in 1h 06m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b02eb03af69546d8b16c7cd52e28b1b7 : SUCCESS in 46m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/55194866e251455fb4b35566c7124747 : SUCCESS in 1h 08m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7023e6f323514eec8f2cb26039b40074 : SUCCESS in 19m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/159df48d3d754ebaa82caaf63f4f5b5b : SUCCESS in 21m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/47f7737409714808b7591e7f564fe335 : SUCCESS in 1h 43m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0704cbd5a6d842f6a038e877784b8d3a : SUCCESS in 35m 50s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3bbdc77a0964c3990f6f6b777ee337c : SUCCESS in 30m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/70616c1e6e5f4f3d9ea1833b27d0e9aa : SUCCESS in 27m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ec67a4f34aa64c2f89aea9ab5dcc52dc : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/897fe22b715f443a8d5620bc2b55fa0d : SUCCESS in 10m 27s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"277c23939fb8d759f61e510a008130790c4511b1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-03 15:56:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bdadcf7fc684186338a6b03e20678f7d94388c50","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-03 16:01:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"baccc05b8af202823ae7d0da61de58ba196b996b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 16:01:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c360631897f260757f4bde0e86232f11085510c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 18:07:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca892872ac6464fbac559ad08ef4f16 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00efaba2222d4722b7646275e819d9ed : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7e559f2028f4745acbd3d230bc154b8 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64d6ea4fa71641aab9bc4b9844939461 : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd37a714d48a4ff69c1ad8956db5051c : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/940b4fa447ef45ca82f2c0055ab5269e : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5c786d12353463ea0fbe2f4ab104672 : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d545c093402418eaf559bcd84bd9f6d : SUCCESS in 20m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d58ebc294af4411fba3b5bce80a9c024 : SUCCESS in 1h 03m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b13728f1e0c548e8ab2a0e534e2fe917 : SUCCESS in 47m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6ed268f1c7964f24895172ddb7b0b89c : SUCCESS in 1h 11m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f4c2400bb04d48faaf09ae898f7147ac : SUCCESS in 20m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c395ae04613f4125b1d9d002507349a2 : SUCCESS in 16m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a95007404de54d49bc06f9ac655da242 : SUCCESS in 2h 00m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8d2544451e5643e98dd152e685ba2052 : SUCCESS in 34m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fd7ed71c9dd4f1bbfc30519f689de0a : SUCCESS in 36m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e8fc9a82c8bb4cae805c84a54571cbf7 : SUCCESS in 25m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1a19a5c500c243c386dc5e50f1a6a8fa : SUCCESS in 13m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe966a039143466e9d42059eba1feafe : SUCCESS in 10m 18s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"0ba89b63057c6c2a7dd9a2fe0428770072f8017b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 18:07:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ba03c9c235000bff46167cf037f64c8e90829fa1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 18:08:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daee975bbe4f43bc9a78fd0f3fdfd42e : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52ae8a1574624cb0afdb447834788f5b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af6495fa26087fb9cab904636da3ee8df708bb28","revisions":{"af6495fa26087fb9cab904636da3ee8df708bb28":{"kind":"REWORK","_number":1,"created":"2021-08-03 12:47:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/803283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/803283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/803283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/803283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/803283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/803283/1"}}},"commit":{"parents":[{"commit":"8ac837d9bc8b0aac0e52e62fb106713154e823bb","subject":"Merge \"Remove stale DIB_DISABLE_KERNEL_CLEANUP references\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ac837d9bc8b0aac0e52e62fb106713154e823bb"}]}],"author":{"name":"Lukasz Zalewski","email":"lukasz.zalewski@kcl.ac.uk","date":"2021-07-25 13:10:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-03 12:47:32.000000000","tz":0},"subject":"Ubuntu: Set MTU for veth in the network files","message":"Ubuntu: Set MTU for veth in the network files\n\nConfiguring veth pair MTU in the systemd-networkd .netdev files\ndoes not set the correct MTU for the peer and there isn\u0027t\nconfiguration option in the [Peer] section to do so [1].\nThis results in an MTU mismatch with peer having the default\n(1500) value.\n\nThis change moves the MTU configuration from the .netdev file\nto the [Link] section of the corresponding veth pair .network files.\nSame approach is already recommended in [1] for the tun and\ntap devices.\n\n[1] https://www.freedesktop.org/software/systemd/man/systemd.netdev.html\n\nStory: 2009072\nTask: 42882\nChange-Id: I082e1b807497d17ae40ddf86e1473fbfcdf625e2\n(cherry picked from commit 5a0206aef1f78d54845593556c13acbde99a7419)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af6495fa26087fb9cab904636da3ee8df708bb28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af6495fa26087fb9cab904636da3ee8df708bb28"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
