)]}'
{"id":"openstack%2Fkayobe~803365","triplet_id":"openstack%2Fkayobe~master~Ied01d2797420f844134dff8714e77a1ca6c1f7aa","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ied01d2797420f844134dff8714e77a1ca6c1f7aa","subject":"Update ntp role to match code style","status":"MERGED","created":"2021-08-03 15:48:43.000000000","updated":"2021-08-03 19:31:21.000000000","submitted":"2021-08-03 19:29:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803365","meta_rev_id":"50b8bdb4bd3447ab0b710a874b3d4d4f1134c186","_number":803365,"virtual_id_number":803365,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-03 19:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-08-03 15:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-08-03 15:59:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-03 15:59:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-08-03 17:33:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41ff77256e3cc00cab882694278f0adfc3b869e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-03 15:48:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0357968cb894c7764d96b742d415c6c1f37d78d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-03 15:59:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1efe41e821dd2450beaaa1feaac18ce746df69b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 17:33:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df6ce6a7c6034ea5b3d04cabd03e3efd : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c7eba15b4394711a3d5f9e84365420d : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c823668a1dff4d8f94c0cc9f2578729a : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4680beed75f84253b7426391347556a6 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e723ca0c9ba046d5ac9110ce4e0992f5 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e61a7a5a8c4448b8b97fa01e97880ce : SUCCESS in 9m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c6465fa75b749feb7c75777294da189 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/706a580484524a1697839c6bee1d3cb7 : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e743561809714fc68f821826befe24cd : SUCCESS in 18m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4c063f762ea243c598f370bccd7a829b : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4211c4f167214ef8a277830424c3348f : SUCCESS in 41m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/69d98459e02342d78bcf2b8f4ae9068e : SUCCESS in 1h 04m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/76984e577d3c4f39aaead8c818a3c194 : SUCCESS in 17m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee7745d8f2fe43a3890e6fab1bcf9e1e : SUCCESS in 13m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7b2758cdf1d44413b41a70bcc8427f8f : SUCCESS in 1h 41m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d440ba08d101469abd96c3e0c4cdbd4e : SUCCESS in 34m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf5b6a5aa8894e798831bb014c8f491b : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/236e106373654e499bfdda9e5eabd7ab : SUCCESS in 24m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d299da1f8eea401394919c7ccdd4f010 : SUCCESS in 12m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98164f61fed94793b4b6a836ab42b40d : SUCCESS in 10m 02s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"5cc2b8b656026d299251ce7dd895b7db96a2f098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 17:33:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b448aa5d8c21dde4beb61eaa450b4c8166a2f70e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 19:29:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b10e4af8ca4512863c9334d241a576 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c43464fe2d5f4b5fb50f665f018f3636 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/119886e49d674e5784b97e0268bca50a : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f452756067649418b1a3ddd0a9e97d6 : SUCCESS in 9m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/303304923abd413eb9a6e35d83704129 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ec9718de67c64e489683b899f3900da9 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/65c0179eb9a44dac8fefa7cd62fae680 : SUCCESS in 19m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8f3efc69c0064b5ca9fde9b25bb23f68 : SUCCESS in 1h 11m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee7e1e0b431447508246404841dd97cd : SUCCESS in 43m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/80b7f4f6c2894b2da98292c9c89c3169 : SUCCESS in 1h 08m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f6d1f9df3079469abb925854e500508c : SUCCESS in 20m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1310784b2b694dac9f8dddfbc0df4f28 : SUCCESS in 13m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/330627fc01414908a5d40625350b8863 : SUCCESS in 1h 51m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/785f30502e8d427c88e1f3c410af47a6 : SUCCESS in 36m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d1405edcc434a8fa89cdd8d51933c0e : SUCCESS in 29m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/885a33d27865431e85b782845bf72993 : SUCCESS in 27m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1e5bd9414f924e1ea13884853bdad13a : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2b9cc0a98064281920a2cde2157d26d : SUCCESS in 10m 44s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"c50d8d87227ac687a691556805728db3fad1f217","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 19:29:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"50b8bdb4bd3447ab0b710a874b3d4d4f1134c186","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 19:31:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97877646762f46d2a2341b55cb87df7f : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ea2287466a43f39e03a801e3f7f205a8618a030","revisions":{"6ea2287466a43f39e03a801e3f7f205a8618a030":{"kind":"REWORK","_number":1,"created":"2021-08-03 15:48:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/65/803365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/803365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/803365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/803365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/803365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/803365/1"}}},"commit":{"parents":[{"commit":"5a0206aef1f78d54845593556c13acbde99a7419","subject":"Ubuntu: Set MTU for veth in the network files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a0206aef1f78d54845593556c13acbde99a7419"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-03 15:48:31.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-03 15:48:31.000000000","tz":120},"subject":"Update ntp role to match code style","message":"Update ntp role to match code style\n\nChange-Id: Ied01d2797420f844134dff8714e77a1ca6c1f7aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ea2287466a43f39e03a801e3f7f205a8618a030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ea2287466a43f39e03a801e3f7f205a8618a030"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
