)]}'
{"id":"openstack%2Fkayobe~804852","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Ib267b648f011a7efa6440b6adf9ea8b9b5a0931f","project":"openstack/kayobe","branch":"stable/wallaby","topic":"dnf-metalink-stable/wallaby","hashtags":[],"change_id":"Ib267b648f011a7efa6440b6adf9ea8b9b5a0931f","subject":"dnf: add metalink argument to custom dnf repository","status":"MERGED","created":"2021-08-17 08:25:58.000000000","updated":"2021-08-17 12:15:04.000000000","submitted":"2021-08-17 12:13:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804852-dnf-metalink-stable/wallaby","cherry_pick_of_change":802703,"cherry_pick_of_patch_set":3,"meta_rev_id":"b2c22c6c8127e04351a21fbde84710e606283a15","_number":804852,"virtual_id_number":804852,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-17 12:13:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-08-17 08:40:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-17 08:41:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-17 08:41:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-17 08:25:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-08-17 08:41:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-08-17 10:14:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"603fbdee95c3c2c5bd2b7a41d526f22cc4f8209b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-17 08:25:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"116093cb8aea08781543a5a4607af26a1d77723c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-17 08:40:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5a6ff305e0980e54d1e8aa65894a13eb43767b7b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-17 08:41:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8bf50cb30bcba329f9e00d4b5d463b2026fd26e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 10:14:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22d4d22bc50a466985992afb57b8a523 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef64aea00dd548a98ab85581f08f8ae0 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/758c8b7d928f4cfcbd32609d6c0b0abd : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d175c14766cf484b97d6fdba4bd63010 : SUCCESS in 9m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a611610a08d43c8820666a8bd54a379 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1858eddef9f4826adb59b41af8eb4e8 : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9320d172c4b4b07939339ad8c300fd1 : SUCCESS in 4m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b8700c1a6924bd1a083d73ef580424f : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/01235d97f34d41459016b3567f30097a : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7757a28c15c49808b9301d8f62bff38 : SUCCESS in 21m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a053fadbd6e34e23bfc1e84b77162576 : SUCCESS in 55m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b7d58df096a4eff9d23472b9a410676 : SUCCESS in 46m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/581bac304db24714a7b84b2de3b0f58c : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/972dfeb2fdde41bd941b4bc3e31eeb4b : SUCCESS in 18m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b4fd34f49124052952660fc6098ab22 : SUCCESS in 15m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3e44088a7ff54a5da7e9e69f50396c49 : SUCCESS in 1h 48m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5b41f7a5268b4d09afbf609474135b75 : SUCCESS in 36m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bf475229a3a4be5bca2f4d0ec1c8e0f : SUCCESS in 42m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fbb02fbbf7ca47f09bc3acac0f631e88 : SUCCESS in 27m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/32123c3fd21941df83b348a3fd9106c3 : SUCCESS in 13m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5afa39a1ff549ff85409242e605c073 : SUCCESS in 10m 19s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"5fa669dc1ae56fab0c27eb56a64370fd129fcd1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 10:15:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c05692794fd10a43e65619e7af3bd7b0d833e8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 12:13:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a0e2d7eb5d548d18ff9b59d9f2070f6 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9761b48f15e642709a38449861fa1685 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a533fc2737b040c0828d298743379b34 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38dc27531fdf4686b7400d6fe1d0d391 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9a3602d37f04c60b6c49fff0390834f : SUCCESS in 4m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92b5dd9c8f4743669b15e267c5804252 : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a7f216e097ca4e58b6b24784e34015ea : SUCCESS in 12m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f5bf9be40af143e48803d97d02acf2e1 : SUCCESS in 20m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/56ec6106deb74c02bb13f5ee35f82b18 : SUCCESS in 1h 07m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8336596f3a394e2e82307013e2137cd0 : SUCCESS in 48m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/11b31f7242824570815ab18ace12291f : SUCCESS in 1h 09m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0a5e2d357f0242158bbd87456c9a85da : SUCCESS in 20m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2683144659eb49c6a09834ea56cf744a : SUCCESS in 13m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5c929e47cf474203a0c3309fe47fdb7b : SUCCESS in 1h 54m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a1474f9810f44eefa129a82d98d80657 : SUCCESS in 34m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9776ad62f947409abbe4a874f484923d : SUCCESS in 32m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cbcc6c2211d14a56b00b156189929f48 : SUCCESS in 27m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cb42c46342634bc88c0b72a5b5223043 : SUCCESS in 13m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90bfadefee3047469ed92b6747c942f3 : SUCCESS in 11m 03s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"66bf2b5aa8167c04bfe5b0f2bf33336de2483472","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 12:13:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b2c22c6c8127e04351a21fbde84710e606283a15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 12:15:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d097ac46cf714c68a13decbb4412a259 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f3ee05296174b6abce018a7ab44a749 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e14231c8eb4c25f12608b21f16c59c7c0a8874a2","revisions":{"e14231c8eb4c25f12608b21f16c59c7c0a8874a2":{"kind":"REWORK","_number":1,"created":"2021-08-17 08:25:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/52/804852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/804852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/804852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/804852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/804852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/804852/1"}}},"commit":{"parents":[{"commit":"8aa98acb524f5ea35d3ef5870fc09130130c4cd4","subject":"Merge \"Sync Kolla Ansible feature flags\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8aa98acb524f5ea35d3ef5870fc09130130c4cd4"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-28 13:06:55.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-17 08:25:58.000000000","tz":0},"subject":"dnf: add metalink argument to custom dnf repository","message":"dnf: add metalink argument to custom dnf repository\n\nAs per [1] mandatory configuration option is either baseurl, mirrorlist\nor metalink.\n\nEPEL repos use metalink and it\u0027s impossible to configure them via DNF\ncustom repos functionality right now.\n\n[1]: https://docs.ansible.com/ansible/latest/collections/ansible/builtin/yum_repository_module.html#parameter-metalink\n\nChange-Id: Ib267b648f011a7efa6440b6adf9ea8b9b5a0931f\n(cherry picked from commit a48cc2498859f5949d304266d32ce3e620ad7f36)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e14231c8eb4c25f12608b21f16c59c7c0a8874a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e14231c8eb4c25f12608b21f16c59c7c0a8874a2"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
