)]}'
{"id":"openstack%2Fkayobe~804853","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ib267b648f011a7efa6440b6adf9ea8b9b5a0931f","project":"openstack/kayobe","branch":"stable/victoria","topic":"dnf-metalink-stable/victoria","hashtags":[],"change_id":"Ib267b648f011a7efa6440b6adf9ea8b9b5a0931f","subject":"dnf: add metalink argument to custom dnf repository","status":"MERGED","created":"2021-08-17 08:26:07.000000000","updated":"2021-08-17 14:06:47.000000000","submitted":"2021-08-17 14:05:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804853-dnf-metalink-stable/victoria","cherry_pick_of_change":802703,"cherry_pick_of_patch_set":3,"meta_rev_id":"379a754b2e9db394e3f4e798f1b8be6c5675463c","_number":804853,"virtual_id_number":804853,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-17 14:05:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-08-17 08:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-17 08:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-17 08:59:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-17 08:26:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-08-17 08:41:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-08-17 11:56:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ccb792ebcb1be7166d2678338c31ca0f8cbd7cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-17 08:26:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1bd6f1eca89ec1d28525851e09b678e716b47608","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-17 08:40:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"060515ded896bea65ff4502a4f599ce3dfc42909","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-17 08:41:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"48d6375a986f3fdba7001b72921d5c359a8ceef0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-17 08:59:27.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7aeb762757ae8ffffc9548d50bcee9c4eccda89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 11:56:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81c8099c76984096afb6b8226aefccc2 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ee3017d9c6f4643a8b82b5870c864bc : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9977ec05e6944d58996e480334d2f4f : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ed0db210d6149bcbfa46f8a55f1aa4b : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a71376602949e78e573a82a5c9f909 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c49aeea4fe447c1b8d7c3ed4661c3fe : SUCCESS in 5m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe9b7dd5e5884fbb8785082e28dca14b : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8892a3a1672f4b7982b079b4832e2130 : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e7a5e79f9a84366baf28f20c698dc7c : SUCCESS in 18m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8e98cd66139a42cea64e42e725cd1949 : SUCCESS in 1h 01m 11s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b55550c7e3764b7fabfa85c345d6a208 : SUCCESS in 58m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3197b141e819459d9a2b6adb60f762c9 : SUCCESS in 16m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/baac7ad857c246e0bb1052b121d4093d : SUCCESS in 2h 11m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cebf0672f3984094b94377aa6ae11d70 : SUCCESS in 29m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/46904683831742a4a7dfcd673034ffed : SUCCESS in 29m 46s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/bb201a0bbb534d228cefa8d98ffdd636 : SUCCESS in 11m 28s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"78c70b638b8c8c829ef01641adc661a62caeb77c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 11:56:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d400d4d597d30a588ee0e5811beb6330c2aacc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 14:05:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dfcf24dfffd44d9aed5b001e4084cca : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/baee1cc1a7ca4e37841a6fb7e8f83c18 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/145fbe3ce9544af5a42ed755f1406dc6 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59693f4a744845c3a3e57a46a160a052 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06cc26301ad940ea95cb5de0d6c21d04 : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba8c9b269e0840b690e3b580e43c1c88 : SUCCESS in 6m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2ad6d90b961e44308dd0b051196bf921 : SUCCESS in 9m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b56656beff714565bd78b45522530e68 : SUCCESS in 20m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/585b6b3daab84ebba73af9497cd8e7b4 : SUCCESS in 1h 00m 51s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f849f8d78aa247b7b2499daf1d95fa2f : SUCCESS in 1h 03m 45s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c4c994fe163c466fa81aa8d28699deb3 : SUCCESS in 15m 58s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2f9ee90b9f4342acb0d7182da9bd9a64 : SUCCESS in 2h 06m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b286aa125a51421983ac5978086ec9ba : SUCCESS in 29m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c0950773d0ec402ea4ab32248ef0bb09 : SUCCESS in 26m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/17bead771e09430eb9b74c782f55e942 : SUCCESS in 11m 36s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"e870b3fd7ba22a0f52a03b286c74d1e0bc4aba90","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 14:05:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"379a754b2e9db394e3f4e798f1b8be6c5675463c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 14:06:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95d58f4018248c38d6e75e5ab2d5266 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d26d8149e3c4bc4b08e22a8743bde74 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"761665ffa127dabc8fe7cb7afc229ed2d0b951ee","revisions":{"761665ffa127dabc8fe7cb7afc229ed2d0b951ee":{"kind":"REWORK","_number":1,"created":"2021-08-17 08:26:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/53/804853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/804853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/804853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/804853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/804853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/804853/1"}}},"commit":{"parents":[{"commit":"90f07dce6acc0c5569d8f77bb2466aa234f0abb1","subject":"Sync Kolla Ansible feature flags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/90f07dce6acc0c5569d8f77bb2466aa234f0abb1"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-28 13:06:55.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-17 08:26:07.000000000","tz":0},"subject":"dnf: add metalink argument to custom dnf repository","message":"dnf: add metalink argument to custom dnf repository\n\nAs per [1] mandatory configuration option is either baseurl, mirrorlist\nor metalink.\n\nEPEL repos use metalink and it\u0027s impossible to configure them via DNF\ncustom repos functionality right now.\n\n[1]: https://docs.ansible.com/ansible/latest/collections/ansible/builtin/yum_repository_module.html#parameter-metalink\n\nChange-Id: Ib267b648f011a7efa6440b6adf9ea8b9b5a0931f\n(cherry picked from commit a48cc2498859f5949d304266d32ce3e620ad7f36)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/761665ffa127dabc8fe7cb7afc229ed2d0b951ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/761665ffa127dabc8fe7cb7afc229ed2d0b951ee"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
