)]}'
{"id":"openstack%2Fkayobe~805263","triplet_id":"openstack%2Fkayobe~master~I4a15578135a48cf282672288f4db24ded65f3b82","project":"openstack/kayobe","branch":"master","topic":"story/2009129","hashtags":[],"change_id":"I4a15578135a48cf282672288f4db24ded65f3b82","subject":"Fix overcloud introspection data save","status":"MERGED","created":"2021-08-19 18:28:00.000000000","updated":"2021-09-02 08:36:16.000000000","submitted":"2021-09-01 11:31:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"805263-story/2009129","meta_rev_id":"499bf049cdb7f8d2cae906d14eb3e3d122a386c2","_number":805263,"virtual_id_number":805263,"owner":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-01 11:31:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-01 09:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-31 11:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-09-01 09:02:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-09-02 08:36:16.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 20:24:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 21:10:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-08-23 08:51:49.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-08-26 12:44:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-08-26 19:18:07.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2021-08-31 11:29:01.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"b48fd5b15520791cc3ef4dacb871bf3b208c9a8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2021-08-19 18:28:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"252b3acbcdda5d1308a5c5ce27fc2acc2320eefe","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2021-08-19 18:30:17.000000000","message":"Patch Set 1:\n\nAs seen first on Victoria and reproduced on master.","accounts_in_message":[],"_revision_number":1},{"id":"dbc76663a03407732a21d6ff0ec37f908a618574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 20:24:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e50acebd9864fc39bd7a158862881e6 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af98d09b68e5415eafe997e64ac538dd : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/884e7c693d934aa2a4277c1dcc091ded : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1978caae3adf4caa90ac814ae05f2ca0 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc3c643a6c7c4861b8d9309ead28b4e9 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224a24d7693640e2a85b9e9d76ac5989 : SUCCESS in 8m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/07974c8f578f4a6b9e2be92f9f8c24d6 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4ec750a50abf4bcc93ee38c087503750 : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4a82416bf5a84b228b5500220b929711 : SUCCESS in 38m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5b5a21933e754b6e84e1397ea92fe95a : SUCCESS in 1h 03m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6441502615f46e9aaed65bc1eb92a9a : SUCCESS in 49m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8f663cae88884a61a44e8fcd8c95e118 : SUCCESS in 1h 09m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4891ace77e9e4997b224719429df087c : SUCCESS in 18m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6632e9f36003439c8f9fd5cea66c8971 : SUCCESS in 14m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e3a25218e6b447468d8e79e625cd2f4f : SUCCESS in 1h 51m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/757909bdcf4d4bc395f0ac71efe0c56a : SUCCESS in 1h 31m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08e191f5c8954f3785dcc85669724114 : SUCCESS in 29m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/58bc0ab2f8a24315aa197c9214509b5c : SUCCESS in 28m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2f828cfefbc94a349ed3bc1538b63c9f : SUCCESS in 14m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68b71f269d274d8ebea856a6b7a5131c : SUCCESS in 12m 23s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"bf5a0634592e473554ebe8023449906b558e3fbf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-19 21:10:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcd284848ed7fddef4a79e2bf714b640e5c4581f","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2021-08-20 10:56:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62007f6abf10085e7fe45ecb1f1e6732efb50da5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-23 08:51:49.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"120e53a3228703a8c14ad1cb718c42db1a067181","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-26 12:44:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"91b06e21f80beeef17be4a16fe15d0fb7ce71d77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:44:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2147fe7eda1d6ea0d0f46976bb56aba38847216a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 16:17:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfcdb51c90274f8bb257fc7b4afbccab : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e62efc0918c34fc8ba37262a3bc895c1 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adf47526330d4e54834352bbc67c3b2b : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/354fc66098654a22958c7c461a926615 : SUCCESS in 9m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46c905448a14418982f403acbfb66b64 : SUCCESS in 7m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f816062e4be94109a7103cdfff0702fa : SUCCESS in 13m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3697b27006fa485fa17b3153e6b7c6f3 : SUCCESS in 33m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d1740d2f99e8474986403510977bbb22 : SUCCESS in 1h 10m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07a18a0125384fcabd56ff3a870e1690 : FAILURE in 41m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f3799ffb8bf44d9aacc49f3507f6c7ad : SUCCESS in 1h 09m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d67108487219427493394220d923dd9f : SUCCESS in 21m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42bd1dd083a7477dbe8b93c50f0bafde : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/da3639347dc7417b9be1163a48d8f338 : SUCCESS in 1h 57m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/35ca47735a3940d6a9561d4b585d8c57 : SUCCESS in 35m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12505a2629e241ed976c8c71bf1a584a : FAILURE in 31m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/89802fc238a9465899aaa961cb98c077 : SUCCESS in 26m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d2e568ee287747ef9d4e21295aafd0c7 : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f30832571fc24f6caef3708f7978918b : SUCCESS in 13m 07s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"be548d394a86d89e5381da65db4cfb95426d9c9f","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-08-26 19:18:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5e986c02046290e384738cce0e5a19d62b58930b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 21:11:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/207546ad53184d529413fdc5f9acedc4 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7d1bd14c2554c37a70b2dd50791cde2 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/005f2c531e1540d382ca32a18b413798 : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eacfe912f82b4b82870e22a378d7fa6f : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef9a4591daf24ecaab9dffcd0a4b3bc5 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96078163b8734f2398d26ee26fef9032 : SUCCESS in 9m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d7ed8d8fbd5849f695f4274cfa420d32 : SUCCESS in 6m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2289134e35204cfb83c11e1a4a00b021 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/69269a3f596445d9857bd6b3cd6444fe : SUCCESS in 32m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ca5c31c2345d485786826db11dcfcead : SUCCESS in 1h 06m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8393220ceccc48369adb3181563206f1 : FAILURE in 36m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/72a2e3ed6ca648d3b62a8df3314bd449 : SUCCESS in 1h 09m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/080d5b7b7e2a4a6d8ae10b277fbdaa51 : SUCCESS in 18m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1725c4262feb430ca9db0b6cf0a0c1a1 : SUCCESS in 13m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2bfc91cb99ad467b8aea58825744041c : SUCCESS in 1h 52m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fb4e7fa958aa4bec9409a7f34a217108 : SUCCESS in 33m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/197e9424785d48559479585e6901234e : FAILURE in 29m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7ca8bc2189c54575ab2d9c809c9a2a73 : SUCCESS in 24m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/00030f2d5a2b49128b539b950ffc2818 : SUCCESS in 14m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e52731a731c546c68595cf4e2012e8f2 : SUCCESS in 12m 36s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"4d77da79c446184136c94629a397c74339f06680","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-31 09:27:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e9da9c8b1e557420424db7f78a13aa851161eb61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 11:17:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63ac16a904a9479bb6bc37bef1de1b55 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/665459087d5849fbae80035eace629e6 : SUCCESS in 8m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22bbe8b192874e688cb3d6cb5d339792 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/290535834722405e80318fd2554981ba : SUCCESS in 8m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/107e40d3c3be4c71adc28908460219eb : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97b155dc9098404da293d0b3e782146c : SUCCESS in 10m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/83d0c0f3f1964427b7eab5df7abaa53f : SUCCESS in 8m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ed265f9fa275414eabbbda54ebfe68b9 : SUCCESS in 14m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/39c036603aa344c5bb3bd14fd521a190 : SUCCESS in 29m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/22111506ffe24c3e8d05ae92f5263b7b : SUCCESS in 1h 02m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b4b58bdff574e08a1cd4a00564ee2f4 : SUCCESS in 49m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/14a273bac6574dcb8a2c11171f8a5768 : SUCCESS in 1h 06m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bd2ed0ab083c4e008062295e8c8e11c5 : SUCCESS in 17m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/837184c1962847a4aa91ccc814cdfc74 : SUCCESS in 14m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0313c080bd0e4a02b5d43391e1acb1a3 : SUCCESS in 1h 48m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6c92d718125a4fd2bbe58f97bd663627 : SUCCESS in 34m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55c67dfb79794109a6f587a37cb9e60e : SUCCESS in 54m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/730b8dc2942f43a49315edf4fdd5dd1c : SUCCESS in 26m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e5da8b32467945da9fea02a219f892a9 : SUCCESS in 11m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54bc53957f31495d9f72106d375ff2cc : SUCCESS in 10m 16s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"776befc68b3dbd8bba685ad3a23949dcf3de62ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 11:17:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32613e863f67ee8b0561fdf662d75357ac724783","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-31 11:29:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"064aa8d09aed50ec248063371ab285c9e8b1d884","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-31 11:30:25.000000000","message":"Patch Set 2: Code-Review+2\n\nI added a reno for backport, reworded commit message while here.\n\nNote that I don\u0027t think I have seen this issue on Victoria.","accounts_in_message":[],"_revision_number":2},{"id":"b2ba4ee85d4b1aa94ed4fe39ba07f4fc7fb0e4f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 13:29:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e042f41c0e1c46fe84c9cd50d296262a : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3fcdd61f2fa490996e90951cb04ba9c : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a1bea35a33941ff868dbc65d66d7d89 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65dc6746593449e0bfbdb114004f9c52 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fb4be08ac1549a3ae384458e398d15e : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2322b26efe114b3a9c84866051382c8b : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b89d8c5488b5480fa1cacbf8f90ec43d : SUCCESS in 5m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f36f37fcf8474f7b87376bd49e2ae874 : SUCCESS in 7m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dfc464e7d2324e8dab9cca30c97e257b : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f7201867703840c8a593eb267f162756 : SUCCESS in 30m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6252b4c9966f454882f1bd1fc80401b9 : SUCCESS in 1h 06m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64ddaaef1de34be9a54431905d36ade6 : SUCCESS in 46m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/87da8edb71a64d9892904ffed4d1fa69 : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5b2b7bb875934e2c9a11f2723c4b4c86 : SUCCESS in 18m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68f88e808dbd40afa9b6f874a5c54a84 : SUCCESS in 14m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a925f90c3365401fb9b6e1c095f18636 : SUCCESS in 1h 56m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4d3dbaf34fb7410fa82cf20a637d69b6 : SUCCESS in 34m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce40ee39a4a2485780c62bd4978ab0e3 : SUCCESS in 28m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8c4bca6775104730b3f2332eac68217a : SUCCESS in 26m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/36a0190d40fe4e64a63921623a110684 : SUCCESS in 15m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a1f0aae37bf4799985d83263c386e20 : SUCCESS in 11m 48s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"dd339026e437277c7c9c66a6c6a7600b90dc8d6b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-01 09:02:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1f42fcaee2d20955abebc6fbae4859a43a949431","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 09:03:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"41747f866d389a78b19d374c4bf32fa8c1f8f3ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 11:31:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba41689afe54aba9a435348e490f06c : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78de19ce49c64c8facdbb72b72c90120 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e571af96b72d482592ef3a58fd2b1089 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84596f3a67a4aeba0334d969d399ba9 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7df11940eea45199a920afa732e3273 : SUCCESS in 4m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1c222da088654cf99dbf518316a54abc : SUCCESS in 8m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff39aae6937442de980d45226d836338 : SUCCESS in 13m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bb36950d19d34f60ab7c57a945bf220b : SUCCESS in 31m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/67547895a4ab43afa65ef889f9ef8e7b : SUCCESS in 1h 00m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61b6b0c44cd640aaa891bb63a881d994 : SUCCESS in 48m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1f59dfbe404f4438868ce8a93e8c0cfe : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9b1ea08f15eb4f27b6d6e4a322a3ba14 : SUCCESS in 19m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05fa00269b224e1da470e48452ba1d8b : SUCCESS in 13m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ed039b5a164b46bb94bf8b3ad265bf81 : SUCCESS in 1h 51m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/edd93fbf20bb412d8cdb3162db9c1936 : SUCCESS in 33m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91be2aef96d24379999ad12c1bc01db4 : SUCCESS in 33m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/df68dfcd8f6243b0b1e5f49db2dacfc7 : SUCCESS in 26m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fb2ecc4c0ce8479c82abde6ba8b940cb : SUCCESS in 14m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b0c948227184387b4c2de362cea87bb : SUCCESS in 10m 11s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"24f43c684370dfc184bae68b9732452c25022b3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 11:31:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bb1a7d29cf5232146f255a87d3e560396a3c79ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 11:33:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3180a8fe14f645798c05bd0c7d570db4 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad5a5b7ad07546ef9c7e45422b92125e : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2},{"id":"7c4808192ec7ed73d03bf670ec8037071ae3bbd1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-02 08:36:16.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review+2\n\u003e \n\u003e I added a reno for backport, reworded commit message while here.\n\u003e \n\u003e Note that I don\u0027t think I have seen this issue on Victoria.\n\nActually I tested it again and it does fail on Victoria. Mark already proposed the backport.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3947997ed24a34f8913bb536a4f48a06058a0da4","revisions":{"69c29fbe2198fe62c7517ed8bc60592fd0afc557":{"kind":"REWORK","_number":1,"created":"2021-08-19 18:28:00.000000000","uploader":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"ref":"refs/changes/63/805263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/805263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/805263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/805263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/805263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/805263/1"}}},"commit":{"parents":[{"commit":"a48cc2498859f5949d304266d32ce3e620ad7f36","subject":"dnf: add metalink argument to custom dnf repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a48cc2498859f5949d304266d32ce3e620ad7f36"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2021-08-19 18:25:28.000000000","tz":60},"committer":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2021-08-19 18:25:28.000000000","tz":60},"subject":"Overcloud introspection data save broken","message":"Overcloud introspection data save broken\n\nThe command kayobe overcloud introspection data\nsave has been failing due to lack of OpenStack\ncredentials when running the bifrost dynamic\ninventory.  Add OS_CLOUD\u003dbifrost.\n\nStory: 2009129\nTask: 43059\nChange-Id: I4a15578135a48cf282672288f4db24ded65f3b82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69c29fbe2198fe62c7517ed8bc60592fd0afc557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69c29fbe2198fe62c7517ed8bc60592fd0afc557"}]},"branch":"refs/heads/master"},"3947997ed24a34f8913bb536a4f48a06058a0da4":{"kind":"REWORK","_number":2,"created":"2021-08-31 11:29:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/805263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/805263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/805263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/805263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/805263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/805263/2"}}},"commit":{"parents":[{"commit":"37524ff2a67c2619eaea3934a0635f09232326d6","subject":"Merge \"Fix documentation links for routed control plane networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37524ff2a67c2619eaea3934a0635f09232326d6"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2021-08-19 18:25:28.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-31 11:27:46.000000000","tz":120},"subject":"Fix overcloud introspection data save","message":"Fix overcloud introspection data save\n\nThe command `kayobe overcloud introspection data save` has been failing\ndue to lack of OpenStack credentials when running the bifrost dynamic\ninventory. Add OS_CLOUD\u003dbifrost.\n\nStory: 2009129\nTask: 43059\nChange-Id: I4a15578135a48cf282672288f4db24ded65f3b82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3947997ed24a34f8913bb536a4f48a06058a0da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3947997ed24a34f8913bb536a4f48a06058a0da4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
