)]}'
{"id":"openstack%2Fkayobe~805764","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I0c41d59123aba3351ac3b0a5a76cff517cca93e4","project":"openstack/kayobe","branch":"stable/victoria","topic":"inspector-port-desc-stable/victoria","hashtags":[],"change_id":"I0c41d59123aba3351ac3b0a5a76cff517cca93e4","subject":"Add twentyFiveGigE to port descriptions in inspector rules","status":"MERGED","created":"2021-08-24 09:01:19.000000000","updated":"2021-08-26 16:20:00.000000000","submitted":"2021-08-26 16:17:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805764-inspector-port-desc-stable/victoria","cherry_pick_of_change":802709,"cherry_pick_of_patch_set":2,"meta_rev_id":"43c5cad00172ba0552f9065ea78330968d500aaa","_number":805764,"virtual_id_number":805764,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-26 16:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-08-26 12:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-08-25 14:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-08-26 12:42:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 09:01:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-08-24 11:50:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 12:42:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"cad154dabd45e991f58ed53638ced2c214c50be9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-24 09:01:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7332201a93bfc8beb96452302638e1c07997039b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 11:50:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/861fbd18527c44de903a58040e9b936e : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe32dd05ec504206a7d2e8eeef0fa524 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9187b1b005bf418da76e66061ec83aa9 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39b31e04db9843b9b6104c5f42e901a7 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1b04bb96f6d43e3be897945fccac9ba : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e918669117d04dae84275d0eb1febde1 : SUCCESS in 4m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/34e7049509944d819d31272d79b5acf2 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c03d18c1ed4c47b7a8472975c9ecd9c8 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b24455ea5d6449faaaca494fd605893 : SUCCESS in 18m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/03f1ac5b56424c18973df737fa4b333d : SUCCESS in 59m 21s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9f204b03ec70420ea124d4375133d2e4 : SUCCESS in 1h 05m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7130b76d7e4b491a93cd07cb6df9a20f : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3d605db9d6a64cecad324bf746ba69e6 : TIMED_OUT in 2h 37m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7e25fa6b29004942a3d48c74df6237e0 : SUCCESS in 28m 33s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4758275bd2154546b70d40ae146fbbc3 : SUCCESS in 26m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1f13c398e87745409065c21beb9fb3dd : SUCCESS in 11m 11s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"c2b7cab3d42ce0e84c3fb75037bb0bba47f3c5fe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-25 09:11:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a87e6fc5a5cd7fad27419a00f7d33390dd99d931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 11:23:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c55f581c67e747899d5b70e483e10343 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88ecb1ce250c4e3d98cd1887cf3f2fe6 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b3ff0671a7645a08d5f11a9dc2a26f4 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3460e9f68a184158857c9f55a1bf94e4 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71482a64993b409fa2580e1a9a6fc76e : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/947bb6a6432b450dbebbe1a4b4502cf4 : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b214931ffadc4841a7ddca7264f86aa3 : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cdfe41886aac4b22866f4ff74328f91a : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1f6aa01b1b0a487bbc8fd58320da5e35 : SUCCESS in 20m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/978fa1fe66b145cd8c0b4bfdf1d10a79 : SUCCESS in 58m 20s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/4398843ff6a0461b9f302e899875b124 : SUCCESS in 1h 03m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/083551926cd54d5cbcd9f829969fc217 : SUCCESS in 15m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/99cb3440966d4a348a90eb7cb2ea1c79 : SUCCESS in 2h 10m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/909761fcbeef4be7acf1c4be81ab4846 : SUCCESS in 26m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c8c16f0f462c469e8744f3bfe85dfe43 : SUCCESS in 24m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2db8951b2a144ed9b99f454763f0dce8 : SUCCESS in 10m 59s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"be91235fb8053a8513d3cc1879c623a89a94b8de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-25 14:57:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8918ea8ffa26aad9e377e08654b980b1e4fb6546","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-26 12:42:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"028d1d42edcface0fea86cfc726b0e8959f00fca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:43:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6da6b46cac220186489a581b4baa812e80ddb44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 16:17:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17aba996281944e89a46751704da9d83 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa5b0c83eb804fe4a80bca1eb1709a43 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c5308c72f6444998d092a807e863528 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59001b07490d4571af1523513eacb44c : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b581bec4136c42738a5430454bb05697 : SUCCESS in 5m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d176679f3344deaa92dfc0f9e99fdad : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7758ea07b404bbb8d9896f984367076 : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3893af5ac7e14ad6ac7eedbd4cad3790 : SUCCESS in 22m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/810dd39bc9ff40aa8cf9855b61c4bbca : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3dec7fcfcf5d497ea01d61454be0efd3 : SUCCESS in 1h 03m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8972a6d4936d42f1ab24a9b45586908f : SUCCESS in 15m 17s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bfba5ec5004e4b32be6e3e2305fafe03 : SUCCESS in 2h 10m 59s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/39b91bcb78d54d2ca3dde88e47575b38 : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/65cac78e800c4d8fb74412fce7112c2a : SUCCESS in 27m 55s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7f435729ddf2416096d4015c938921e4 : SUCCESS in 12m 56s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"9a5e53d0d424b7803038bcc70abb7e5eccf07c6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 16:17:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"43c5cad00172ba0552f9065ea78330968d500aaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 16:20:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c1359d557349c7a217a0558930e98e : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc0df8988f7f4a28a473665ac0f48935 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e98ce3b64d54a834dd4c1162bb2dbf91ccd4bbc","revisions":{"9e98ce3b64d54a834dd4c1162bb2dbf91ccd4bbc":{"kind":"REWORK","_number":1,"created":"2021-08-24 09:01:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/805764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/805764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/805764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/805764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/805764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/805764/1"}}},"commit":{"parents":[{"commit":"761665ffa127dabc8fe7cb7afc229ed2d0b951ee","subject":"dnf: add metalink argument to custom dnf repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/761665ffa127dabc8fe7cb7afc229ed2d0b951ee"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-19 08:39:35.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-24 09:01:19.000000000","tz":0},"subject":"Add twentyFiveGigE to port descriptions in inspector rules","message":"Add twentyFiveGigE to port descriptions in inspector rules\n\nChange-Id: I0c41d59123aba3351ac3b0a5a76cff517cca93e4\n(cherry picked from commit 91b2dddc415e1a5ca6aa66fdd65d7f97f5781cc3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e98ce3b64d54a834dd4c1162bb2dbf91ccd4bbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e98ce3b64d54a834dd4c1162bb2dbf91ccd4bbc"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
