)]}'
{"id":"openstack%2Fkayobe~805765","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I0c41d59123aba3351ac3b0a5a76cff517cca93e4","project":"openstack/kayobe","branch":"stable/ussuri","topic":"inspector-port-desc-stable/ussuri","hashtags":[],"change_id":"I0c41d59123aba3351ac3b0a5a76cff517cca93e4","subject":"Add twentyFiveGigE to port descriptions in inspector rules","status":"MERGED","created":"2021-08-24 09:01:26.000000000","updated":"2021-09-01 11:33:02.000000000","submitted":"2021-09-01 11:31:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805765-inspector-port-desc-stable/ussuri","cherry_pick_of_change":802709,"cherry_pick_of_patch_set":2,"meta_rev_id":"2a1e2d749acb6013caf0dd844e874d6ab97703ad","_number":805765,"virtual_id_number":805765,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-01 11:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-01 09:01:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-31 16:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-01 09:01:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 09:01:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-08-24 11:51:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 12:43:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"b81af018747267214b523f0125a9540cfcf55a36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-24 09:01:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6f5a0b73f2c6763c97a50dd27c55072a44473e05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 11:51:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43a54686182446d7a7a040e3ee94c7b7 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7bd3bcca54649118164fca6f4e6d742 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0d87d00f61e428a8476483628440e0e : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ec05f9935f44887a10724453055d609 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c37d80f36b41f0926781f61222de2c : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ede7339126c941cba1bd61841a617e90 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7c4782a05c144491aa6e70bdb078dea4 : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9807e9de2dc340019d9a217be71e53fe : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a4aaa43765b42cc903663c16f04b9d7 : SUCCESS in 17m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7e17420960934cf683dcb8dd583ef1f9 : SUCCESS in 1h 04m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b03bf66c9bcf4ca4a5854eb9289d930b : SUCCESS in 16m 47s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/be86d3bb99834879b60bdb761e207526 : FAILURE in 1h 00m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9e4f6c7b577542b3a1350561a2863102 : SUCCESS in 31m 14s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/515a5b94fe3a4a748ec31886e3a80670 : SUCCESS in 27m 11s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2f6f696894244174ac1401f20f2005eb : SUCCESS in 11m 20s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"2899b3e2d0114c5cbb34f8a783f0d0581b9a3740","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-25 14:56:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c82d3ba7f4ea7dea6b5bd0daf59b5941cc317fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 19:56:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae8a945279864e6d8b51d5545e29c434 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a78fe458ec964a75b864e5a3ccc5e83a : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e913a3ff86554db6b446963b33d2b8fd : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd689d7bbe3a4459a7558f4bcf3bdb4a : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7018a99868864f1ba4f6f987ca856304 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/491e7a6201bd4a33a8d8541bed49f679 : SUCCESS in 5m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/473042873d2642cab5ed7c97584348eb : SUCCESS in 7m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/110d4a053d7f4c6a95bfeacb1dc1f6a8 : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/648a48040dda4cc48478f5ded463dcda : SUCCESS in 20m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/47cc3ccb346e4167b9c11eb8d29f4b16 : SUCCESS in 1h 00m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/dd1e2cd08b8545e7b1221344bc996bbc : SUCCESS in 15m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/95497c15225f4399aeb122529be68edd : RETRY_LIMIT in 1h 29m 44s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f55c35bd4af44e509112378a557baf82 : SUCCESS in 31m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f6fc278971f241918690ec62cfc3370b : SUCCESS in 27m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/53e8f68ea1f94d43a20b1d7d3efba45b : SUCCESS in 11m 08s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"2be3530155b85c645f652ec47772d7ebb25773b5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-26 12:43:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d44a56e38a28b7d9df027747e4f254315f505ba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 14:53:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77a76b1553a3402a82c8dc539bbbd14e : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee3b55286c641d9aabd64395d10155e : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/748e448d65594c13bcf32f41a0e18073 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6388a5f572f3413d904669afd8babf60 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840e090ac8bd4cecbff6551ecea2deaa : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f945861c7f4640ef8c053ad315375c5b : SUCCESS in 4m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0fc4bd9e6e4b4665b7fb325b8d830b69 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b981b18907a476b812ceb7ae52dd81f : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f113e7c187b440a86d230ab54cc6e35 : SUCCESS in 20m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d1a2f9a104634ab2a10e78d44eaa8ba9 : FAILURE in 59m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5a089c4703404cdd9e80e25b44b57230 : SUCCESS in 16m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1f8f5f46f892474f8df8937b4b80d71e : SUCCESS in 2h 00m 45s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/494825a1aa04482d8e00099a46db9f29 : SUCCESS in 30m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a952ff46e17c4bf8ab200dde359dfae5 : SUCCESS in 32m 10s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a0afbefc333b45eb96fe49c64701b31b : SUCCESS in 12m 02s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"5376f47db99883d9570016933237dbb03f351563","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-30 16:33:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"29e213e09526ebff5fa6ebb093684840ad05e24b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 19:13:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ade454b69aa4425fbb95cb194db34f8d : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca29a4446d4f461db341c504de251f49 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fd5b07147284fe685328b4e3f0f0fe4 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/322ba2c369df4deba51cb0e933532eae : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2616a4787ddd473a9ec47665bff5075c : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3dc38fd0a6046e7bdc46197f1ecb34a : SUCCESS in 5m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f7014aac8254c26a9a5f3355217b1f1 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98fdc2e4079d4f0a95d90afc90756637 : SUCCESS in 10m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e4b7f60fa35448dcb94535c3152edce9 : SUCCESS in 18m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9f9e758b4c5e4ed6a31bfff4abfe4109 : SUCCESS in 1h 18m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4969d7aab1864c298f5cac9ad337d005 : SUCCESS in 16m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ea912cdb9270494cb9de6b9647e6d46c : TIMED_OUT in 2h 32m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6c0e8f39cd124eae88659b1f8790f177 : SUCCESS in 26m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dfc131de5f2441bd87823e48b1f1b37d : SUCCESS in 26m 23s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/cc5d1cf3349f402eb6f61d6be29491b4 : SUCCESS in 10m 38s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"d68af67b92c741c10aec8f840463e9bab790c4f5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-31 10:50:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"09389c30fd28ff54bca336037be16917d27f2359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 15:39:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fccbfd84692459d88cbb394ef01ae41 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bab4cb8a54a4336af819bb7598ee620 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20667107607c49bc88f93d9a6885aaa7 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc4be092fab04ebfb7b90c0751ee1219 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc916330c6884bc392d36f09dee11e15 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80e023bd30a24eaa9269220fbb6ee384 : SUCCESS in 5m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/51c0812949f5491ebfa5a4b3c42c789b : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5e7b7797d5dc483b92c0579802f1359c : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/28f0a779533842cc9f987f743e81e35c : SUCCESS in 20m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8402397e448640afada2878006463a46 : SUCCESS in 1h 08m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/62593e29e23f42a5a669de7254db7db7 : SUCCESS in 18m 20s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f85f4124fb8942c8b2580235c48d0d19 : SUCCESS in 2h 08m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3b6301ea67f642b28037c821435b616d : SUCCESS in 30m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9b8cce220fbc4fff92f5b86573aad59b : SUCCESS in 28m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d9d210449b05408ab000778f78983658 : SUCCESS in 12m 05s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"d9276fd04df9d5c876c4f307446a96f1cecf65ef","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-31 16:22:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8654f1f2b4e4d2459634a597184c41e2e7417b37","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-01 09:01:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"16120b22c5a5ac0e855fd5ecd42705fd4b2dede8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 09:03:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f0f8dc7fdabedf5e49149cf1fe2219df6f46ba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 11:31:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca797d8bacc4c82a8d2942f3a65ef17 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/917e2e6018db45758c23645d92084faa : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/304d5d79509148169f36fffd348b51e4 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d69d56d28049aeb5c2a22e4bcd7f9f : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43188fc9431442daa98f960f42c38851 : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f253785fc3ef420f853761ed1a9eb6dd : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8734a89c45ed4a28b37abd6302f1507c : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f758f880faa4acca311412daa250642 : SUCCESS in 21m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ea5e14625d534ff886b1f48a171f4a0d : SUCCESS in 1h 11m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/490a27e782c64b9e8a3c5d0fc2a587a5 : SUCCESS in 16m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f4ab8a8e1d4b425a81be1b1fbbb02b82 : SUCCESS in 2h 06m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ec6298c835a942a1a0ff7ff1bd9453a0 : SUCCESS in 30m 05s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d359c63b04324b2f93a0ae166cf091ec : SUCCESS in 30m 36s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9a84ea84f5ee4a139cad9cb0d144bcf4 : SUCCESS in 12m 05s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"794ca9c9737cb87dffb09170dbdca942482552e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 11:31:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a1e2d749acb6013caf0dd844e874d6ab97703ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 11:33:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709eb9cf7849479e9ff56d4f58d57d7e : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86d20af2c512400cb16d401d797787c3 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02ea57de3563e5c401a72175db82fc0e2215d1a9","revisions":{"02ea57de3563e5c401a72175db82fc0e2215d1a9":{"kind":"REWORK","_number":1,"created":"2021-08-24 09:01:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/65/805765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/805765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/805765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/805765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/805765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/805765/1"}}},"commit":{"parents":[{"commit":"d4445e3eb167232c595fd19cab6cc4e16e0f1ce8","subject":"dnf: add metalink argument to custom dnf repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4445e3eb167232c595fd19cab6cc4e16e0f1ce8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-19 08:39:35.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-24 09:01:26.000000000","tz":0},"subject":"Add twentyFiveGigE to port descriptions in inspector rules","message":"Add twentyFiveGigE to port descriptions in inspector rules\n\nChange-Id: I0c41d59123aba3351ac3b0a5a76cff517cca93e4\n(cherry picked from commit 91b2dddc415e1a5ca6aa66fdd65d7f97f5781cc3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02ea57de3563e5c401a72175db82fc0e2215d1a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02ea57de3563e5c401a72175db82fc0e2215d1a9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
