)]}'
{"id":"openstack%2Fkayobe~806892","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I4a15578135a48cf282672288f4db24ded65f3b82","project":"openstack/kayobe","branch":"stable/wallaby","topic":"story/2009129-stable/wallaby","hashtags":[],"change_id":"I4a15578135a48cf282672288f4db24ded65f3b82","subject":"Fix overcloud introspection data save","status":"MERGED","created":"2021-09-01 11:34:02.000000000","updated":"2021-09-02 10:33:47.000000000","submitted":"2021-09-02 10:31:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806892-story/2009129-stable/wallaby","cherry_pick_of_change":805263,"cherry_pick_of_patch_set":2,"meta_rev_id":"7a56e8c5a4124d343e4c5ed028068ae1d36c15a3","_number":806892,"virtual_id_number":806892,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-02 10:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-02 07:50:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-02 08:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-02 08:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-01 11:34:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"REVIEWER"},{"updated":"2021-09-01 13:35:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 07:50:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"d8c302d722efd047ddee440f1952a4bd910e3c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-01 11:34:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"169030f0539dc814f8fd685cf81daacbfd2fc78b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 13:35:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00d2faa940d64d68ae8a2ff0ace355c3 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ec210d4282340979bf4a71d6593929c : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bba384611ece44b89149537b21116ade : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7202f7e56674abba5325baacf299d07 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e59f8e8ddea349cf8d208c9d4ecddeb2 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d5ef567d7c497cb91230f586b35855 : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae8a7856256947368e6d640e993282f2 : SUCCESS in 6m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/517d0acf9b3c43e8adb21daea9bdf810 : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/57474f0dc1f54c398124cd3cf81023bf : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f1f8e5935eea427cbe10c21088889119 : SUCCESS in 22m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fa1fcc15d6a94f3c8057fdc799c9fe9f : SUCCESS in 1h 10m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/289c161a13854f299f718a04e4917b7f : SUCCESS in 50m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/978f8f7f9c8d462392d3fae720207bd7 : SUCCESS in 1h 12m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/709abbede4754d838cb98eea70720738 : SUCCESS in 18m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94f321708cae46acb1459b25f5b158ad : SUCCESS in 17m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b91bd75e4ffe49c7a832b03f4894e9f2 : SUCCESS in 1h 59m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2f998b3becb34f6d9f94e78bc3093fc3 : SUCCESS in 37m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42f5b822ebea46fe9423ff0f1dcfbb4c : SUCCESS in 28m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a9207f4612aa4563839e8c6bdb90e807 : SUCCESS in 27m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c6681fffbfb142f4adaed247b5cdbf95 : SUCCESS in 14m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08e11692b26543bc9084af3e26fa64e5 : SUCCESS in 11m 45s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"72b8937b9469d63a43af0211e4ebe83c2de5de8c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-02 07:50:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d59d6cb520281976f50e2cc438c590b81f6cacd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-02 08:35:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"933d22f039ffbda7a9e5ff9a70b303d120b9c505","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 08:35:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b23196693bc8a8c51b6325cd43981a2d7345915","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 10:31:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/221eb965dcba404fbfe1e4b30a8c9303 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39b49fd2ad7746138266a8f3af41dc47 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61c582ce41794543a2b38d57e8a955a2 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ca99dc4b3a43229bb9548aeeff3b23 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96ce00d087724d5f999f59fb63a5a453 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19f805c5f2804307944061cf8d6ca2b3 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6a445546f23244709429b0a081f375d3 : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/de8a60406f8b4e67a1ef72eb5980b9d0 : SUCCESS in 19m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/888f5a8d40194d43a955ea23701c35d0 : SUCCESS in 1h 02m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8041dde6cd54a479bde3922de05d213 : SUCCESS in 1h 02m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/aa0f28f1ff714101aaec8ce3b094a449 : SUCCESS in 1h 04m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6fa6228d20274db9af76404d1b7ff654 : SUCCESS in 21m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a1eac875ba1407490239ac8efafdee8 : SUCCESS in 14m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a67501fc94384f8aa733e8b2c7e3b54f : SUCCESS in 1h 51m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/09876803acd7403090fdbad1f68288f2 : SUCCESS in 35m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e8a49826b7748b2aab0183f5365ba1d : SUCCESS in 28m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/142c0043455a4eecb6988420a2150a6e : SUCCESS in 26m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/832af48e9d1b4351a0931fb6b32ed6d5 : SUCCESS in 19m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0aa51c741dd445efbe31fc5ff479ecd3 : SUCCESS in 12m 07s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"5b963254864fcb01510a09a2609f69996bcf9e64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 10:31:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7a56e8c5a4124d343e4c5ed028068ae1d36c15a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 10:33:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0234965392a44b1395ef646d3ee4b513 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df775b4bf6b5429789ba5c434492d109 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a906d26f0d68f0644adf50420b42c8207421d7e5","revisions":{"a906d26f0d68f0644adf50420b42c8207421d7e5":{"kind":"REWORK","_number":1,"created":"2021-09-01 11:34:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/92/806892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/806892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/806892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/806892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/806892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/806892/1"}}},"commit":{"parents":[{"commit":"b6ad0e6382170596545db31b648ca3839fa984e4","subject":"Merge \"docs: fix heading styles in upgrading page\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b6ad0e6382170596545db31b648ca3839fa984e4"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2021-08-19 18:25:28.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-01 11:34:02.000000000","tz":0},"subject":"Fix overcloud introspection data save","message":"Fix overcloud introspection data save\n\nThe command `kayobe overcloud introspection data save` has been failing\ndue to lack of OpenStack credentials when running the bifrost dynamic\ninventory. Add OS_CLOUD\u003dbifrost.\n\nStory: 2009129\nTask: 43059\nChange-Id: I4a15578135a48cf282672288f4db24ded65f3b82\n(cherry picked from commit 3947997ed24a34f8913bb536a4f48a06058a0da4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a906d26f0d68f0644adf50420b42c8207421d7e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a906d26f0d68f0644adf50420b42c8207421d7e5"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
