)]}'
{"id":"openstack%2Fkayobe~807157","triplet_id":"openstack%2Fkayobe~master~I70d5f78871867f3a71150eacc3df1857949e6df9","project":"openstack/kayobe","branch":"master","topic":"network-allocation","hashtags":[],"change_id":"I70d5f78871867f3a71150eacc3df1857949e6df9","subject":"Fix typo in IP allocation filename","status":"MERGED","created":"2021-09-02 14:11:45.000000000","updated":"2021-09-06 10:01:38.000000000","submitted":"2021-09-06 10:00:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807157-network-allocation","meta_rev_id":"a8140f81e41f747b3d202dcaf7d7593ac21fb477","_number":807157,"virtual_id_number":807157,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-06 10:00:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-06 08:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-06 10:01:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-09-06 08:15:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-02 15:43:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-03 08:19:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-06 08:15:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9ef942511f2387bcd3aa39473a75ff4972a187b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-02 14:11:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c5d304fa52d1ca3923c3efed5b3d5d927de7684","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-02 14:12:06.000000000","message":"Topic set to network-allocation","accounts_in_message":[],"_revision_number":1},{"id":"0d3635341a97f115dcdd786f98bfce000b0c6355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 15:43:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a89118736e3d4ad7b309bd1cc6cf300c : TIMED_OUT in 51m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9a02a53d58744b99fa0f94d48e2a597 : TIMED_OUT in 31m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b806b4673d174c90ab19e2feab1cdb78 : TIMED_OUT in 41m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0085a2c7be142c5a2bc71c277516cac : TIMED_OUT in 41m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a352399f3d354eb08cf69f25ebf738a0 : TIMED_OUT in 41m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d4897211d240608d656004fcbbe1dc : SUCCESS in 9m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a01ffd75ef234caca14cd913b6bcec41 : TIMED_OUT in 31m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aca3c206c4ce4b528fb2a16e0b86cd60 : TIMED_OUT in 31m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/401d635be5e2429a9550c93bea9d61ce : TIMED_OUT in 1h 01m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/acd531e63f344ef8909aea4f1d5fa497 : FAILURE in 27m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faf9f41de7f94b72a4ce3687f1251671 : FAILURE in 24m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/604d48f3a3804ff2853b0669a6e19057 : FAILURE in 30m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8efe97fb59e24db4a586089fa250cdf2 : SUCCESS in 18m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4430a786dbdd4240b4b34177af8c2e8e : SUCCESS in 13m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d0d1063a869e4e7a929c6c747e8e0127 : FAILURE in 1h 29m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/32537cbc81a54b2c8d7052e34c531809 : SUCCESS in 35m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e72a076233b48ce8fb767e51286014a : SUCCESS in 29m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a62a09ddea1546198f4fa53b2a0dab4f : SUCCESS in 29m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/83d5cbee6eb745719b9f03c1a1d7027a : SUCCESS in 15m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/529d8e140b3649718c7a7396052f6681 : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"33b0ed7c10015f46468ef028d28f3a09873ff201","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-03 08:19:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c440e903c5ffc98e70011e33149ffb7f66374dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 09:47:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adfb19612caf4623a1bff5e4c2d698b6 : TIMED_OUT in 51m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b80cacd3444a47c2bc9f9e2df9cd8ead : TIMED_OUT in 31m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f2ee4a1585d424ca7fa959a06022ace : TIMED_OUT in 41m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c52332b5ca9d4b84b9fb4d1f9e19d54e : TIMED_OUT in 41m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4ce9566f208492eaed1d2b8121589a1 : TIMED_OUT in 41m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/706879190d68455c9192902cf0f4ece9 : SUCCESS in 8m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aedcc4d4837b488893a00b890ff39a77 : TIMED_OUT in 31m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/42247bd8642841d2b450ff98ca374370 : TIMED_OUT in 31m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee3d9a303fa1489fad96ce494da90383 : TIMED_OUT in 1h 01m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7111de7c4bc04aaba30cf6d4d3b2aa16 : FAILURE in 22m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc7316fab51345fc8d25d21f8f96dad7 : FAILURE in 18m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bd8c8d26bb35423cbccaca589738fcc8 : FAILURE in 23m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8dc2f10606ca4ba69d6e030f1deae6e5 : SUCCESS in 19m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa1d383c71b54d76994794cf288aacb0 : SUCCESS in 15m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/09abb6e1c0eb4ac9a7a8145fb2cdec5b : FAILURE in 1h 21m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bdeb64a5154440358769bdaadb9df088 : SUCCESS in 43m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a532874e7a494d09a579d41c60d9011b : SUCCESS in 33m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ff844f5a00144b5e9fb46646a062a251 : SUCCESS in 28m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e2ed6360ef414ffeb4191c342b40acbb : SUCCESS in 12m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd3040e9cb2347b29d99ec74c0ea71de : SUCCESS in 11m 54s","accounts_in_message":[],"_revision_number":1},{"id":"696a0b4af6a269a19630ffd4faf64c7d76889a40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-05 20:42:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e9a7d26202bf264d39ad3cb35a83db549bd85be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 22:28:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d601b55811c44bd86606952a5f03bf2 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0ca20aa0f544d3bbacc635112064555 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac63ee8e3b1d41449603de840da3daae : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb3686a6ae344ce79ccc2ca3a94e5911 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a81e25698527494cae9690bf9e8ec46b : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b19ea224f554bc3b397f6e2aa9b2e6c : SUCCESS in 9m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78faf7d88d6f4c44bc3bfe5c40fd8148 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ec6b114577042278da763c24848e04d : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0a98f725898b4fdbb9fd1306a1f76aa2 : SUCCESS in 30m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0eecee3ba64d40a59ccb93d3aee1374c : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/215d866f213348a385d574e3b45ca008 : SUCCESS in 44m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/91201fd81b9243a08c6dc08aedd6c9e0 : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3dcf67db4be74777aa8333f61adbde56 : SUCCESS in 16m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72d897ddb29b4afd9b7c7ea0b31db80f : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4542585af43f4d59b51f1360335b7386 : SUCCESS in 1h 45m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/92c0ae7a46e14dc5892d0d72334e513f : SUCCESS in 33m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/768df235305144fe84ff4640469e80bb : SUCCESS in 27m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7fd8878d58e649708ce26d1a19162653 : SUCCESS in 25m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7e58ec90676741ba96a710bde7e11460 : SUCCESS in 11m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec89569d91c84262a45196d6cdfd5371 : SUCCESS in 13m 07s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"7b2cb708833baa966416e6ea014df1ecb6cfbc3b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 08:15:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"37829f770a781ece9ae58c1943d8913e818e99ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 08:16:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1df61f2c90e4ca968bdef87f3f3f72793c5ccacc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 10:00:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/148cf3656bc84663ba41d1ff02e5815b : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ad9fa5b217e4fc3b13ed23f4909b4a1 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/082c0c3e402d492bb38e42f085032d38 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e263b34ba2e9465c911dccf2b3122c4d : SUCCESS in 8m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/96d00fa6985d45579b126ad4781205d6 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a9bddb80e99f4a3b9c49a142079835f1 : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2443c087ed5841fa88db8a917e723226 : SUCCESS in 30m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f7a5e26968c141f09558ff937f306bf0 : SUCCESS in 1h 01m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64e552d70a1e4c80b340db5165627953 : SUCCESS in 44m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bb5c95261176412f87181140af2a0136 : SUCCESS in 1h 05m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/292bc740d4e74befb3e3ca18bde6cc26 : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/696a5805be3c46ada4abdd299ff05109 : SUCCESS in 14m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8a706c531ca047d3b2ce811f6a5153c4 : SUCCESS in 1h 43m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dc3298a4dee340d9ab8f605e51749be2 : SUCCESS in 34m 50s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a6dda8e54264f618406b0200be421ad : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/68975f65df464f3daf84de35076c5b34 : SUCCESS in 27m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8d13869f392249ef81e64cc13f432369 : SUCCESS in 12m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f8d5240c2e546d7a2a4120de09072b7 : SUCCESS in 10m 55s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"85452e66405c854c306c767bbf27471b4f59fd9f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 10:00:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6e9ffa38c00f7ab97a830f6c55b229959b196769","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 10:01:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e46e7cd66f34f64a0c0da173d2efc8b : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8769373b6adace69b041a18a4342870a98eaabf","revisions":{"2ad40545fe157121a2c109737063af70cc5db5d4":{"kind":"REWORK","_number":1,"created":"2021-09-02 14:11:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/57/807157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/57/807157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/807157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/807157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/807157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/57/807157/1"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 14:11:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 14:11:35.000000000","tz":120},"subject":"Fix typo in IP allocation filename","message":"Fix typo in IP allocation filename\n\nChange-Id: I70d5f78871867f3a71150eacc3df1857949e6df9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ad40545fe157121a2c109737063af70cc5db5d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ad40545fe157121a2c109737063af70cc5db5d4"}]},"branch":"refs/heads/master"},"e8769373b6adace69b041a18a4342870a98eaabf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-05 20:42:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/57/807157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/57/807157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/807157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/807157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/807157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/57/807157/2"}}},"commit":{"parents":[{"commit":"74eff837cb0870eae78b71bdefe6bfeb6634d6ee","subject":"CI: Fix multiple failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74eff837cb0870eae78b71bdefe6bfeb6634d6ee"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 14:11:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-05 20:42:15.000000000","tz":0},"subject":"Fix typo in IP allocation filename","message":"Fix typo in IP allocation filename\n\nChange-Id: I70d5f78871867f3a71150eacc3df1857949e6df9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8769373b6adace69b041a18a4342870a98eaabf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8769373b6adace69b041a18a4342870a98eaabf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
