)]}'
{"id":"openstack%2Fkayobe~807270","triplet_id":"openstack%2Fkayobe~master~Ibaf9d3b64f78504853c0c6dcf795126977278187","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ibaf9d3b64f78504853c0c6dcf795126977278187","subject":"DNM: run unit tests","status":"ABANDONED","created":"2021-09-03 08:30:20.000000000","updated":"2021-09-06 09:53:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c051e0058a50f9f77b51e1aa0880676f6a6cfd03","_number":807270,"virtual_id_number":807270,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-03 09:19:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-03 09:19:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c1ebb23098e7ec15716e26edd43e49b384bdd9d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-03 08:30:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5159b305bf2f100fc0c9a9c344b715ecbf161945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 09:19:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/734c8991cd954215ac680dc4d24a64b6 : TIMED_OUT in 31m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f05dfb5bbb54da288b72d1f72cbd4f1 : TIMED_OUT in 41m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b1387608452464690a200dc83450b49 : TIMED_OUT in 41m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a83f10bd96a496fb6e0afd1f06eb6c0 : TIMED_OUT in 41m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c051e0058a50f9f77b51e1aa0880676f6a6cfd03","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-06 09:53:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e789ffd2de15e720069e9fb1a2735479a6ca53b8","revisions":{"e789ffd2de15e720069e9fb1a2735479a6ca53b8":{"kind":"REWORK","_number":1,"created":"2021-09-03 08:30:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/70/807270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/807270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/807270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/807270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/807270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/807270/1"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-03 08:30:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-03 08:30:09.000000000","tz":60},"subject":"DNM: run unit tests","message":"DNM: run unit tests\n\nChange-Id: Ibaf9d3b64f78504853c0c6dcf795126977278187\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e789ffd2de15e720069e9fb1a2735479a6ca53b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e789ffd2de15e720069e9fb1a2735479a6ca53b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
