)]}'
{"id":"openstack%2Fkayobe~807345","triplet_id":"openstack%2Fkayobe~master~I4cad83c92c32d1db2a5d3c03a78a1c533a8b9633","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I4cad83c92c32d1db2a5d3c03a78a1c533a8b9633","subject":"CI: Fix multiple failures","status":"MERGED","created":"2021-09-03 14:49:16.000000000","updated":"2021-09-05 18:44:46.000000000","submitted":"2021-09-05 18:43:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807345","meta_rev_id":"8273356901c4d675f2f2babc594da386b94658b2","_number":807345,"virtual_id_number":807345,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-05 18:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-05 17:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-05 17:13:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-05 18:44:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-03 17:43:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-05 17:13:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"63213f0b0fefbe4da56c51ed25d7795d902a0b0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-03 14:49:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00b27f110da6ee9a9d08b05f7b4e318c7aae63a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-03 15:02:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"baf3589ea66dd6e886c772595825a30fa5bc65a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-03 15:08:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"03ff79b92df6aef97dbe930904e37ce1aac59cb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-03 15:09:06.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"204a79fbae59ada792e3ebe0a6567ce4f1778a0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-03 15:24:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6cb352ece77a3836c6ba2235c6b87148a74081b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-03 15:49:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2658bac660ad45695bd523a1e8dbe1b790c54faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 17:43:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/043a942aa5a543afadb0cf2d71eff902 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf431cf437f3409da3f6e775a22ede3d : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6f937a1a21f4f949fdf6b2c5db39ae1 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ae889e579df4530a191ba307d275088 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f55df1701e6412f8c134b13ab66e744 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a248de0ea4354e4c955f6a6029504606 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe508f3e2c7f404b904fe7d017f005a3 : SUCCESS in 4m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef236837e115428eba583e8236bc2a37 : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8e22a4a414ca4e419b28d9c38e169d6b : SUCCESS in 12m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d1ebe9ec86844809baed692839a898bb : SUCCESS in 31m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/14afcb46b099479c9e8880efe9444c97 : FAILURE in 26m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/652125a78098459ab0a41239cc246d09 : FAILURE in 22m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a39e72b7092f43ec8263f1806d79017f : FAILURE in 27m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bad11c33780347c9bba941b5037be79c : SUCCESS in 19m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/653064c4fba14dc7b141598cb60eff93 : SUCCESS in 13m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/29474865b78148f4ab176a259f21a40b : SUCCESS in 1h 46m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/04baa71ebba3440bb11d8e353a3654fd : SUCCESS in 35m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22d6fa9d89e84c519faf588b0fa537af : SUCCESS in 28m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1b64ab6a6e3b4253b1bb86c6e84286f4 : SUCCESS in 29m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dd276f756962488fbb8f68040e5e4483 : SUCCESS in 15m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f289d6cf01134922be13304a1257622b : SUCCESS in 11m 35s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":6},{"id":"44d196de1e98866db5829609aab1dedf42363e1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-03 19:19:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0cc0ebd2fe56056d9c2aea80009246121df9bc3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-03 19:21:17.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"e46aa2b9b83e5f4c338d94da5eb30d434b7ebc67","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-03 21:10:12.000000000","message":"Patch Set 8: Review-Priority+2","accounts_in_message":[],"_revision_number":8},{"id":"14e296341b2ab559601767cf34d68267ef8ec6a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-03 21:10:56.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"3e8a86ac8599c4cc53560777312e6d8821d16e4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 22:57:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcf2bcb33dae4981a19c464e1dd5c645 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6b7c96892524aa3ab9ce78950d3fdc3 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f471a85ce05e4532bbdf928806eb4fab : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f366696d7ef4f3b8f378d3ca3cbc871 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcb924ad2f0f43cbb84d4244e6c8b1c9 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8136e49e43b84fc994f61432b9759daf : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69a64755a3d14e3cb2c6993703ad8b57 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff8f8de7f1144702962e4325c308b451 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e6cee1c100be4a15a1bc876973465b4f : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bdb89c58e44141a6b180cc60087029d6 : SUCCESS in 30m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/dff0ca152b574fc29690d28b5c4bb2f7 : SUCCESS in 58m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb51c88758034eddaa34c141c5b6650e : SUCCESS in 43m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6ef862de53ac4f5891f87c0b86a60f70 : SUCCESS in 1h 00m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c2a58f05208f4dc6bbb497b6c28767ba : SUCCESS in 17m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cdc9b57c4b34458b6ae6bb1510c3938 : SUCCESS in 12m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0e8d56ee6b5247f88b8f98a2d911144b : SUCCESS in 1h 42m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e783595b12eb48a99c78a5492ca9e0eb : SUCCESS in 31m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b15a43e0b183464b871bb2861150cba3 : SUCCESS in 29m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e78a233b3ec14d4783be73c625466240 : SUCCESS in 25m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/86bfdf104b3e45eb967c6c49fc903d40 : SUCCESS in 13m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1ed3b1cbcf14d39855ba464a72361c3 : SUCCESS in 10m 38s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":9},{"id":"3eba1f9520b11b606d9cda9e0162b59a5b0c10a0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-05 17:13:27.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nThanks","accounts_in_message":[],"_revision_number":9},{"id":"e3949e283e8ab6eaae0e3a3e1cf4b297892a2d91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 17:13:51.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"65f9fbd11fce38761fca6308eb826af7990059e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 18:43:09.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cefbcf6ecb6436c88a5381acb3ee40c : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fae9d1f6be66455e8a0193b095c0114b : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b69f507924fc43b78438b1f9d56d9e2c : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f9589497604713bcd656ee231b453b : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/862aed9ffb0647dca94fa5d267f67c83 : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2129ab6b698546beb7f98a98256a9d95 : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4491fea3297644bca9707b4cfe14e0f6 : SUCCESS in 9m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a1deafc5aa044f21b21708ec8a46fb8b : SUCCESS in 28m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f346436dc8b14a4192764ed7fc2b4021 : SUCCESS in 50m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5da84723ca44b3a9960e1c11d9118a2 : SUCCESS in 39m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5f3e2dab663f4281ac5a39ea81faea30 : SUCCESS in 54m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1fab668acb3244109e73ee22a89d3439 : SUCCESS in 15m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dec3d2c258c94115aa76532d64e344df : SUCCESS in 11m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a6d40904d4334690a83339a248d53241 : SUCCESS in 1h 27m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/75fa8971ab864f28a6f932d3c92f0979 : SUCCESS in 30m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6be74e47ae1a4636abeffb6742a30b4b : SUCCESS in 23m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d35b70e5eafc46a6bcd7df4f2b325fc9 : SUCCESS in 22m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/23e164b3d36a413f87633796a584cc74 : SUCCESS in 10m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0e4a58242384cc889e09e6ca2d9e26a : SUCCESS in 8m 41s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":9},{"id":"f8735fb3679311f8f9df0753f1f3b223ec6d1bcb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 18:43:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"b28098e0efb7605f5ffa1aaf1b7cbc4a7b610d7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 18:44:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98373ebef7994622abbaa09b2c78a756 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f3bb2228bde478ca3640fa524653e64 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"74eff837cb0870eae78b71bdefe6bfeb6634d6ee","revisions":{"4a6ee7d775b24af78523d7360b214cf0866f80e0":{"kind":"REWORK","_number":1,"created":"2021-09-03 14:49:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/807345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/807345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/807345/1"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 14:47:52.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 14:47:52.000000000","tz":120},"subject":"CI: Fix tox jobs timing out","message":"CI: Fix tox jobs timing out\n\nTox jobs are timing out in CI due to pip dependency resolution\nbacktracking taking too long. When run locally, pip fails to resolve\ndependencies due to upper-constraints now allowing pluggy 1.0.0 [1],\nwhile molecule still depends on pluggy \u003e\u003d 0.7.1, \u003c 1.0. See similar\ntripleo bug for more details [1].\n\nSplit molecule dependencies in a separate requirements file and disable\nits CI job for now. This allows other tox jobs to complete. As a bonus,\nit should speed up execution of the jobs.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/806680\n[2] https://bugs.launchpad.net/tripleo/+bug/1942508\n\nChange-Id: I4cad83c92c32d1db2a5d3c03a78a1c533a8b9633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a6ee7d775b24af78523d7360b214cf0866f80e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a6ee7d775b24af78523d7360b214cf0866f80e0"}]},"branch":"refs/heads/master"},"ccf7ef60c0351b611c3ec8fc69a72da9344f8ed2":{"kind":"REWORK","_number":2,"created":"2021-09-03 15:02:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/807345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/807345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/807345/2"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 15:01:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 15:01:49.000000000","tz":120},"subject":"CI: Fix tox jobs timing out","message":"CI: Fix tox jobs timing out\n\nTox jobs are timing out in CI due to pip dependency resolution\nbacktracking taking too long. When run locally, pip fails to resolve\ndependencies due to upper-constraints now allowing pluggy 1.0.0 [1],\nwhile molecule still depends on pluggy \u003e\u003d 0.7.1, \u003c 1.0. See similar\ntripleo bug for more details [1].\n\nSplit molecule dependencies in a separate requirements file and remove\nupper constraints from its tox environment for now. This allows other\ntox jobs to complete. As a bonus, it should speed up execution of the\njobs due to installing fewer dependencies.\n\nThis also fixes the other CI issue caused by docker 5.0.1 [3].\n\n[1] https://review.opendev.org/c/openstack/requirements/+/806680\n[2] https://bugs.launchpad.net/tripleo/+bug/1942508\n[3] https://review.opendev.org/c/openstack/kayobe/+/807128\n\nChange-Id: I4cad83c92c32d1db2a5d3c03a78a1c533a8b9633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ccf7ef60c0351b611c3ec8fc69a72da9344f8ed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ccf7ef60c0351b611c3ec8fc69a72da9344f8ed2"}]},"branch":"refs/heads/master"},"34d3d589e00179bc0570930c02688d0cc32db933":{"kind":"REWORK","_number":3,"created":"2021-09-03 15:08:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/807345/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/807345/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/807345/3"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 15:05:42.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 15:05:42.000000000","tz":120},"subject":"CI: Fix tox jobs timing out","message":"CI: Fix tox jobs timing out\n\nTox jobs are timing out in CI due to pip dependency resolution\nbacktracking taking too long. When run locally, pip fails to resolve\ndependencies due to upper-constraints now allowing pluggy 1.0.0 [1],\nwhile molecule still depends on pluggy \u003e\u003d 0.7.1, \u003c 1.0. See similar\ntripleo bug for more details [1].\n\nSplit molecule dependencies in a separate requirements file and remove\nupper constraints from its tox environment for now. This allows other\ntox jobs to complete. As a bonus, it should speed up execution of the\njobs due to installing fewer dependencies.\n\nThis also fixes the other CI issue caused by docker 5.0.1 [3].\n\n[1] https://review.opendev.org/c/openstack/requirements/+/806680\n[2] https://bugs.launchpad.net/tripleo/+bug/1942508\n[3] https://review.opendev.org/c/openstack/kayobe/+/807128\n\nChange-Id: I4cad83c92c32d1db2a5d3c03a78a1c533a8b9633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34d3d589e00179bc0570930c02688d0cc32db933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34d3d589e00179bc0570930c02688d0cc32db933"}]},"branch":"refs/heads/master"},"2237ea0bb4a1209ea823766513daf9c7486e4753":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-09-03 15:09:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/807345/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/807345/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/807345/4"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 15:08:53.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 15:08:53.000000000","tz":120},"subject":"CI: Fix tox jobs timing out","message":"CI: Fix tox jobs timing out\n\nTox jobs are timing out in CI due to pip dependency resolution\nbacktracking taking too long. When run locally, pip fails to resolve\ndependencies due to upper-constraints now allowing pluggy 1.0.0 [1],\nwhile molecule still depends on pluggy \u003e\u003d 0.7.1, \u003c 1.0. See similar\ntripleo bug for more details [1].\n\nSplit molecule dependencies in a separate requirements file and remove\nupper constraints from its tox environment for now. As a bonus, it\nshould speed up execution of the jobs due to installing fewer\ndependencies.\n\nThis also fixes the other CI issue caused by docker 5.0.1 [3].\n\n[1] https://review.opendev.org/c/openstack/requirements/+/806680\n[2] https://bugs.launchpad.net/tripleo/+bug/1942508\n[3] https://review.opendev.org/c/openstack/kayobe/+/807128\n\nChange-Id: I4cad83c92c32d1db2a5d3c03a78a1c533a8b9633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2237ea0bb4a1209ea823766513daf9c7486e4753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2237ea0bb4a1209ea823766513daf9c7486e4753"}]},"branch":"refs/heads/master"},"bf8cf6494c29fa66c3414073068a88d6a4febd70":{"kind":"REWORK","_number":5,"created":"2021-09-03 15:24:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/807345/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/807345/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/807345/5"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 15:24:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 15:24:47.000000000","tz":120},"subject":"CI: Fix tox jobs timing out","message":"CI: Fix tox jobs timing out\n\nTox jobs are timing out in CI due to pip dependency resolution\nbacktracking taking too long. When run locally, pip fails to resolve\ndependencies due to upper-constraints now allowing pluggy 1.0.0 [1],\nwhile molecule still depends on pluggy \u003e\u003d 0.7.1, \u003c 1.0. See similar\ntripleo bug for more details [1].\n\nSplit molecule dependencies in a separate requirements file and remove\nupper constraints from its tox environment for now. As a bonus, it\nshould speed up execution of the jobs due to installing fewer\ndependencies.\n\nThis also fixes the other CI issue caused by docker 5.0.1 [3].\n\n[1] https://review.opendev.org/c/openstack/requirements/+/806680\n[2] https://bugs.launchpad.net/tripleo/+bug/1942508\n[3] https://review.opendev.org/c/openstack/kayobe/+/807128\n\nChange-Id: I4cad83c92c32d1db2a5d3c03a78a1c533a8b9633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf8cf6494c29fa66c3414073068a88d6a4febd70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf8cf6494c29fa66c3414073068a88d6a4febd70"}]},"branch":"refs/heads/master"},"3d7f914d78f8899723e4a273cbc9ef91d056dd73":{"kind":"REWORK","_number":6,"created":"2021-09-03 15:49:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/807345/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/807345/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/807345/6"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 15:48:40.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 15:48:40.000000000","tz":120},"subject":"CI: Fix tox jobs timing out","message":"CI: Fix tox jobs timing out\n\nTox jobs are timing out in CI due to pip dependency resolution\nbacktracking taking too long. When run locally, pip fails to resolve\ndependencies due to upper-constraints now allowing pluggy 1.0.0 [1],\nwhile molecule still depends on pluggy \u003e\u003d 0.7.1, \u003c 1.0. See similar\ntripleo bug for more details [1].\n\nSplit molecule dependencies in a separate requirements file and remove\nupper constraints from its tox environment for now. As a bonus, it\nshould speed up execution of the jobs due to installing fewer\ndependencies.\n\nThis also includes the fix for the CI issue caused by docker 5.0.1 [3].\n\n[1] https://review.opendev.org/c/openstack/requirements/+/806680\n[2] https://bugs.launchpad.net/tripleo/+bug/1942508\n[3] https://review.opendev.org/c/openstack/kayobe/+/807128\n\nChange-Id: I4cad83c92c32d1db2a5d3c03a78a1c533a8b9633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d7f914d78f8899723e4a273cbc9ef91d056dd73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d7f914d78f8899723e4a273cbc9ef91d056dd73"}]},"branch":"refs/heads/master"},"5224f46fd977e11d897724445aebe44c9742e8f2":{"kind":"REWORK","_number":7,"created":"2021-09-03 19:19:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/807345/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/807345/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/807345/7"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 19:17:31.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 19:17:31.000000000","tz":120},"subject":"CI: Fix multiple failures","message":"CI: Fix multiple failures\n\nTox jobs are timing out in CI due to pip dependency resolution\nbacktracking taking too long. When run locally, pip fails to resolve\ndependencies due to upper-constraints now allowing pluggy 1.0.0 [1],\nwhile molecule still depends on pluggy \u003e\u003d 0.7.1, \u003c 1.0. See similar\ntripleo bug for more details [1].\n\nSplit molecule dependencies in a separate requirements file and remove\nupper constraints from its tox environment for now. As a bonus, it\nshould speed up execution of the jobs due to installing fewer\ndependencies.\n\nStop using upper-constraints when installing docker python library [3].\nThis part of the commit can be reverted when docker 5.0.2 is in upper\nconstraints.\n\nUpdate Kolla inventory template following renaming of haproxy role to\nloadbalancer [4]. With the existing template, haproxy/keepalived\ncontainers are not created and service deployment fails while trying to\nconnect to MySQL using the VIP.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/806680\n[2] https://bugs.launchpad.net/tripleo/+bug/1942508\n[3] https://review.opendev.org/c/openstack/kayobe/+/807128\n[4] https://review.opendev.org/c/openstack/kolla-ansible/+/770618\n\nChange-Id: I4cad83c92c32d1db2a5d3c03a78a1c533a8b9633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5224f46fd977e11d897724445aebe44c9742e8f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5224f46fd977e11d897724445aebe44c9742e8f2"}]},"branch":"refs/heads/master"},"f2bdee9a3edf9bd5a4de8f5b69880e17fa14e4df":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-09-03 19:21:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/807345/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/807345/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/807345/8"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 19:20:59.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 19:20:59.000000000","tz":120},"subject":"CI: Fix multiple failures","message":"CI: Fix multiple failures\n\nTox jobs are timing out in CI due to pip dependency resolution\nbacktracking taking too long. When run locally, pip fails to resolve\ndependencies due to upper-constraints now allowing pluggy 1.0.0 [1],\nwhile molecule still depends on pluggy \u003e\u003d 0.7.1, \u003c 1.0. See similar\ntripleo bug for more details [1].\n\nSplit molecule dependencies in a separate requirements file and remove\nupper constraints from its tox environment for now. As a bonus, it\nshould speed up execution of the jobs due to installing fewer\ndependencies.\n\nStop using upper-constraints when installing docker python library [3].\nThis part of the commit can be reverted when docker 5.0.2 is in upper\nconstraints.\n\nUpdate Kolla inventory template following renaming of haproxy role to\nloadbalancer [4]. With the existing template, haproxy/keepalived\ncontainers are not created and service deployment fails while trying to\nconnect to MySQL using the VIP.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/806680\n[2] https://bugs.launchpad.net/tripleo/+bug/1942508\n[3] https://review.opendev.org/c/openstack/kayobe/+/807128\n[4] https://review.opendev.org/c/openstack/kolla-ansible/+/770618\n\nChange-Id: I4cad83c92c32d1db2a5d3c03a78a1c533a8b9633\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2bdee9a3edf9bd5a4de8f5b69880e17fa14e4df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2bdee9a3edf9bd5a4de8f5b69880e17fa14e4df"}]},"branch":"refs/heads/master"},"74eff837cb0870eae78b71bdefe6bfeb6634d6ee":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2021-09-03 21:10:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/807345/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/807345/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/807345/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/807345/9"}}},"commit":{"parents":[{"commit":"bb05adbdcfeca852085573ca257dadf607ff1905","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb05adbdcfeca852085573ca257dadf607ff1905"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 21:10:39.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-03 21:10:39.000000000","tz":120},"subject":"CI: Fix multiple failures","message":"CI: Fix multiple failures\n\nTox jobs are timing out in CI due to pip dependency resolution\nbacktracking taking too long. When run locally, pip fails to resolve\ndependencies due to upper-constraints now allowing pluggy 1.0.0 [1],\nwhile molecule still depends on pluggy \u003e\u003d 0.7.1, \u003c 1.0. See similar\ntripleo bug for more details [2].\n\nSplit molecule dependencies in a separate requirements file and remove\nupper constraints from its tox environment for now. As a bonus, it\nshould speed up execution of the jobs due to installing fewer\ndependencies.\n\nStop using upper-constraints when installing docker python library [3].\nThis part of the commit can be reverted when docker 5.0.2 is in upper\nconstraints.\n\nUpdate Kolla inventory template following renaming of haproxy role to\nloadbalancer [4]. With the existing template, haproxy/keepalived\ncontainers are not created and service deployment fails while trying to\nconnect to MySQL using the VIP.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/806680\n[2] https://bugs.launchpad.net/tripleo/+bug/1942508\n[3] https://review.opendev.org/c/openstack/kayobe/+/807128\n[4] https://review.opendev.org/c/openstack/kolla-ansible/+/770618\n\nChange-Id: I4cad83c92c32d1db2a5d3c03a78a1c533a8b9633\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74eff837cb0870eae78b71bdefe6bfeb6634d6ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74eff837cb0870eae78b71bdefe6bfeb6634d6ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
