)]}'
{"id":"openstack%2Fkayobe~807864","triplet_id":"openstack%2Fkayobe~master~I4c72e32e5379237340284a09874b0c500e41ad0f","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I4c72e32e5379237340284a09874b0c500e41ad0f","subject":"Allow passing arguments to dev/environment-setup.sh","status":"MERGED","created":"2021-09-08 12:25:59.000000000","updated":"2021-09-09 14:42:52.000000000","submitted":"2021-09-09 14:38:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807864","meta_rev_id":"4c411ab47e47299f606aae2a77f6b69ee9cf5745","_number":807864,"virtual_id_number":807864,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-09 14:38:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-09 11:24:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-09-09 11:24:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-08 14:24:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-09-08 14:38:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 11:24:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"89b7b291a1be5e738c21a43cc4fe2cc695a70900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-08 12:25:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc750ce1f916e770c8d15c6754127cae2c7e6a06","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-08 14:24:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e7e65099bf963ab2f97f29c7aee9c3c021e8bcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 14:38:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c741d3b7bfc7417182c72d8be90bda43 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1e044a6f9664083bf48aab4e6a9739e : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fc0da8a7b7e4e80b5cd51500394563f : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9ebd69444144c67b2dbf29982594a07 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0beed68baf24fa295aa11a628513943 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12829c33249d403095263743048314a2 : SUCCESS in 9m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d925a28db4124ecfbb048a489c3cb4b8 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/44aa8b8104ff48f78503a0bd5da1cf9c : SUCCESS in 10m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0ee19dad26c14351b48b97909a4fdc3f : SUCCESS in 29m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bc1c2789ac1347e2b207f30dbc773685 : SUCCESS in 1h 00m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/018c5a7879fa41648cc189615a6d0984 : SUCCESS in 44m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2bb942e2bc1a46eda4d97fd6b90d12a8 : SUCCESS in 1h 03m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ec834cade5854fb28e398e66a4c34496 : SUCCESS in 16m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83caaa77357843a9b6ec6d920a0c412d : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ade1862148534e6a8fe8767bcdae2878 : SUCCESS in 2h 09m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/66c96e0e93cb4033916a0a0a4dc66df0 : SUCCESS in 36m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c11d52120216407d8d4937b9e198125a : SUCCESS in 31m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f6142484362046c68396b764301488ef : SUCCESS in 26m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e2a7bec95ef04962856deee99d326901 : SUCCESS in 11m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e310b4defab44708cbef8786f90ad78 : SUCCESS in 11m 05s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"92f3a928a9cff5cd54b78b091bed454bbf3c57f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-08 14:43:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa8b1e5b86e79821b31f4821b948511bb38df813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 16:40:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bce6c51d278843f0a163f227d276eafa : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad29abd57c754ae7beaee68ea3f60011 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/867a847b363c4d74b98ac3c33aefe4f1 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7b7af3935c84ec598f3ff3c0e63c1d1 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f988b79d598a4e098cf678cc11d377ba : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb66173a2c34b8fbc58e732a7e6070c : SUCCESS in 9m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8e601a44af64bc3bb67377c3f6005bd : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/58789a06fedf4c539708bda04e55171b : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fcfe62e951674178afc8b5e41200331a : SUCCESS in 27m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d87b0485ad19436886dc19f307cf86ad : SUCCESS in 1h 00m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18a3622509434ec1807f87b5f8740274 : SUCCESS in 46m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2d1310e60ca041569c4e9445da16d98b : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ee77600cc95f4aa59b577c2ffa4ca318 : SUCCESS in 21m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a14f4afeae174aeca3dd198fbbcb45f2 : SUCCESS in 14m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/87e07fe6ec894bfe8faff58f8415a6df : SUCCESS in 1h 52m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f70a6ccf666d4c6eb54905e6b47976d4 : SUCCESS in 35m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daa25ddddfbe4292a98a3a90b4da7c85 : SUCCESS in 25m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ec1aa78c4cca48bfa0e606c95ecda662 : SUCCESS in 26m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/37301fb405d94ca4b6d8e9f9ec91f8d5 : SUCCESS in 15m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6a4970f609942c2b8d8b0691029afd1 : SUCCESS in 10m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"99d4a09027979eade0ff5a122a68d6b423972e21","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-09 11:24:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"73fd54504cb113db63bf5d0118e8ff5160d1ac56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 11:25:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"17356e28f5e3853c897e9e585d3257d85556f816","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 14:38:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f76d4793c5374841914fe32e5ffc6cc3 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67efa404fc864bd68d539c87c05badb9 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d9f75771e4f4c49a15bc93846d3ce14 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd8864996a964f2e8bcb878cbc26a240 : SUCCESS in 9m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28b25855868940d2b4b18c13870db72e : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa6d5353e20e4bf2b217e8e4787c35ca : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/491c7d9d090c44b9b624cdc8dc9b627b : SUCCESS in 32m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c8f2af2c19db42ef9c98ea1afe3d6cf8 : SUCCESS in 56m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebc5755c3c6f424b86c12d8cc0460f8f : SUCCESS in 44m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fc89b82131b942efbd6b7d75a0b0b92c : SUCCESS in 1h 04m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bd259d1d06524419a737043545851a8d : SUCCESS in 16m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/989d6c0fc66b4342b2528ff6eec2fc00 : SUCCESS in 13m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/95bfce82667742ae8df4f055ea1d9b58 : SUCCESS in 1h 46m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/482c881166c34291a069ba29e2fed9a2 : SUCCESS in 33m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1751d155529246d39e5e151294a2878d : SUCCESS in 27m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2f55be5d21c6459fb073da72aa8f33e5 : SUCCESS in 25m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b4a5141c74864b8681b4cdf17c27a8d2 : SUCCESS in 12m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ff411a618cb41e68dadbd5d88af8b48 : SUCCESS in 10m 05s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"652f0dbf0b89e1593ee381aadf1c5ee30c6f9098","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 14:38:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4c411ab47e47299f606aae2a77f6b69ee9cf5745","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 14:42:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d119f934ce4ddb84354c556f455751 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b16300c8f1af73809f2d18cd75a172c562ad4d1","revisions":{"539d7127b12ee640c83e524609884c01f2a28c12":{"kind":"REWORK","_number":1,"created":"2021-09-08 12:25:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/807864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/807864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/807864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/807864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/807864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/807864/1"}}},"commit":{"parents":[{"commit":"e8769373b6adace69b041a18a4342870a98eaabf","subject":"Fix typo in IP allocation filename","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8769373b6adace69b041a18a4342870a98eaabf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-07 07:44:35.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-08 12:25:47.000000000","tz":60},"subject":"Allow passing arguments to dev/environment-setup.sh","message":"Allow passing arguments to dev/environment-setup.sh\n\nArguments are passed through to kayobe-env in kayobe-config, which\nallows to set the Kayobe environment.\n\nChange-Id: I4c72e32e5379237340284a09874b0c500e41ad0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/539d7127b12ee640c83e524609884c01f2a28c12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/539d7127b12ee640c83e524609884c01f2a28c12"}]},"branch":"refs/heads/master"},"7b16300c8f1af73809f2d18cd75a172c562ad4d1":{"kind":"REWORK","_number":2,"created":"2021-09-08 14:43:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/807864/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/807864/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/807864/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/807864/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/807864/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/807864/2"}}},"commit":{"parents":[{"commit":"e8769373b6adace69b041a18a4342870a98eaabf","subject":"Fix typo in IP allocation filename","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8769373b6adace69b041a18a4342870a98eaabf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-07 07:44:35.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-08 14:43:30.000000000","tz":60},"subject":"Allow passing arguments to dev/environment-setup.sh","message":"Allow passing arguments to dev/environment-setup.sh\n\nArguments are passed through to kayobe-env in kayobe-config, which\nallows to set the Kayobe environment.\n\nChange-Id: I4c72e32e5379237340284a09874b0c500e41ad0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b16300c8f1af73809f2d18cd75a172c562ad4d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b16300c8f1af73809f2d18cd75a172c562ad4d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
