)]}'
{"id":"openstack%2Fkayobe~808029","triplet_id":"openstack%2Fkayobe~master~I45e937ec7c102e89edee20679b8959bf7f6c7152","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I45e937ec7c102e89edee20679b8959bf7f6c7152","subject":"CI: Fix get_logs.sh docker logs output","status":"MERGED","created":"2021-09-09 09:32:49.000000000","updated":"2021-09-10 09:45:59.000000000","submitted":"2021-09-10 09:43:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808029","meta_rev_id":"8f32fdc4a24c8f89dc1692712ae340f334a795dc","_number":808029,"virtual_id_number":808029,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 09:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2021-09-09 10:33:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-09-10 07:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-09 16:04:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2021-09-10 07:48:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-09 10:33:43.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2021-09-09 11:20:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 16:04:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-09-10 07:48:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"71da9f75a0dcdf3d6ac68f0311017b1700ab4553","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-09 09:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c8a0234088425bfb37cf9f8126b95025a715c4e","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-09-09 10:33:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"05cacc826a1ee57cfedc83f88e1c49f2b1cb6d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 11:20:33.000000000","message":"Patch Set 1: Verified+1\n\n(19 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a40c69a2ca014ceb88a709ad26639c2f : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1a87c1c632483d8ebf6ab969483c5d : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a03f0e231f448489a773d42ec7032eb : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94e436e001f04b869a4970c62a51cd0d : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/933ace8440514f9fa854cd1d49917a8b : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c7a797a84b64cffacb9fcc1ae3648e6 : SUCCESS in 9m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a8b5f886cc9d4e4b8eb635137352f1c4 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb528a231d144ebca66e5219b32f5020 : SUCCESS in 11m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba6f2d3fbee140bba6e2ea93215e6390 : SUCCESS in 37m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4968ba9608be472fbc87911d0614f0ed : SUCCESS in 1h 18m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/258bb077b5464758a0af52860bc80e29 : SUCCESS in 44m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f9de5f583f6f477b8a4e5564163471fd : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5384b425e1594449be86016b8cb891fb : SUCCESS in 17m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50dbc7b84f47428cb3865c17e5bead41 : SUCCESS in 14m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4a3ae6f4effc411a9f0914821e2b6bec : SUCCESS in 1h 46m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e4749a7f2824488397f9514582e36814 : SUCCESS in 50m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f2d1faf4b454543a10882a547483b9c : SUCCESS in 27m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/32718b289a32480b9e09f576cef496c8 : SUCCESS in 24m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b37d87150341492490455d83f405c05a : SUCCESS in 12m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf1994c2dbb24449aee55175d5390b43 : SUCCESS in 12m 08s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"d11ac200e4f523e9feb02accca7568bc6329ab3d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-09 16:04:45.000000000","message":"Patch Set 1: Code-Review+2\n\nVerified by comparing with logs from a recent CI build.","accounts_in_message":[],"_revision_number":1},{"id":"4eba71c4441a2e39d991005f7834ff1f1dc6cd96","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-10 07:48:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44127274daca2f8f4a89a056616306e1a2e4ba59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 07:48:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9619514b82cfe67255ab648f90226a7a5b9778be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 09:43:26.000000000","message":"Patch Set 1: Verified+2\n\n(19 comments)\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe65f12daa54b10af5baa9986ad9ef3 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/281e68379d9147a9b1fa7f7989ca9190 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3af4e55f78f4852be29125cf6243f6e : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc2154cf6e264153967ca46d32402e2a : SUCCESS in 9m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8fb5433c005f4ee1bcd543ef70017c7a : SUCCESS in 7m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/38d3d51fc70c49df9376075e69c6e3c0 : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/10e49432e5944cba854ee64244fba5da : SUCCESS in 31m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a0d52ede11544e0194b9f5126fce321f : SUCCESS in 1h 00m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/772d40e48d4c4e4fba7822af02e7dabc : SUCCESS in 44m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7ceede18ea2547ec9333a45e54f65d7b : SUCCESS in 1h 05m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6b6016a2b5924b11afd729db4969c7a9 : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4db5ba492abe4c95b02502d121543ddb : SUCCESS in 16m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cbf302f32922492d977b28fb95e9c0e5 : SUCCESS in 1h 53m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bb0ba0fdc84644f387ee830f75f5ec72 : SUCCESS in 32m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2e60c5e153443428630177732622b1e : SUCCESS in 33m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9581cf2a4fdd42c9b2d9786d7f8f8770 : SUCCESS in 24m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/81a57c9a0f46481cae08f888582546c1 : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/735b0e0ea0a745b69ef28b4fac14a53d : SUCCESS in 11m 19s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"365c36f8bdf0b226653000ecfc994e1ce5fa9cec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 09:43:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8f32fdc4a24c8f89dc1692712ae340f334a795dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 09:45:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ffe7b70c7cb4ebdad572f111639f340 : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"923813ac4820044723814b777e4a39cb6eb273fe","revisions":{"923813ac4820044723814b777e4a39cb6eb273fe":{"kind":"REWORK","_number":1,"created":"2021-09-09 09:32:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/29/808029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/808029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/808029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/808029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/808029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/808029/1"}}},"commit":{"parents":[{"commit":"e8769373b6adace69b041a18a4342870a98eaabf","subject":"Fix typo in IP allocation filename","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8769373b6adace69b041a18a4342870a98eaabf"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-09 09:30:38.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-09 09:32:45.000000000","tz":120},"subject":"CI: Fix get_logs.sh docker logs output","message":"CI: Fix get_logs.sh docker logs output\n\nCurrently the logs get printed in Zuul job stdout and only a small portion\nof them is redirected to a file - this patch changes that to send all output\nto desired files.\n\nChange-Id: I45e937ec7c102e89edee20679b8959bf7f6c7152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/923813ac4820044723814b777e4a39cb6eb273fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/923813ac4820044723814b777e4a39cb6eb273fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
