)]}'
{"id":"openstack%2Fkayobe~808111","triplet_id":"openstack%2Fkayobe~master~I6ae44dcae78424b4638762d015cf24336444707f","project":"openstack/kayobe","branch":"master","topic":"bifrost-deploy-kernel","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-10-11 14:07:42.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-10-10 11:43:48.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-10-11 14:07:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ae44dcae78424b4638762d015cf24336444707f","subject":"Add introspection rule to update deploy kernel location","status":"MERGED","created":"2021-09-09 16:43:18.000000000","updated":"2023-10-11 20:21:54.000000000","submitted":"2023-10-11 14:07:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808111-bifrost-deploy-kernel","meta_rev_id":"b3d280862a3243a7281b66da1520e080d5c85b92","_number":808111,"virtual_id_number":808111,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 14:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-10-11 11:30:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-10-11 20:21:54.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-10-10 11:43:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-10-11 11:30:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-09 18:29:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-10 08:14:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-06-10 09:22:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-08-27 00:05:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-10-11 11:30:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"8a1f431b985923eaf63de7695693ad68056fed61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-09 16:43:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"642038d22be65992215b993c2a24cda8f05a5e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 18:29:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/009cd09943694bfdb59a6fcbc73ac1b4 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b82ab59d1818420299bdd2c2b21c6bcf : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e743d22cc18243ca9a5888ae4315f3c5 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e03cb1ab77c64a8d91779a4c9a4436b7 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a282e6856ccf4702b3a9849c62a5a452 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d5dd3d06e804ad496cea4351433cffa : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfd58fb33fac41119740e69e400555e2 : SUCCESS in 4m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/736f8c5c5f4c4a2684e81411c2789323 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9165caa7be4645a38213756338182b0f : SUCCESS in 14m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e566335917714c41a97dcfe8d7fc2400 : SUCCESS in 31m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4d1fdc38588d474d88cca8ca1035a85a : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f51523e6892b49d38559a51d10097fd3 : SUCCESS in 42m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/687f4d0eeb7f45eba6db0ba11ba2146f : SUCCESS in 1h 09m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/02af1d9fd9794ec2b1bb00bc88d9e5c6 : SUCCESS in 17m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1df258279284483987e60d1874a954be : SUCCESS in 13m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c5e9126b48ba474ea699ef9caeb5de3c : SUCCESS in 1h 41m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b8170a35023b4f9997d373d180112244 : SUCCESS in 32m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6ef8cb820af4ba788503d2b7b793128 : SUCCESS in 27m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ceca623260c940fa96cbbd17dff88324 : SUCCESS in 26m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f70aa38edf084f65bf4da73df8a993ac : SUCCESS in 13m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc9ab3c36fb6488f82c0372d6ad25593 : SUCCESS in 12m 47s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"8d2170703b4be6082fd30d63c6c21b5090ec2786","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-10 08:14:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99ba1a4f678992b25c031391af647e39904bf0d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-09 20:20:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1a74493854d309e3a0e08bb3ba834a6f93143503","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-09 20:20:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13897566dad82638e8dee0f2606b71da25781bf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 22:34:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a36a727177d4f9f8fb4513ede741489 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e62ce0c54d4615b7358b2a3673e45b : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09496a2e7b53443f8ddf983bb467748f : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/987f33e1d96044c19b31cde4983968ce : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b34e4a413cc64650bd2cc454258b8252 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a82bf53edf54d83a18330f66f951926 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ac51ff20abc44dbbe605e53b2dd35c9 : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/058b486d4c854c019f188e518d22898e : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0c8699f133494ea59909e986224978f5 : SUCCESS in 13m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a3b1f84722cc4cdda006402d64aa86b8 : SUCCESS in 15m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0158e8bcc474479ca1063af803875e43 : SUCCESS in 1h 44m 12s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f47d0b9ec0d14ec681fa4d1475a79174 : SUCCESS in 1h 08m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47429ca02eff404e9786f1bcff4a8314 : SUCCESS in 1h 29m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/57ee6a8506584966a9b8e6072744fe1f : SUCCESS in 1h 44m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/690841b36df444f7b05df2cfaa35623e : SUCCESS in 20m 36s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9d7aa0ca15b0433cb183a0ce4c89961a : SUCCESS in 21m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0faa21c2ef04b7fbd2cc3b44aafe1d9 : SUCCESS in 13m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ebaf29dfc4f841e992880e45442474d7 : SUCCESS in 1h 51m 47s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63a4e5c3c85446ecb88d38f65978c180 : FAILURE in 1h 23m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a52f51f402644fc8af77b1d76bb5eb01 : SUCCESS in 18m 23s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/990e2da0d23d43e094cd063fef4df6f0 : SUCCESS in 18m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5036c2983484bbfaee9d5b27c64d90f : SUCCESS in 30m 44s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/e775f192e52b43ba931490be64308659 : FAILURE in 1h 09m 00s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d883590b4f9746b38b764fb9926d0820 : SUCCESS in 1h 18m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffe78cdbb6d847a39adc82f4b4ee6eca : SUCCESS in 38m 45s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8118219d475849409449788be2e4c37a : SUCCESS in 24m 31s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/581027ed5df14d8888a2f0dfa38abe04 : SUCCESS in 15m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/233b90d37cdc40f59c295acd32959b64 : SUCCESS in 16m 52s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8b34dace0991400fbd67da043167d4d2 : SUCCESS in 17m 55s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9ee91036a716408fbbfaea9bd665bf2b : SUCCESS in 17m 07s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2477d8d4bfd44209ac64719695baeeed : SUCCESS in 24m 20s","accounts_in_message":[],"_revision_number":2},{"id":"31c12dc1955e4c136a6c5d86e082465abe1ce46d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-09 23:35:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a2e781de7bef9ed1b0bfd18736efc36b59a7d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 01:02:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc44b834372a46d6b35be513628910a0 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c8fbcd0f84f4834b12c134bdd343a1e : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59495632a31848d791881777242770ea : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cb7e23d70b241549a66c6eff750a65f : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b830c6deaf764ce6b5e8e15f7214b40f : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1454e51726d2496e9ace319b6555b36c : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7897bd2f17f1452a8625ccf7837c9959 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c1c408b5ee4347e8b7b0b69308b9a5ab : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/29b8f93aaf2e49beb73632cbb9efa309 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b980af5451e04e2c9cd36d2965c37f67 : SUCCESS in 15m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/14a4964a478948a396ee5ab3d4401af5 : SUCCESS in 54m 45s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/0e0ff5172b3c4254891b6d636429a3c9 : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2651d294f27b4e15aebf02419ac1f7fe : SUCCESS in 48m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/47a89a745d5a42e180c86528067d3778 : SUCCESS in 1h 03m 37s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b550972b851b43709eb15a219f83f0a2 : SUCCESS in 19m 49s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/0c5962e00ab14e358161ded0a1df4929 : SUCCESS in 21m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9327c1da531a44d794912370e6ab84d2 : SUCCESS in 16m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e5b3f811f47f4dea9e91c5905068922a : SUCCESS in 1h 25m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28ceaccb63964c20ad460e8855a7bbdb : SUCCESS in 1h 04m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/07382eb487f045079e38c89ea32b5ba0 : SUCCESS in 19m 33s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f908ce866cef4181955c54442c0c4aaf : SUCCESS in 20m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35d7acbf2c46462a803e9f3b92244403 : SUCCESS in 13m 42s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/1b4f9a9a8adf44bda55a76c9f80b59ac : FAILURE in 29m 51s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8a077cac73244434b89ceeb7207b7c0b : SUCCESS in 39m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/231cbefaa511429fa4451baf7fb0eb9a : SUCCESS in 26m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6b305f031daa4a1497d8ddbe084f9c6d : SUCCESS in 15m 55s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/81541138a9464e85ba5fc16d5617b9fe : SUCCESS in 15m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b7c21df57dd440fbce6dd135079a6ee : SUCCESS in 12m 41s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0738b922d1524ba8ab4bc5792b83a05d : SUCCESS in 17m 02s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2a815594f66a45e39ffe277b69699362 : SUCCESS in 17m 19s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6eecaf95b5b45658fcd816c4b367164 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":2},{"id":"10a0408938e7e496fb1160c02f54d35ec996f151","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-10 09:22:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4935df109e5ae9f909e71c436d84fe047bb53638","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-10 08:46:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63d1fef9ace14304628d02b3b67a492c8cb67a84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-10 11:09:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c40238c5c9bc646eadc100c423846604f5f842ab","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-10 11:09:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3661b04aba7b7de598c0b73f463b34d50d2c2da1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-10-10 11:43:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a1361c69e2f7d6203a56d356b913c4becca5cd12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 13:18:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0beeb0c6d35e4302aef4b5b293b07e05\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c94dfb0f2fe1404bbbd394457c628b79 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a608c22ab4847e6bce52f758a4b2266 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49458dea3ad4479d8b473648dd0c017d : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b05641fffda4235a1f9ece1e9638496 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb76d6e116604059b2f200fb69db9840 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8105698d09f7459ba9ef886114674550 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e073a818d0c472c94a6617c939acd26 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1cb0c1c02b04524af30eb5880823ef3 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5cdb979d23dd4bc8b2e3b3d38b3b7da8 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9640b34144aa4633a6ce8716f30782ed : SUCCESS in 12m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e9016a050842453ebe20e1de69855a96 : SUCCESS in 11m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4ea31a6f4f2a4d48afe51164748d8899 : SUCCESS in 1h 13m 30s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9233234bca3a43e1b6433b47aa074df7 : SUCCESS in 55m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40440b2d22c047fcb522a0b640dfaaf8 : SUCCESS in 41m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4b1f20cdac8d420fb2da520e041fdf94 : SUCCESS in 1h 09m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d434692221a44abc88b84f0fa925b7d7 : SUCCESS in 18m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/47ceb471982843af9642ea186e791067 : SUCCESS in 14m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fb450d276634c09a01f8e185dd4e245 : SUCCESS in 11m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4872cbe266e14e22a406796fa9302d1d : FAILURE in 1h 53m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6620e01b36745a08c30f1049c95436e : SUCCESS in 2h 07m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5e4ea3f9f3c84eb7a45523e19c31d08a : SUCCESS in 15m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a311ef3596e44bd94667a76ce773bc0 : SUCCESS in 14m 28s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1930aa9c7d43458d987f5691c4bf1547 : SUCCESS in 31m 26s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d7568c0d29c64160bedff3f29bc05581 : SUCCESS in 32m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db742799356e4a4fab457dd9732f5934 : SUCCESS in 24m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e159681cbd4349d98824bb40cdad730b : SUCCESS in 12m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edbdd70eec1f46b8b570e3a787da21f5 : SUCCESS in 11m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/50b2e3ff23fe4e259e15e27d09686f1c : SUCCESS in 18m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80142f0539d04eecb49ad80e1b532213 : SUCCESS in 15m 43s","accounts_in_message":[],"_revision_number":3},{"id":"045787ae6eae87a3130868c906424071263a8186","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-10-10 15:57:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea50461e3bd5b6ad519fe234991c56900e8b12a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 18:29:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9197a3563b04c59bb1bae280fa092b7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b21dfa100464e0caacdf1ec984723cf : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/043356bd17e047ec82dd864a8dc1e6f7 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27ff759b99494b348dcbff61ab8368a4 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8e40f69fb864dcba1f9a36e0ce4bedd : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0fd41e0455748c5ab79b6233016ffd7 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/580d3a9604914c209d0f86282f800de0 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c0cce4ca89149f69fc43954ce539ffb : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e2fc05597b84f5d9e22e5a843280f82 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f9abf376812446578140c553617dcfbf : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/79eda0af2fff45dc888b7ba62626a555 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5caf93be51a74d13aa5a74cb966ac61b : SUCCESS in 10m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6e2e806790244c429f1f993a5031e7eb : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7f7e327df7f645a994a0ff2b24c03803 : SUCCESS in 1h 07m 03s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1738c1fffc4c4026a5ddea2135ff48f0 : SUCCESS in 53m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/82ff31407ff14a06b39dceba978f4664 : SUCCESS in 1h 09m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7adc594585554f48a879403895e93b4e : SUCCESS in 16m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/af8f56959ab74159ac8463d16c3ca5fd : SUCCESS in 15m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16db0e4c848e492b96c02be2021aca1c : SUCCESS in 13m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/87a37929acaa45ec81037610f219bc74 : SUCCESS in 2h 26m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e47c7efd0564c438dca4ae51a92bf31 : SUCCESS in 2h 07m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/812beaf53dcf4c54b15ad0132235839b : SUCCESS in 16m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71bc1418637c4b279dfc466a6c46f837 : SUCCESS in 15m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0639df3f07294366998c0d4810df7355 : SUCCESS in 24m 25s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/65149bfc10884fb9b91d789980045138 : SUCCESS in 27m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb9b6bb8d2284964b10a82e218a2e3d8 : SUCCESS in 22m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5728c682ad2543a999b52fac3b83bf04 : SUCCESS in 11m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/305c2469a6f24a9f90335810bb43588d : SUCCESS in 12m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3824a815ece044e7add53b70d48af76b : SUCCESS in 18m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e68f02f0b104253a23c61e576d921ce : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":3},{"id":"381e7711a5e100da6a1e34c6ce0b9b6785d4ee13","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-11 11:30:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4b13036b1f368800fa215148f308e8c163a2b053","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:31:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7d5c8548b6e3af84ce4cd6d300f432e6dffff05b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:07:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e59f305a7d04d788870bf3c9044f0c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e37d9c583c34af0b3b6bf49f52fcad4 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9162734c71e42ddb7bffec1122587bc : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4911ccc09c7f4c038af317b909520808 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb94e486653746f08d222ff15cf527e6 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7875e89d55394f8bb91a4ea7d849503c : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/335ec41173744ab2bf23d0cda2c1158b : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/682d8f47c12a4a01ac00e3162cc1d313 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a042099ebb414a3dbcfed0e66e8bb670 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c0de168ad1974c1c8f935d3bdc88434a : SUCCESS in 10m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f45f2ea7b61f440084ca14458d39475b : SUCCESS in 52m 15s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/574484bb129a4209964350b46ad4914d : SUCCESS in 56m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cffa98c19f7b4289ac86359d1111b5b8 : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/833f005544734765bd2548b6cdcfe814 : SUCCESS in 16m 05s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e57dba8920d41e9ae6bc836668f037d : SUCCESS in 14m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6a3f716797994b809751256f32ff6678 : SUCCESS in 2h 17m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7901dff3b5c443bf9756f2297fdb59b3 : SUCCESS in 2h 29m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6d12d83dbb794146b8c93017ca6b7b69 : SUCCESS in 16m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52288f3bd39641e2a3228c668c963caf : SUCCESS in 12m 31s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/af92a45cc6264457b7eb590cc267387b : SUCCESS in 29m 57s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4450e6e1a1d94b04a400aff8d2d2110d : SUCCESS in 22m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1fdba4662a1643d5b4c15ce60c6ad62e : SUCCESS in 14m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fe31cd2d26e425296a4cf87ee50bc75 : SUCCESS in 9m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6627a6fc7dc34a28aef86f85d69e0b34 : SUCCESS in 15m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/830fc82ab4ef483dbcd99c008baf6fb7 : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":3},{"id":"c03b6fe116dcd8acefa2923a5fbe8c058062869e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:07:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f57607c48e7597030f787191800c5d48c96ed712","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:08:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71385662e06445d8803adf1dfde0a80f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e371c2c9cec94c72b42431baff2beac8 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60557b2dc0a8473bb5089e8329b72387 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3},{"id":"b3d280862a3243a7281b66da1520e080d5c85b92","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-10-11 20:21:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"44af704836799e9b211e3eccfa5efa72f9713f32","revisions":{"d1e494fc0e0b19ff53d0975e5a45095569e03af1":{"kind":"REWORK","_number":1,"created":"2021-09-09 16:43:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/808111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/808111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/808111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/808111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/808111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/808111/1"}}},"commit":{"parents":[{"commit":"4163f0f5c0871328d90517493de3a6f9ee64db71","subject":"Merge \"Allow passing arguments to dev/environment-setup.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4163f0f5c0871328d90517493de3a6f9ee64db71"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-09 16:43:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-09 16:43:02.000000000","tz":120},"subject":"Add introspection rule to update deploy kernel location","message":"Add introspection rule to update deploy kernel location\n\nBack in the Ussuri release, we changed the Bifrost kernel location to\nuse ipa.kernel instead of ipa.vmlinuz. While this works fine for newly\ndiscovered nodes, any node added to Bifrost in Train or earlier will\nhave kept ipa.vmlinuz as its deploy kernel.\n\nThis can cause issues since upgrading Bifrost or building new deployment\nimages will update ipa.kernel, but not ipa.vmlinuz, resulting in nodes\nbooting with an old kernel and a new ramdisk.\n\nFix by adding a new rule updating the legacy kernel location and\ndocumenting how to update node information.\n\nChange-Id: I6ae44dcae78424b4638762d015cf24336444707f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1e494fc0e0b19ff53d0975e5a45095569e03af1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1e494fc0e0b19ff53d0975e5a45095569e03af1"}]},"branch":"refs/heads/master"},"b8457c17345731697d029fb4e9fe0ac9ae9f95ec":{"kind":"REWORK","_number":2,"created":"2022-06-09 20:20:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/808111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/808111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/808111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/808111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/808111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/808111/2"}}},"commit":{"parents":[{"commit":"b77033970a79587b3be5c04a5b81e8b88e21dadf","subject":"Add missing declaration of kolla_enable_hacluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b77033970a79587b3be5c04a5b81e8b88e21dadf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-09 20:19:53.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-09 20:19:53.000000000","tz":120},"subject":"Add introspection rule to update deploy kernel location","message":"Add introspection rule to update deploy kernel location\n\nBack in the Ussuri release, we changed the Bifrost kernel location to\nuse ipa.kernel instead of ipa.vmlinuz. While this works fine for newly\ndiscovered nodes, any node added to Bifrost in Train or earlier will\nhave kept ipa.vmlinuz as its deploy kernel.\n\nThis can cause issues since upgrading Bifrost or building new deployment\nimages will update ipa.kernel, but not ipa.vmlinuz, resulting in nodes\nbooting with an old kernel and a new ramdisk.\n\nFix by adding a new rule updating the legacy kernel location and\ndocumenting how to update node information.\n\nChange-Id: I6ae44dcae78424b4638762d015cf24336444707f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8457c17345731697d029fb4e9fe0ac9ae9f95ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8457c17345731697d029fb4e9fe0ac9ae9f95ec"}]},"branch":"refs/heads/master"},"44af704836799e9b211e3eccfa5efa72f9713f32":{"kind":"REWORK","_number":3,"created":"2023-10-10 11:09:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/808111/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/808111/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/808111/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/808111/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/808111/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/808111/3"}}},"commit":{"parents":[{"commit":"990370a3673b8bdf4882816926868dd7b422db60","subject":"Revert \"CI: Disable bare metal testing on RL9/c9s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/990370a3673b8bdf4882816926868dd7b422db60"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-10 11:09:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-10-10 11:09:12.000000000","tz":120},"subject":"Add introspection rule to update deploy kernel location","message":"Add introspection rule to update deploy kernel location\n\nBack in the Ussuri release, we changed the Bifrost kernel location to\nuse ipa.kernel instead of ipa.vmlinuz. While this works fine for newly\ndiscovered nodes, any node added to Bifrost in Train or earlier will\nhave kept ipa.vmlinuz as its deploy kernel.\n\nThis can cause issues since upgrading Bifrost or building new deployment\nimages will update ipa.kernel, but not ipa.vmlinuz, resulting in nodes\nbooting with an old kernel and a new ramdisk.\n\nFix by adding a new rule updating the legacy kernel location and\ndocumenting how to update node information.\n\nChange-Id: I6ae44dcae78424b4638762d015cf24336444707f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44af704836799e9b211e3eccfa5efa72f9713f32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44af704836799e9b211e3eccfa5efa72f9713f32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
