)]}'
{"id":"openstack%2Fkayobe~808161","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I45e937ec7c102e89edee20679b8959bf7f6c7152","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"I45e937ec7c102e89edee20679b8959bf7f6c7152","subject":"CI: Fix get_logs.sh docker logs output","status":"MERGED","created":"2021-09-10 09:45:35.000000000","updated":"2021-09-13 19:55:16.000000000","submitted":"2021-09-13 19:50:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808161","cherry_pick_of_change":808029,"cherry_pick_of_patch_set":1,"meta_rev_id":"6c400e39f68c191d7f0e44b153dda6ebc5aa90dd","_number":808161,"virtual_id_number":808161,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-13 19:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-13 08:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-10 10:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-13 08:58:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 10:22:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-09-10 11:12:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 08:58:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6a7bdc9eed25eca5e0c0701075b7ef477028d255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-10 09:45:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d70aaa25947bf4878690ec75bfc3c247544e89f8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-10 10:22:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"71f1d2ec8de6432f96991d7243fc1ee9aec12bd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 11:12:06.000000000","message":"Patch Set 1: Verified-1\n\n(19 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4964ddc27d8e4e77a5c5bd34f069e861 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f659d828e6ec401d84ae655da8c651d8 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fb8d39d099e403f870a6f66a67b41f0 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b3dc9aa78414405b846cb900668da50 : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b69340894f44bd09d5b733ecd1d94be : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78ad139a7e9b4d0893cb22bc72cb69cd : SUCCESS in 10m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/62a303b67bee41f0bc7d4aba14c31375 : SUCCESS in 8m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72ee832d522548b5a1c906d128016638 : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/14c6abcb659f4586b7d1936de9ba4ead : SUCCESS in 22m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1652cadc43c34e55a13816f5abcb2758 : SUCCESS in 1h 09m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0deeda6b11f545a0bbcd98c2994d65ab : SUCCESS in 48m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9716b47cc03f41d3bf7d72d6e7354945 : SUCCESS in 1h 10m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/237a7b8ce3bb4f1c9599e43af4ca414e : SUCCESS in 21m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54ea9d05fc1e467d93bbc73e50a9b1af : SUCCESS in 15m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4e8d8780a76e4dd89b72219ab3aa889b : FAILURE in 1h 21m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1649a1045aef4db3815b3161550ac723 : SUCCESS in 35m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90f6975001634e4190dee07addde6e6b : SUCCESS in 31m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/65c041f44c884f71bd9a27edd39581be : SUCCESS in 27m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/edc0259b13974c59bf5f729bb4b96151 : SUCCESS in 14m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ef2c0ffc26a4e08a2b1e9f9284993a8 : SUCCESS in 12m 22s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"03cd4058d735c62bc66bd2e219223b639cc9ffe2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-10 13:44:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b3d8255e1745363647e0803f972d2049b6301af5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-10 13:45:21.000000000","message":"Patch Set 1:\n\nout of disk space","accounts_in_message":[],"_revision_number":1},{"id":"3bb405d58802335513651dc8f7bfebb0ec24e3a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:38:02.000000000","message":"Patch Set 1: Verified+1\n\n(19 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfc2fe6ae19a43c9a166b6e4be8ec37f : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43d36946eb394266aedf70cae69bb0aa : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18e16af3b2fb49bc8bcc2a45d344d921 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/100ae8cbde3f450ebbdd60d9b9b823f7 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/531507f6448d472c87785709d802ec10 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ccbb82d91f4132b6b20e96f163d83e : SUCCESS in 9m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6d0e7f8a565b47568eea68556d341f10 : SUCCESS in 7m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eb7e5afb112e4592a9ccafbc8b4c4723 : SUCCESS in 13m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa380453e8654db2990e675f151c1aff : SUCCESS in 21m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/68733aeaf2bd45feb533cd9b7b29fac8 : SUCCESS in 1h 11m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f316549080374d6d8a77c900d3f75bf3 : SUCCESS in 55m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/22a4ef6a3d5c46ac8cb5e171fec347a4 : SUCCESS in 1h 24m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1b2a417c376e408dbccded5708ccca14 : SUCCESS in 22m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7326e6b683744588d8bc2d169165483 : SUCCESS in 19m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6c985d8e34dd4e4c8002f312284c47ee : SUCCESS in 1h 52m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2c0f422247fe4b3ca08be4759962ee6f : SUCCESS in 34m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8e0ca68868441469ad969ee7b82c4ae : SUCCESS in 30m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e794e6de77854be8b29985cd10b54099 : SUCCESS in 26m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f759ea7dd3bd4a2fb7a8180fb80b18e3 : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc71fcdb1395461f9248c0ae05e9308e : SUCCESS in 11m 59s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"de61ebccb90d3cc3ff180b3d5b5fc9fc4c9addec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-13 08:58:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5f4cb541b11aab775f2c1b6a35805de279e75f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 08:59:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30bb5406e7ebfb0aad0820bcd1673633a21d380c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 12:38:13.000000000","message":"Patch Set 1: Verified-2\n\n(19 comments)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcad473a274a4471ad60051b35f6db82 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47036b5eb6884717b52f7f920baaee17 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c3d283597fb40618cfe46cf35cf8ff7 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba3298833c34f928ae482868a814ba8 : SUCCESS in 10m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/66229dffb10840559cf6996e00ac31b5 : SUCCESS in 7m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7aa6fb5c62a940efbe5a552fe986dd5a : SUCCESS in 11m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d646246d37574603852efd4797fdf666 : SUCCESS in 21m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4546b67ea41e40ba9c244ba643ad88e3 : SUCCESS in 1h 03m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbaf4c426b384852afc1aa9c4963c121 : SUCCESS in 44m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b696edd9bf2440a0b56c10c6984d48d7 : SUCCESS in 1h 05m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/23b635efd0704746b7d2608d9bcab4bc : SUCCESS in 18m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/712c80fb33f54daca4b838263d793b71 : SUCCESS in 13m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2f96978e80774c408f959ca46dc6d1d9 : FAILURE in 1h 26m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/92849c711c4b465383c81a51cc500f63 : SUCCESS in 38m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9ba4782af9743d6b55e67d0673bdf3c : SUCCESS in 27m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d523a73198834b159147adaff84d7e33 : SUCCESS in 27m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/52382e000e304c168582a8f910fadfc8 : SUCCESS in 13m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50a0c66c93bc4e08ad78629df70dbbb5 : SUCCESS in 12m 29s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"546a98f4477246a221ec4979d735b489e55219ed","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-13 15:37:20.000000000","message":"Patch Set 1:\n\nrecheck\n\nNo space left on device","accounts_in_message":[],"_revision_number":1},{"id":"0170f299ec9480c692298a34e9cb45e313c39e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 17:27:35.000000000","message":"Patch Set 1: Verified+1\n\n(19 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29ca60f6dabd4c2fbc4203966338ee18 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d529b146e3a42f693614a72462ab033 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed5413f7252a44f6ad4fba86fa5f0c4e : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba6ea4700e9a43588553fbf982814c32 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f1696616711488f953e1f8040679e6f : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49134ac749584ca9949ded3e514b77b1 : SUCCESS in 9m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5537f0de00094a4e96f0661e1e3feac0 : SUCCESS in 7m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28f833f206804ba6a25d54fe7720c5d2 : SUCCESS in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d217377914b941539619135cbb3b6beb : SUCCESS in 21m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b013bb35d021415b8452a80af8267b59 : SUCCESS in 58m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/920916ba7a5c46c8bea1a72872f3e362 : SUCCESS in 47m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8fc0ed0b2e4a4902adb068705a299bc7 : SUCCESS in 1h 00m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/53d9180ddf144be9a5c47e77a3ac94d3 : SUCCESS in 20m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4514e0052c8a43b3870a94b76fd9271f : SUCCESS in 12m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/679717ed57864a9b8ff2838356a156ca : SUCCESS in 1h 46m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d86cf9113afa458595efa79a5cbf472a : SUCCESS in 47m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/528b50eed49841bdbc72f7a0ada867de : SUCCESS in 29m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7efe4a8d22454de594c04941337eab73 : SUCCESS in 26m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/33f005d9db734f41b6137fe43401cbf9 : SUCCESS in 13m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efd0bd9c5136486eadd0cd53823da3c5 : SUCCESS in 11m 07s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"36d5f485a77ef8fed0f569da0afff857370ecf08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 17:28:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34ff09a18936e2d9c3ac66e700206461940f13a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 19:49:59.000000000","message":"Patch Set 1: Verified+2\n\n(19 comments)\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce81e7156e34b23b185333fe7f8b1f1 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6cf6cc8287e494ea212a0f2a4bfd6c8 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1c7d3b047304cf99a89f6ff25580af3 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578ce6448b744524b1b36f9669d5b9ff : SUCCESS in 8m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6e433a4e0d6940e4bd3950de3ac8c1fe : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8642481be16c42c28208562d2ea3b6b5 : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/768f967d9949431b9f06d5427011fcb3 : SUCCESS in 19m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f9a0b2a96d0549f0925ef492ad239cb2 : SUCCESS in 1h 05m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08758ff5641d4455a8565f3e63d5a1b7 : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/07410586f9e44c09af2eb405d3cb178d : SUCCESS in 1h 08m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1028877bd4fd4011b98d4b8f338dea79 : SUCCESS in 20m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f2cfc41a99545dfb97d7a2cd822566b : SUCCESS in 13m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/138dfe2149b44a59a62055f8f4325358 : SUCCESS in 2h 14m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b669f3dc4ce7489c9dd70deeef0faac0 : SUCCESS in 32m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df0a6af9ebb147e3a61c20bed0d5bb23 : SUCCESS in 31m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9d6081f0e42b4d75ace9c79befcf6111 : SUCCESS in 25m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/90262fb81d3b4f40aca250008f3f090f : SUCCESS in 17m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d786b13906054d729d4cc4d7db1b4015 : SUCCESS in 12m 10s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"0dcc4fc13fdb31c0895d0804f0c7e1a714c4bdb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 19:50:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6c400e39f68c191d7f0e44b153dda6ebc5aa90dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 19:55:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03760a1360464ea585f254f8edefc785 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2e980d18c9a5dbbc2d5ec86836fc93d1e3d1e20","revisions":{"a2e980d18c9a5dbbc2d5ec86836fc93d1e3d1e20":{"kind":"REWORK","_number":1,"created":"2021-09-10 09:45:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/61/808161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/808161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/808161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/808161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/808161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/808161/1"}}},"commit":{"parents":[{"commit":"a906d26f0d68f0644adf50420b42c8207421d7e5","subject":"Fix overcloud introspection data save","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a906d26f0d68f0644adf50420b42c8207421d7e5"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-09 09:30:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-10 09:45:35.000000000","tz":0},"subject":"CI: Fix get_logs.sh docker logs output","message":"CI: Fix get_logs.sh docker logs output\n\nCurrently the logs get printed in Zuul job stdout and only a small portion\nof them is redirected to a file - this patch changes that to send all output\nto desired files.\n\nChange-Id: I45e937ec7c102e89edee20679b8959bf7f6c7152\n(cherry picked from commit 923813ac4820044723814b777e4a39cb6eb273fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2e980d18c9a5dbbc2d5ec86836fc93d1e3d1e20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2e980d18c9a5dbbc2d5ec86836fc93d1e3d1e20"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
