)]}'
{"id":"openstack%2Fkayobe~808162","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I45e937ec7c102e89edee20679b8959bf7f6c7152","project":"openstack/kayobe","branch":"stable/victoria","hashtags":[],"change_id":"I45e937ec7c102e89edee20679b8959bf7f6c7152","subject":"CI: Fix get_logs.sh docker logs output","status":"MERGED","created":"2021-09-10 09:45:45.000000000","updated":"2021-09-13 12:39:58.000000000","submitted":"2021-09-13 12:38:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808162","cherry_pick_of_change":808029,"cherry_pick_of_patch_set":1,"meta_rev_id":"28b100e1fb3d8e4adace36351e0b49c78ebeb819","_number":808162,"virtual_id_number":808162,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-13 12:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-13 08:58:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-10 10:22:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-13 08:58:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 10:22:04.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-09-10 11:55:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 08:58:12.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"f5b4f9a5570b9705ccc2c1ce47bf004d3eccfc6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-10 09:45:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"035db3199dc60648c98248d78f979ca9c0c45b49","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-10 10:22:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf1a6bb7d7ae87e2fe70a24539ee237cf60baf7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 11:55:23.000000000","message":"Patch Set 1: Verified+1\n\n(18 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a6564deec0a41869d2d3bab3c9a425c : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c80be4d7c8d84200a0fb83f9b5875dd9 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fb74ea961b74ae3b588d568ca62c1ff : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8686aca304549daab464e492a4f28cb : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f40822d8636424ead4828eca60ea334 : SUCCESS in 8m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/028fa0a847ff4878a37781b124903e24 : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9dd44aa2d21e4186a0ae2d96a8ce8058 : SUCCESS in 9m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d0a7613c005b4743a91b6cb3058fda4e : SUCCESS in 17m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c7bbb8b826bf4103902bbf325db4853a : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c1fe8ae77ab74ebaa45718306af6e8b0 : SUCCESS in 1h 04m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e10174a9a50b411780e9112400b0288d : SUCCESS in 19m 02s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/efa64056376c4b9b9941af115e1b3a9f : SUCCESS in 2h 07m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/18af9cdc25d545d79975132574008c6f : SUCCESS in 28m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/aa07287f89c0430c8f9c7ccb1ca6e85b : SUCCESS in 25m 33s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1f73ffa67349421b9d37591c90cc40fd : SUCCESS in 11m 06s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"675abe960fdf2906db7fd364bf19cddbdd620396","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-13 08:58:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b08d375db6825c2d7faa94229d8ebb1f554fdb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 08:59:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb8233aa7cd525da503ac8aebb707a3b2452eae4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 12:38:11.000000000","message":"Patch Set 1: Verified+2\n\n(18 comments)\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca29cd9031fd4be4ab7e5eefee50cdf4 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcf0278d209a4f459a23a87cdc902969 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5630a61d9db54c3fb76128f7499a27c3 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea5ec2be8f444a58640dbfb7a206692 : SUCCESS in 9m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1f63a901e1ef4a05be77e75e89699566 : SUCCESS in 7m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/56697d35e9f14d8cbe88be44269b136d : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c10dac48b5554ed2bf8203fd82f4e5e1 : SUCCESS in 18m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b2e0a3d4fd224bd1884d63819c893436 : SUCCESS in 1h 03m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f5cf286a92bb4d3ab6a243d5d3e96b11 : SUCCESS in 1h 09m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/260ce99382784f21a7269131d3bc4ea0 : SUCCESS in 1h 01m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e5ee77e1c1ff4e85b909a2bf4980018f : SUCCESS in 1h 16m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f0f0c4b8a923421684ccc3b76c9a0219 : SUCCESS in 15m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0a15cd9cb557421cb0e57a40e287a079 : SUCCESS in 17m 52s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/24b74c2376c1475e8545000d6137c2f4 : SUCCESS in 1h 56m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c1bcaa9e7f8b4765a55c8814725e66ce : SUCCESS in 2h 11m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/85b9865c1a214684a9299018feaea8a3 : SUCCESS in 28m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bedffa0d958e4e6e80d6e6e89b63bc5c : SUCCESS in 27m 28s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3b81cb5377cb4a95b64ce7cc6e39c05c : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0600181aa6924cd0b84929191506bbe8 : SUCCESS in 25m 38s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/03ffe3851fcb479ea92e19b92e4379e0 : SUCCESS in 10m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/430d0548a2ba4958ae21e6441c175bf7 : SUCCESS in 12m 01s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"b948b3915d8ec4322e63d0db10d45718d6af7c06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 12:38:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"28b100e1fb3d8e4adace36351e0b49c78ebeb819","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 12:39:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aced912f9f34d84b822aec389430d93 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18460ecf3b3bed33f5fcb5b29163c6a188439dd5","revisions":{"18460ecf3b3bed33f5fcb5b29163c6a188439dd5":{"kind":"REWORK","_number":1,"created":"2021-09-10 09:45:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/62/808162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/808162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/808162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/808162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/808162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/808162/1"}}},"commit":{"parents":[{"commit":"750a17c48a4a48acda02aa664931b251a7fa2786","subject":"Fix overcloud introspection data save","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/750a17c48a4a48acda02aa664931b251a7fa2786"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-09 09:30:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-10 09:45:45.000000000","tz":0},"subject":"CI: Fix get_logs.sh docker logs output","message":"CI: Fix get_logs.sh docker logs output\n\nCurrently the logs get printed in Zuul job stdout and only a small portion\nof them is redirected to a file - this patch changes that to send all output\nto desired files.\n\nChange-Id: I45e937ec7c102e89edee20679b8959bf7f6c7152\n(cherry picked from commit 923813ac4820044723814b777e4a39cb6eb273fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18460ecf3b3bed33f5fcb5b29163c6a188439dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18460ecf3b3bed33f5fcb5b29163c6a188439dd5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
