)]}'
{"id":"openstack%2Fkayobe~808218","triplet_id":"openstack%2Fkayobe~master~Idaea4a984391a8cd05a5b0eee254ac6bad531a3e","project":"openstack/kayobe","branch":"master","topic":"ci-fact-caching","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-11 11:31:37.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-28 09:25:26.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Idaea4a984391a8cd05a5b0eee254ac6bad531a3e","subject":"CI: test fact caching","status":"MERGED","created":"2021-09-10 14:12:03.000000000","updated":"2022-02-28 13:11:31.000000000","submitted":"2022-02-28 13:09:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808218-ci-fact-caching","meta_rev_id":"019d43f8ca37505429f09a6d7fcef0e91a559566","_number":808218,"virtual_id_number":808218,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-28 13:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-25 13:40:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-25 13:40:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 15:35:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-14 07:52:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-21 11:56:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d0a9c43c5a278c70feb60015c8d68387d43cb536","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-10 14:12:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a92bb297ad5e7601aa0cc3277d7d7981125b387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:35:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c138f2f77c814e2fa776f857398c04ca : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2042665c485c41ff867a6d30c43d2361 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a06d3bf5b12548a0ad2ad1ff38e53843 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/920f123b0adc4b34bba77992ea60a443 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdda9147447a439e82d891f04607deb6 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9153496e79048b0bc4bd098e1c156a7 : SUCCESS in 9m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4bad71814b5e45e5868c75217146d332 : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/47216ce933aa47248e8ce57b61842ffe : SUCCESS in 13m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/670fa6471cac44059bd90875db03fcb1 : SUCCESS in 32m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/489e1b25b80b4e71941866b50d37b731 : FAILURE in 22m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8a7e498fd71481db6a4138725d01c5c : FAILURE in 17m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6ca4db9b595e443fbeefb0308c16a4cc : FAILURE in 22m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/760774f6b4d84855af76f1ecbf0fb582 : SUCCESS in 18m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/917e92d0266d4d73945210ca62543894 : SUCCESS in 14m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/124eed14a9304803bddcd13795d8e1b6 : FAILURE in 1h 19m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/944c77c1b2224e388b61cc202bb069f7 : SUCCESS in 35m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec8d8d54ad6440e68c19d9dddf75bb14 : SUCCESS in 26m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d63d0a7a781b4a53b67a9d8a872a9822 : SUCCESS in 26m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/082b992002584d80bdd1d186d97f237a : SUCCESS in 12m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1052fe485c674a7681b20763ef50034d : SUCCESS in 10m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"814dbb4da6a518d8e5777b4ace4d3f8a1f320a77","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-14 07:52:47.000000000","message":"Patch Set 1: Code-Review-1\n\nYou need to use a separate fact cache for kolla-ansible, as described here: https://docs.openstack.org/kayobe/latest/configuration/reference/ansible.html#fact-caching","accounts_in_message":[],"_revision_number":1},{"id":"51553e537b4e5fc3b9962871581f50a65489902d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-10 08:03:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"59773c3e3281c2703ae45799226f67e033bab6be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 09:52:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f148b60c504f48d0a3ea108d120b4b72 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c80b9d3203043ff90695aba1291e1a3 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa27fdd32c02472b862e80211ae07b7c : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70a924e3c34d468292ffde03d4b9920b : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e80f53dddc1146139a27be3b2241154a : SUCCESS in 7m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e373990cac694df9a8b4c1260fccc23b : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d2776e8034844b5a830e7dbf58db08e2 : SUCCESS in 12m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7987f1a84df74a6088b89682db3c9065 : SUCCESS in 32m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0edff735813744c6b26dc6d3db6a6a66 : SUCCESS in 1h 04m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0b9b4e886d549679637c627f2579975 : SUCCESS in 47m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a927d07543b34614945095143aaaffc0 : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6d007e42cede45ad9f54e1bdaa6d6136 : SUCCESS in 22m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a799e13b77444eb18aa48a4e7242b6c7 : SUCCESS in 17m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6014a99007f24a2c82d24a81220aed9d : SUCCESS in 1h 47m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ceb3bb656e1c47aca247361313db245f : SUCCESS in 1h 09m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7b3fa0e4af9e4b928ae180a7a9981025 : SUCCESS in 43m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/586313dff53e44b1b894882ad026a4c2 : SUCCESS in 28m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6a896873fb974ffabd89dbcfa2962947 : SUCCESS in 29m 39s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a3d3055f0de4b1fbda8dced2dc96e8c : SUCCESS in 18m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5c20af1ff8404615b0cbce3ea1d9593c : SUCCESS in 15m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e109352137ff4b5ea9785da6c61aab7b : SUCCESS in 11m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/26e4c33c3cd04a189901c0eca0449cd6 : SUCCESS in 17m 20s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1a96d35591e4aeaab136e0aa55abef9 : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":2},{"id":"34cb70d8faa135bd6f60c4f3922d3e6d71c75b58","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 09:20:31.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a003d10879810fb81e144dc5480570e578adb21d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-11 09:28:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d8b9a123bd36216250e3f9a04c0b048e98767813","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-11 09:31:42.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"aa7eaabc28039a588b2de3a7d2daf034611da1c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 11:20:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14012851e6e9439ba676b7410a795e7c : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aea21c783ee4faa9e46451c64d6443e : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eba02513d3ed4c849c960975fa9c8b35 : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9d940649f4b4fbb8b1b9f33ca427204 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f51a9ebbe44c6083f244eb302b2117 : SUCCESS in 9m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b34ca32597c54d54b78e049b4ad12e5d : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/30a6ea03a2ab491794a8b41a230f9b52 : SUCCESS in 12m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2990574f92774624b04a076568e0b49f : SUCCESS in 30m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/40a09d61d27840c4b8a67f786001fe19 : SUCCESS in 1h 17m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a3785f08ffc48d69e4b0e02115eecf3 : SUCCESS in 47m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e2e8de1945d744d4a727a053ed2cbb44 : SUCCESS in 1h 07m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7faa8081f2594ccf93b5591877159110 : SUCCESS in 24m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3592a52e7c434a849224c32b16010f72 : SUCCESS in 13m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8f0bbade59164e97a2c7cd62cd730bc2 : SUCCESS in 1h 47m 01s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d0312588ef04460bea15e28de463000 : SUCCESS in 1h 11m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7e2746df116b4714a0bbaf77108e430e : SUCCESS in 45m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17e41c819cfd4ac9bebe4163b2e1c1cd : SUCCESS in 36m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5d3296b83d9447b78181060c932ec5b4 : SUCCESS in 28m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c9fcfc76d674e97bc2e08ef42606754 : SUCCESS in 22m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fc31eb9fc0ad4fa7b8ed2de2130e26ca : SUCCESS in 15m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b5b2a964fa64c44a17d3cc1fa9b81ca : SUCCESS in 12m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b3de31192ad64bf8a57c0391d79a4259 : SUCCESS in 16m 49s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/994ed5bd9a73413b8e1d0202e65f2064 : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":3},{"id":"8fcd0e5868d632d10696f8164182f92655b261c8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 11:31:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"990389ed74ceb3753a8fb344257e0e51b92e8545","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-21 11:56:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4be580f56c7d2303758a4db3a933a4611e49d3df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 11:57:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"17978613ff37b72896ada52af5c368b0db54a9b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 15:32:57.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cafe0daf6c6455389e93035d15bb403 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffd88dfd65f94f3eab53bdb133049b5d : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9733e89626e4a48bd13885507de445d : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58bfc20496c643af9da11c4dbb520393 : SUCCESS in 8m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/95f62481b42342cdba99115be6c3a8d0 : SUCCESS in 8m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23e4a90ddff74bcaa2f5d2e9eab24ef0 : SUCCESS in 12m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c0d1c54e640e4e2b98ada601edde1b8c : SUCCESS in 35m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0f05454f56fe458993b3daa34e8034de : SUCCESS in 56m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eac315143dd342bdad0e96fbf40ad490 : SUCCESS in 52m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ca2b2a09b3f34c97b7fd756b95f195a8 : TIMED_OUT in 2h 04m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/54e66927c222426e94e1b982d9a5224e : SUCCESS in 20m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6241d7d503f14628a8ec378b76bfca02 : SUCCESS in 16m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fb57927b84624ebf96dfa8a1a90ba0b8 : SUCCESS in 1h 31m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfe55cbca8de48c18243e38e79f3923f : SUCCESS in 1h 16m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/27be33118c9f49599e6dba9d95c361e2 : SUCCESS in 52m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f49ab3ab7ca4973a2311413cb556fc2 : SUCCESS in 33m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6367529dc57b439f98c8e9d8737f55b6 : SUCCESS in 21m 00s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08bcc00c61fe40e5a48ae25a023f02b0 : SUCCESS in 24m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/14620899f2ae4de19aaeca1347474890 : SUCCESS in 16m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/300f3ab6ba724b388e9b0f5f7ef8ebb5 : SUCCESS in 14m 35s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e603007eaafc4adb9b8119d7702e07f6 : SUCCESS in 19m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/887734aa0c0149758865e8523bb4d708 : SUCCESS in 16m 20s","accounts_in_message":[],"_revision_number":3},{"id":"cc794234dcf96ed806673c404378399d97a92b86","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 15:40:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f969066071523603e4e133b21bf864388d42716f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 17:45:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d59b8f51e71044e28528903c02a982fd : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8bb185a8d0d47a9aa07c6dfa9aad625 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4120713d17924343bdf7f58885fc93fd : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f476d12114244eb85924c234d779b27 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e284ad734e4714940fc252a4013187 : SUCCESS in 10m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/386f39183286494b8ff79e2872febd70 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fef16ac8721c439e82052f85e135ecbb : SUCCESS in 15m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa5587efb14b4f57b6ea2195090e6706 : SUCCESS in 32m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/35af03b93269447a88e6b2e4a22648e9 : SUCCESS in 1h 10m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06ab4a68e3bd48c3a03ac702cc0be33d : SUCCESS in 52m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d0a992208be040e6b4b9813316de9e27 : FAILURE in 1h 06m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6e6f77a5cc224e1ea7b33df3f402fa9b : SUCCESS in 19m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c00252e0ea642f8ad723cd210c7cea0 : SUCCESS in 14m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2d9bd46b2a745f59d67c370f34e2eea : SUCCESS in 1h 57m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39bed8d056c74dfca964a32b2c62212b : SUCCESS in 1h 01m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d0291b2515c9443997b195135786ce5a : SUCCESS in 51m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2a192213a444c4f8ae28689f6e42899 : SUCCESS in 40m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7da8430e2c9243be8e43bbb86f629bf5 : SUCCESS in 27m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15c5416d66e0447fadec0d087b13dcd8 : SUCCESS in 19m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/816d625641514780ae5a05dcb0276198 : SUCCESS in 17m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c422ab1f8484c4196b0718b6a0edd96 : SUCCESS in 12m 20s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/15cd7b5a12654c25b2cfd10a3b4357a5 : SUCCESS in 15m 36s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c28190c336bd489b88e012a860702d81 : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":3},{"id":"21a49763ce8cc0a1bfd9e15594667477471b9780","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 22:02:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ffb8ba64d516fdebb054459fefd91aa9282bf95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 23:23:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17bb39acf6c242ad85c0919178993ed5 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd57e3523ab643478c0b4df89953332e : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2009ab5447554ad7a5fdccd266f86365 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3045de79db42460c92ed7d543b882e40 : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c761c46924554b028338bce26d4b7ed8 : SUCCESS in 9m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ad45cc4c21f54e40bbd3365b7570e5a1 : SUCCESS in 7m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7e0c10290134477baa2fab147d244a49 : SUCCESS in 14m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/56bf29b1159747c8a311435b225d2901 : SUCCESS in 34m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d5a5437b04e34f48becc98c09df38d20 : FAILURE in 17m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11dfbe5476184d15a1565951d5cf339e : FAILURE in 12m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5163777b2ab946e7bb936ba22029b3f8 : FAILURE in 14m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cb988e4336e7404a9a089e2a3e45c8bf : FAILURE in 18m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8328cb88d274408937133550886f622 : FAILURE in 14m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/53fc3805cf5d4086b1be2fdd7a33123d : FAILURE in 1h 19m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d872d46d16444738784bce8122de2ad : SUCCESS in 1h 12m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ff4b7b20b5474d68832c74538516d67d : FAILURE in 14m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02449c53377845daaf6e1797bd0fc986 : FAILURE in 11m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/56c5284049084ca794ba93e424277e6c : SUCCESS in 33m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14dadc426567486a9e135d278c042b76 : SUCCESS in 33m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/90cb204779fb40f0a299acc67a8d8ed9 : SUCCESS in 20m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da758f16d7f947c3a6bd69279b6f85de : SUCCESS in 18m 12s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3ad8a27d621046a589c4601a5e5e29f3 : SUCCESS in 22m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04fb840c4ccd43a4a3421ba819e91e8f : SUCCESS in 15m 49s","accounts_in_message":[],"_revision_number":3},{"id":"963ac77fdf46b6891a5569c12883bc2a2e93b016","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-23 08:30:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"02acbd9e7a566f9a94c1bf6b9a2892079235655b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-23 08:30:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a95b940ff1be6d1fadd2baeb03ee12a21e461a08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 10:17:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af50560658d0452995cd830d0125e0dc : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d56bb515d1ab482ebcfc6139cc921e87 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45ae0c29d15b4240a38d008f90990511 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbb5627ee0e4424b8ad91515e0022dd6 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b2cf906f2444f099e4bf5c47c8c14bc : SUCCESS in 9m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8adaf52c2b4444ddb51d586b46f07394 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/517b28690f04478d9644cf3980dd4969 : SUCCESS in 12m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/979bde939ca0445a883a6a1ed28bd62a : SUCCESS in 34m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7c4d99be125b46b0b4fc611617be6c13 : SUCCESS in 1h 07m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cf4a7dd775544ee9e0e7676c9905434 : SUCCESS in 53m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/66dee8059a6f42f0baf807cb68dcb9a2 : SUCCESS in 1h 10m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e7ca7627433f475884003c73883cc087 : SUCCESS in 26m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e52feacff7444a4b86d8bb74978ded0f : SUCCESS in 15m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2b505fb8d18540e49e58fdc6742a947a : SUCCESS in 1h 45m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/252b5ff31d904809bce91c8da8ed2575 : SUCCESS in 1h 08m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/20b1418bdf514f73b450d536be1c0f26 : SUCCESS in 50m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb7e57949de74868b075c24e325e6cfc : SUCCESS in 35m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/939dacc3c35d40e7920a6d00c439488c : SUCCESS in 27m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68cf058d6eca4f14bc507599eb0c253c : SUCCESS in 25m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/db19e4859f9f4034a95287449d09e8a7 : SUCCESS in 13m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eafb21cdaf945c493590e228a0b63a9 : SUCCESS in 12m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/bd136d4699dc4e24b785ab15bac29a3b : SUCCESS in 17m 09s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91c267f9239a4cebabf9b3081ee1e056 : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":4},{"id":"8a008e8d74c08efb76c9d1be8ea4bf000c24e2f0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-25 13:40:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f93e689d92ae0a073cc7062018c0f5d712e9c94c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 13:41:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7bb8a93157f566f08362f4e142f94410135bad67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 15:27:27.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17afac5b4dc2429eb96e9dd9328c5efc : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6285ef57da143989fbff2f344eff2e7 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ae4a62a57264d3bb7d08c42949811ea : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b648bdd009a3449daa1e5201a177a066 : SUCCESS in 9m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/120fe858ca9840eb8a4b941e3eb9f716 : SUCCESS in 7m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dbf55910bb2b474690abb82df19dee21 : SUCCESS in 12m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d3d2f416ca30425599e5e51e4f25b22b : SUCCESS in 34m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2de918bd127347a2be216e705eef78cd : SUCCESS in 1h 01m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7150205b0e8c45b2b77a08439c181ac4 : SUCCESS in 46m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8e42d02b5e55455cb7af4f0daf535aa9 : FAILURE in 1h 07m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/73d94ad6f5dc43bcb2e1eb0fa9167f34 : SUCCESS in 21m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8dbba0f8c9748419613c71d78d3b219 : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3463fb6ed32d41549153b624ee75916f : SUCCESS in 1h 43m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eef6d10007144cfa7f9d1aeb28eb763 : SUCCESS in 56m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fe193734a2b148a3a22a0369305d7cdb : SUCCESS in 47m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/070d2e688fd74fc7876e724452d8be48 : SUCCESS in 37m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6c21e3a3d66c4c5bac043eeae4d46d2b : SUCCESS in 28m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb47cee19d9a4a949ad9851fdfb909de : SUCCESS in 26m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0f5900248a294958adc69ba926b88c52 : SUCCESS in 13m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72ebce8b6dfd450e894eeaaffc6a4887 : SUCCESS in 12m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8d33cecbe3fd47b9aa4968549973df2e : SUCCESS in 13m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07a0fb0d64ed41a1b7a3826d79c7624b : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":4},{"id":"8cb6b3a053fb9836fc2d4939be732f7549be5ff5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-28 09:25:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e0ec37c981c804e237e08b4a0d04ad762d8cb0e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 11:18:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea585e86146247a085f80d7978a504a6 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a52f607f3bd4e589ed39f59616dacd0 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85139d2564d34280b53b21a772dae1f8 : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f92e09360dec486fa927c6e16ca757eb : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67124a2c2c354594a0b3a7cecd4ac40c : SUCCESS in 12m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f70bde4a22d44039580fb5100ddbd0c : SUCCESS in 7m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b92a202ae344dad9658a8f309fcde9a : SUCCESS in 14m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/56987419eb5b41469bc265d69a222abd : SUCCESS in 33m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7377eb4d810f41128960285907acf93e : SUCCESS in 1h 08m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4387f87b92648fea1f8f592943b8788 : SUCCESS in 55m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5204499735c64aa3b2346b30355bacc3 : SUCCESS in 1h 12m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/dce33dce60d54d30944880dddf34ef89 : SUCCESS in 22m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ced19826fe4428388449766f4af736c : SUCCESS in 16m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c1cdf2fa0e9841918a4d1ab21838e29b : SUCCESS in 1h 48m 56s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfc4dc3e5aad4f62a312f5e25f04915a : SUCCESS in 1h 16m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/415a5fba9390480f827fdd90d4603d0e : SUCCESS in 53m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f45d268f67af400f801e2e28dbfc5392 : SUCCESS in 38m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f6e17e172af04ff8af6e9ec164ee3f3c : SUCCESS in 29m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee05a0867a6f453b8e869bca4190e528 : SUCCESS in 39m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/888781e13daa4f79817896482227b0f0 : SUCCESS in 16m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c07a71b10fc74b1da06e568b9a804d2f : SUCCESS in 13m 21s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d4ea946696ad418b84cfc83026f4bca1 : SUCCESS in 19m 19s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e41dd8430d204f08a14101df72b9465b : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":4},{"id":"a05fc0b431108fced2d6dbbb376ea281a20f216d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 11:19:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7d80b2bb72ed570430cdb5bc89e2716bf68abe78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 13:09:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01bfec5a8c7241eebeb27ffd8ce81837 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/464f970d85114035b5be0d758a1ae55a : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee1a2e559b574167910889519e6ac50b : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed88abfa4c34dc6a64a886ffcfc5f01 : SUCCESS in 10m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d1d787c93c57429c92d310d781fe0a4a : SUCCESS in 7m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cfcf2c1e1d92408dab00bd2350dca2d1 : SUCCESS in 14m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/48ff4cd7c98845c5b8b2151b53e7a27f : SUCCESS in 33m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/052a9bffabd64881a61b372f018b41ca : SUCCESS in 1h 06m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b94ec1a647d471e84a4ca35771c741d : SUCCESS in 53m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0116fe7e0dd94c8db402b922d405557a : SUCCESS in 1h 09m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/48c2ed9072884096819c9e7576cf5442 : SUCCESS in 23m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3a5119b93af47f1841cd8b9ec15594f : SUCCESS in 15m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bffea214f9d5446b8e9bf97e32d6e481 : SUCCESS in 1h 47m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10d4c31ce58a40498c5e5b8ddac2fbd3 : SUCCESS in 1h 10m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c89d12a50eb14f7f82120dec98ad6509 : SUCCESS in 47m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ebf077a800e4ab4b800f2971b1d950a : SUCCESS in 36m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5de1fbfb2b714b97a7c4544fc3b1e776 : SUCCESS in 30m 22s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/768616d3a7884923a649b864c5b2a130 : SUCCESS in 23m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ad0aa407abd44a27a97b13a412b062b3 : SUCCESS in 16m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/000e7e0b89c742f996a57617127cf94b : SUCCESS in 13m 13s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/bf4b359a763042f0974fa7b0ee16636f : SUCCESS in 19m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c35c83923d9e43289d6a6d37fbb0e53b : SUCCESS in 15m 01s","accounts_in_message":[],"_revision_number":4},{"id":"05a710841730b68a25e7726a6d77a0fb1f7ac2d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 13:09:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"019d43f8ca37505429f09a6d7fcef0e91a559566","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 13:11:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef544d0e0c3540a6b8972a925827ff42 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"959bef67452093ce7af298a971dcca4dbc0507e0","revisions":{"24b74cec6f4648206e16deb780a9cbe2554fb972":{"kind":"REWORK","_number":1,"created":"2021-09-10 14:12:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/18/808218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/808218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/808218/1"}}},"commit":{"parents":[{"commit":"65546beb0fc85c9c74ad0e877321230689b364a0","subject":"Merge \"CI: Fix get_logs.sh docker logs output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65546beb0fc85c9c74ad0e877321230689b364a0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-10 14:11:43.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-10 14:11:43.000000000","tz":120},"subject":"CI: test fact caching","message":"CI: test fact caching\n\nChange-Id: Idaea4a984391a8cd05a5b0eee254ac6bad531a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b74cec6f4648206e16deb780a9cbe2554fb972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b74cec6f4648206e16deb780a9cbe2554fb972"}]},"branch":"refs/heads/master"},"2a7f9a3472caf025603be88c081b3a755ceb2289":{"kind":"REWORK","_number":2,"created":"2022-02-10 08:03:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/18/808218/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/808218/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/808218/2"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-10 08:03:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-10 08:03:21.000000000","tz":60},"subject":"CI: test fact caching","message":"CI: test fact caching\n\nChange-Id: Idaea4a984391a8cd05a5b0eee254ac6bad531a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a7f9a3472caf025603be88c081b3a755ceb2289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a7f9a3472caf025603be88c081b3a755ceb2289"}]},"branch":"refs/heads/master"},"151160318895b2abe1ef5abe2f90bea15a378334":{"kind":"REWORK","_number":3,"created":"2022-02-11 09:28:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/18/808218/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/808218/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/808218/3"}}},"commit":{"parents":[{"commit":"37a9e87041d0bfb2ae43a0e4ae597b3047454e96","subject":"Merge \"Support elements repositories for overcloud host images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37a9e87041d0bfb2ae43a0e4ae597b3047454e96"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-11 09:28:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-11 09:28:27.000000000","tz":60},"subject":"CI: test fact caching","message":"CI: test fact caching\n\nAlso synchronise Ansible settings between Kayobe and Kolla Ansible.\n\nChange-Id: Idaea4a984391a8cd05a5b0eee254ac6bad531a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/151160318895b2abe1ef5abe2f90bea15a378334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/151160318895b2abe1ef5abe2f90bea15a378334"}]},"branch":"refs/heads/master"},"959bef67452093ce7af298a971dcca4dbc0507e0":{"kind":"REWORK","_number":4,"created":"2022-02-23 08:30:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/18/808218/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/808218/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/808218/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/808218/4"}}},"commit":{"parents":[{"commit":"f477473bd700cc4453ba17d487f17dfc2ad6ba48","subject":"Revert \"CI: drop jobs requiring kolla-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f477473bd700cc4453ba17d487f17dfc2ad6ba48"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-11 09:28:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-23 08:29:43.000000000","tz":60},"subject":"CI: test fact caching","message":"CI: test fact caching\n\nAlso synchronise Ansible settings between Kayobe and Kolla Ansible.\n\nChange-Id: Idaea4a984391a8cd05a5b0eee254ac6bad531a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/959bef67452093ce7af298a971dcca4dbc0507e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/959bef67452093ce7af298a971dcca4dbc0507e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
