)]}'
{"id":"openstack%2Fkayobe~808931","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I14db53ed6e57d37bbd28dd5819e432e3fe6628b2","project":"openstack/kayobe","branch":"stable/victoria","hashtags":[],"change_id":"I14db53ed6e57d37bbd28dd5819e432e3fe6628b2","subject":"Use ansible_facts to reference facts","status":"MERGED","created":"2021-09-14 11:18:20.000000000","updated":"2021-10-04 17:10:17.000000000","submitted":"2021-10-04 17:02:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808931","meta_rev_id":"ab855a243e82da77d8ca19f35ba171864a781f32","_number":808931,"virtual_id_number":808931,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-04 17:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-01 08:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-17 13:30:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-10-01 08:16:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-14 12:23:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-17 13:30:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"ab9917333e84a8f361ad348f79486b621234d6df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-14 11:18:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed1109d08957cf1647ccfec7107ecf5670f7a16c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 12:23:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2f88e13124b4b83aa293dfb09c7f23e : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f267a7b318384eada32dae8a751b29cb : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9873a95198334affaa5e8b83af0c7d4c : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/831bd40965b54f16bc64ccba50667cfa : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e3f55ba794e4bb7988a113b38953a0c : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc5cd60f68bd49f587c2ea453dcea0a4 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/64d4c584d31d44afa6364957f27d0c60 : FAILURE in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f400b9f01ad54eb58558b30cc323360d : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1bc1d5a5ddf14a7f9549df6d1e0f2b20 : SUCCESS in 17m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/07989a7db4c04af583d3043d28046ab5 : FAILURE in 7m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/682a015a848e458bb38d1f909bd15894 : FAILURE in 7m 44s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/5622d96250ef472ea64802404404f0e7 : FAILURE in 8m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bcf74197c7ca4d1882400c004013da3c : FAILURE in 7m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c3f1c3d25cb347249be1e5b3a605770c : FAILURE in 9m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1970a5b5ed5a4f4c80313ea4ed75a2ac : FAILURE in 9m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7581212d0d4142b4b180226939292cfb : FAILURE in 1h 02m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/91b0a338db3d4685b86387cd26db89b5 : FAILURE in 1h 02m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2ef85e3b213147d78a6be1519ea82518 : FAILURE in 6m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dc4277d747d443029eb6b25d3725580a : FAILURE in 6m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ffe99223b2604472bfa738ac5e831a1d : SUCCESS in 21m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a6284d3e28e6426dac541fe412e1c62c : SUCCESS in 24m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0b4ae11fa29d4651a20b644b46132d27 : FAILURE in 7m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/735fa09f60e14ff386e4b403cfccc5ee : FAILURE in 6m 39s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"f419eb69728ce14971d7bbd0541ab14169511fea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-14 13:25:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"389a69a46c23d93b982529e2962d85ad091aa899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 14:40:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0c3e155f1ff493ead10e6bfb5e9f040 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091c76d65dc74402bd9264f601ca8d19 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc0c4fa984c54fc9abb1e76728065fe8 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1270b2f41b494fad878b923d3487ebe3 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f6791c4d9341f6980d7860f1b61a5e : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce5c9e6ff0ae473484bee89957c459b5 : SUCCESS in 4m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/713c4c56c86340a3acfc159f4f98ed77 : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a12c66b9da1a4ee78eadaf740a05bbc6 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/deb0f216897546b8a80e2c10d7abe13a : SUCCESS in 18m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6f03907c5cef4f65a4281a1fc609010c : FAILURE in 41m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/dde8e215321d41009f902e0caa0e6493 : FAILURE in 42m 03s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b6e849a664c04fbf92ecff58842c43ea : FAILURE in 44m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/691fc836f7484081878078fcb1ed972d : FAILURE in 45m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1b1340bd973147f8bca59f97dd84fc70 : SUCCESS in 14m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/212aa049230b42f69ba7f44da45a04cf : SUCCESS in 14m 58s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3aea9465a5f24357af7eeca6ec767d9b : FAILURE in 1h 07m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5dd9538e5fac490db561c7d23b822502 : FAILURE in 1h 09m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/34375239c1cc4ef999811e472a2589f8 : SUCCESS in 25m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/608cee338015482081d57c4371edeb74 : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a048e46bec1c4c58b79a7af38eb4ac2b : SUCCESS in 25m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2023b722324440aeab2c295feffaa8c5 : SUCCESS in 24m 28s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e1a9defa339b41f79ba493bb9d81fbfa : SUCCESS in 9m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/72380e403092492d9dcc2b96c69093e8 : SUCCESS in 10m 51s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"09c2004793e20431e156c2ea67c53e7a0a6d2c86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-14 16:14:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bc664784e16069dff6e7100d15d217260fd4540c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 20:28:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7bd363f4ecd4175af0f0b86a15bf72d : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f22df13374340f882a8dfea8fcf25dd : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f55c4939724485bb14d13c7947f21d7 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0da65292d4474f39af2b6fbcc8de2f4e : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c9ad8efe3464f699f580944e9fa7407 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6de342fa1aae4effbc8a827f575e5bd0 : SUCCESS in 4m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f171c05c9f6849d3b09a6b422e69de70 : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/91d98d83d32244c8965510de1a52a5ae : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7635c5ea3bda4c86ab42a611a1523a21 : SUCCESS in 17m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7feda1cfb33e43e689ed932464d0a2e6 : SUCCESS in 57m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/295e86e22f2f4cad9ff497bb27c7d079 : SUCCESS in 1h 01m 43s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d64509e72af2405fbf2898f2a98da4ab : SUCCESS in 57m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/35de1de0957445b9883ef3e8ed274175 : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3bc8cc0510604c768595b8cddd4c7c58 : SUCCESS in 14m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/231eeca02e104c9e85ab475be409d60b : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/61c42430196f4b0baf6e51d2abc49d9a : RETRY_LIMIT in 1h 24m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3d71a851ad4d407dbcb207e256e24416 : SUCCESS in 2h 04m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8664cf99c9914c619a40081528820171 : SUCCESS in 25m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4036d13233924d93a4542781ecf73600 : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/46ac34562e104f898879b4d926b36e45 : SUCCESS in 23m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2b89ddced04541fbaac4a435209f9c58 : SUCCESS in 23m 47s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/cf2f4c06f9e147f6b29951d372309fa0 : SUCCESS in 9m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d94a09ed0cf44d6a88ae7bd7afc774c3 : SUCCESS in 10m 32s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"062138b7d7eb130cd88adf06a2d7d37079e1c7af","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-15 08:06:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"aa30046cd24d70983b6d68716c634c7007af5bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 10:16:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18fecc1f2a43489ab4d4db7f31de5fcc : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9777395778c4429bb0f098dee9f3f09e : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e4fc205f18a48f1bbba9686a6067d60 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/000293aa15a74edab75762dc8432c3e5 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caab1caa6660442398b3fc50c281dbdc : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6f2a62eb22646b593f27c405d91d5a7 : SUCCESS in 4m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/837a1c791c76439287fa90bfeb391445 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d0cc10b7ce924e6290636974e8cc8964 : SUCCESS in 10m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2bb14346c4374987ad4beb04b99857d3 : SUCCESS in 19m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e4c6a591ea314f1fa9ffd255062c24a5 : SUCCESS in 50m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a6312b47e7bf46549c0974d7cf4273ec : SUCCESS in 1h 00m 09s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/587ae6f3495f4899a419ed5d02980c88 : SUCCESS in 59m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9638ab78d2bf4b22b60b608c3d179a31 : SUCCESS in 1h 00m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/72deb98067b94d41ad2b1dfd292e273e : SUCCESS in 15m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2f3eb668c2eb4541a3ece31c6eb9281c : SUCCESS in 16m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/585f43e1ae0a4b64965335da57ac6bb5 : SUCCESS in 1h 55m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6f3359095c844597a928f0bed61ef724 : FAILURE in 2h 01m 49s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f9c5b23ae6a34d73acd05b96f45ff4bb : SUCCESS in 29m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5017ce5586f74b9b9f3f22f754fcf584 : SUCCESS in 29m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1df7fd96b9704c7da4029da7f4347eb4 : SUCCESS in 26m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b74215c7ebf048b6b5169878b4bf349b : SUCCESS in 27m 28s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/08bff7bf62c447fbb27e5f06ca5835c7 : SUCCESS in 11m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ba9042c902cc4e259c8995c1da2d1d95 : SUCCESS in 12m 20s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"3a833c6fdf97fed3211748a67a8e98ab269a8f3f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-15 11:07:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a8fac327691cbc5d3520a830aacdcd924ea0993c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 13:19:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d65227f268c4e0ebbea534bdd541cad : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2b90cb1347940b0bf6b35d59a2f9165 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/519adf5fecf743c1817956bdfe6fea75 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86c192cbe04e4cefbea08f001499fd4b : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cad659f66bc14cacbdbf66a3600a7207 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e731e1538884e9f99f1c49e639bf2e4 : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/13c7d1c525434441b8d149513a2b9153 : SUCCESS in 6m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3da0bd74576e403eb75cf0bbb62be974 : SUCCESS in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3547723240964caead26dcb8cd6b7511 : SUCCESS in 17m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c17b4345c4cb4b1293c5e4db5443af77 : SUCCESS in 58m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/157a3f49e06e4e49b551edbcadf29dd7 : SUCCESS in 58m 23s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/51ceb9b79f7f4bf9bcdba7c51e52463b : SUCCESS in 1h 01m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f25656706b0443fc89c604c9f82027f2 : SUCCESS in 1h 14m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8ff9a8e7c5764bf0acf259c7e919688d : SUCCESS in 14m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2027a4c6847f4b53a7f1c809aa6d9f73 : SUCCESS in 14m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e188697e4be04716a569581102c2fd2f : SUCCESS in 2h 07m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0b6fbbfbc33c409f873bd91c1b13da28 : SUCCESS in 2h 10m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e2dffbea054045bca4eae2cca0b9bcbf : SUCCESS in 25m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/036b7406bfd040eba05e6a222f7d452c : SUCCESS in 27m 13s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/27164ea4f8c340328b83161800c18a6c : SUCCESS in 25m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/51b07edbc3a74f14b36a1d1d1a48748e : SUCCESS in 37m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e4feac127c7c45828a7aaea38c22ab7c : SUCCESS in 12m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c3c8a39f88a64379bcb92ad96571e29b : SUCCESS in 10m 46s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"b40d6844899e0a1698314b094517102df8fb491f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 13:30:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f7c3870fd4af0ae24432fe7ef246942068f6d73b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-01 08:16:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1c23fa99f27ca8e05c1731c9193d6c12ffe8b75e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 08:17:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4a319c674db9cdcae086e24103d32b5a30121768","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 13:07:40.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb0e3e482dab4c5ea5bc5ea326299572 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/926f1a15e45241c1b23d35cb0fda1a68 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a8bbc9328bf43d5ba1bc4b9f2d399ad : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/765c78b973ae40c594c6e50635b301aa : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c91709aeb74348048295e08823d7bdf6 : SUCCESS in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d79a4f09a6024797a3e0baab119f8660 : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8aadb46c273549da8c6606b8e63f97c5 : SUCCESS in 9m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d39907d9980f4a7dbf48b316fea2bdb0 : SUCCESS in 19m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/97c830486d7b48d1a235ba657b699880 : SUCCESS in 59m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d0721550b17849c7a8e1b21337c56376 : SUCCESS in 1h 02m 28s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/38b73b44d5e74d1abec018a0b8b8f6e7 : SUCCESS in 57m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8ef4f862f8ac4f86b14acd0deb65a5a8 : SUCCESS in 1h 02m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/11b9fc1c4c75438d975aa348b4dead1d : SUCCESS in 15m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/30350b631a49496294b427dd350f8dd3 : SUCCESS in 14m 44s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/18692f26a77a48c1bd9c276b0bce7bcf : TIMED_OUT in 3h 01m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/33623df4297d4274ab41d7274fce36ff : SUCCESS in 2h 17m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/07fb227704e14cc69f1ae4f8dde985d3 : SUCCESS in 25m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4af2a876c5584c2fbfec8626f1196092 : SUCCESS in 29m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1aff272a9d30484a844b08ba3c4edaa0 : SUCCESS in 24m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2685685a21a34756824720e52a137cf7 : SUCCESS in 28m 55s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c9be8658559343ff8f5468f0976d1c83 : SUCCESS in 10m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/547a54110a4745b7b5a3d98c597db329 : SUCCESS in 12m 39s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":3},{"id":"5cc610c671353b3bd3eca3f1bbba5f342f01454d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 13:17:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e28cb5062d3b891300f471b92c4b664455e741f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 16:00:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c29cf8b71b584eb09c2cabbdaad0b753 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9624e5544fb042bda4a62b1cb939ec24 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1de22c64c2c947e2ba31c3cf0a1a1508 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ccb62267a0248249d587fc80efdeeed : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51bc897e8bd44f9192e14c6f19ce56c7 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a8eaf4cef0e41f5b9b7f672eaf21398 : SUCCESS in 5m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61066ee49e884b6ea02a8d6acbdc10ac : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6de25745afdf42eda7358757e53989c3 : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75eaae78580442fb8099d7fc63df5936 : SUCCESS in 20m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/56714a19aedf4d08b31abfa92a6a99af : SUCCESS in 1h 01m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b896d89ad94d4e319246a8fe1dc46798 : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/54f9f18bbc7d4926bc46ffc1ae12d41e : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/100fd0fcefa549178321b69249efc04a : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bce8f94f611c4123b1e1fc2bbd2bb17e : SUCCESS in 17m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b010780937de4de79d119b64255f95ba : SUCCESS in 17m 25s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/afdcec195f43437d83dda438c931e3b5 : TIMED_OUT in 2h 39m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ecaad2cac5d841068dc822c03008dac9 : SUCCESS in 2h 12m 38s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/940b5e704156442ba851e6a5e516930d : SUCCESS in 28m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/218faa11db8d4600ac1a169a16bcf0a9 : SUCCESS in 31m 11s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/84c31d3301c24c7aaa962b0d0c4c09ac : SUCCESS in 28m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/71407d697f3e418097469b1404d1750c : SUCCESS in 27m 45s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1664985a474f4228a98c8c83d40fc946 : SUCCESS in 13m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0598214ab865466aa8bba5e107ba4fcf : SUCCESS in 13m 43s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":3},{"id":"e067e29b0482ebcaaa22beb3901194008fc15a58","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 21:08:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7b5d6a3d85ff848d8a92bd586edf165c17c85d64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 23:00:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92d6c418a58243c1bf6bbb0ff28fd187 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d972d29226114960a5c985745367ef06 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a38f6ef4a0dd42fabac265c590446944 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bc11a5b7fcb46d09280f846cebfd764 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a68f0d2be2439b8c6431cf8242e74c : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a054b7014bc5411e878093fde08a7d0e : SUCCESS in 4m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/baa4165ffc3d4e8d98e25f0a99ae8a23 : SUCCESS in 7m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/05cbc1a91ff04f2c961b3ee8f4eeb5d2 : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/35e2318d91a9410fac66d0b56f4e74f0 : SUCCESS in 17m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/dcbfc0aec4e64389894b2704849bc9ed : SUCCESS in 59m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cf29f96454ee46b1ba12a147adcfe7da : SUCCESS in 57m 43s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/29ddf9904cb548f29ec44965f4cf5fdc : SUCCESS in 1h 04m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ec3788a5135b4e8892e6b13048244b08 : SUCCESS in 59m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8dc505b1c4b54cd59d543b21c7c22d09 : SUCCESS in 16m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/339d535b96934d0ebc3e875d836d355b : SUCCESS in 15m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6075a8621be046638e0c6d994e41eea8 : SUCCESS in 1h 47m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/36bd6feff440432691cb6be0652e18f2 : SUCCESS in 1h 49m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c1e7295e27df444fa62a049564b6872e : SUCCESS in 28m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cc2bf137364a44c5ba500d6c3709cfbd : SUCCESS in 33m 01s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e5b4babd9acc408a9a9f3efc536864e9 : SUCCESS in 23m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0ba03dc042ba481fa4c3dd37fc5ac4da : SUCCESS in 28m 26s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c8091cfc0d24461aa4000f1dda19e87e : SUCCESS in 10m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1cd66ebcfef64452ab37e74664f7de50 : SUCCESS in 11m 58s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":3},{"id":"932fdb2bbc7d5c05018f812e53178335443433a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 23:00:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2d7084351dc41ce7304d08614cd4ac4b0c761b0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-02 03:24:56.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf98bdb63dc4c1192c3363eed0b4d1d : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fda837b64724469bee88c1479587015 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2074962ccee450787374e8bbe5fb57d : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7979d760d1a048818c62e737c0508dcf : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7fdc06725df487e886a6c821b821cc4 : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/562de61e36ea433383e7e213dd5ce8a5 : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/773f29d3ee944d8bb9436bd81f9a3b75 : SUCCESS in 8m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/88ba079e1b6047b5a7f28e7a3ac00e48 : SUCCESS in 17m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2d43f39d2e6347068345ff0a48a54fcc : SUCCESS in 54m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ecd46385a6a84e069bff0d00eb834a76 : SUCCESS in 56m 02s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/de872e65d5be4d1eb6517f2db96ce810 : SUCCESS in 58m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1c3e1b4afda143c7abce9512c43cbac8 : SUCCESS in 59m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/97568363152f4c06b3b28d000ef75323 : SUCCESS in 15m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f9ac59353691451991b34c7460d0ece8 : SUCCESS in 15m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/feb5d8fb107841bda7463cd5795a3ad5 : TIMED_OUT in 3h 01m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4920032627404616bd37f0a7bfcad2ee : SUCCESS in 2h 29m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f3f09b50914043aebc7cb35524510ddc : SUCCESS in 26m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fa568f6b55a4451cb0cd5c9ebfca4ae8 : SUCCESS in 27m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b25372e1c1604c778a2e5b7dafb2e7c8 : SUCCESS in 27m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4a082b9d51244988a575c0b1fe46d7de : SUCCESS in 25m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0998b316c9df4c9eb465b2b0f9861609 : SUCCESS in 11m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/092c0c3fa88f40d5a1b916b68dee5fd6 : SUCCESS in 12m 57s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":3},{"id":"c479e29096d715ca3e5939ac1f60efbb000778bd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 07:59:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e348ccae43d2398800df00abcdffcd286de231a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 12:16:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d7255c9890941b5a7329d79e96b7c69 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef6911f6b62f406e9e923a40f4dfdf06 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f535680de7b04d1f8c71e93ec0dff4c1 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8832604a3d7a4d67a0bcb6821cb1265c : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2832aea9cf05478eb97b611390799e43 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f13e121e4394409bb32231448ed5f187 : SUCCESS in 5m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a1e9f009fef94007938a9128914e48a0 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e21a5d52116493abd5225c90bac6a1d : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/228528c87a174108adca639c0a0d5d05 : SUCCESS in 17m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fe1cba6bd14a4dfb8b43fa9338533129 : SUCCESS in 58m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8ba9ebf7a8944dba81fb67f8cf5eaf53 : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ec9a5ee9b6ec4f3ba8e6e29f4759846d : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a58df9fd161a441f8cbd32d1174a0313 : SUCCESS in 1h 04m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bd90371d3f4e406991c27dda1b2a1eb4 : SUCCESS in 16m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cbc61457c145437a8cc696718ac88a2a : SUCCESS in 14m 20s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/805a8cc12ea0485ab01ea795ecc873a4 : RETRY_LIMIT in 1h 10m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/97733b66b0184159a3abb86beb25b02d : SUCCESS in 1h 58m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bdd6c5dd1c5d462891977bc1598e869c : SUCCESS in 25m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fcf32de40d3c4c188d21d678c0c76fe2 : SUCCESS in 26m 30s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/86fc235a52004e449da141abcf0e0d24 : SUCCESS in 25m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d9db54f16de745baba5a5325fd3251e9 : SUCCESS in 25m 01s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e1fc87ef3bb94fa98d3caa6bc961a3a4 : SUCCESS in 12m 09s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1a6a19bd0ff244399d7498f57a0e8ed8 : SUCCESS in 11m 40s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":3},{"id":"751c7be7e822d32c5b202d326aea5b9892fd7bd7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-04 12:27:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6b8f78ab147aea198c6e1a8fc15028af8e460b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 14:43:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8d635a5b3de4e0a927c5309b5e8dd32 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c970c34daad84c019a4814f6f3affa6b : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a2c9c67c54049178f7e9ce950a15f88 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9cb87239a2641efbd867b4500b4fe29 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/172f918605e441669bad168fa4445360 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3d846da877b44f889c8164a5cdfd2a0 : SUCCESS in 5m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2091e7c509304263b969156cc0b97690 : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ade21b8b7ea449d7b80fd6861fa98f1f : SUCCESS in 9m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/520ae0f28a574f7faa5867b803e0ef3e : SUCCESS in 19m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2ad033fa1a0c473998c8a553eb731268 : SUCCESS in 57m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/110944ad9e4b40c2919e052a36c4625b : SUCCESS in 59m 45s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/66ee7a10a0a44bfd876918f702aa7814 : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f99efbd055ec45299ac270b35ac4c825 : SUCCESS in 1h 00m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c8e7c978eae14f758061f2e09157afdb : SUCCESS in 15m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/339fb6f401f64a319c8051afb49c1c7c : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/329e5c22959446fbaf0eaa0e9af70bc0 : SUCCESS in 2h 14m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fd62bf2a494b499e97f2fc2387676be9 : SUCCESS in 2h 13m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/323dde3dd90b4b8b851e0e4bab89b6cf : SUCCESS in 28m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cda2ac15b2e54cb39531635d7c9cd3a8 : SUCCESS in 33m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e1079fd9d37d49408a6448a16884713c : SUCCESS in 28m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7458ee96607a42708aec0e3f68f4fd6a : SUCCESS in 29m 33s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2f38f0d5afb14dc5924cd504c0654a98 : SUCCESS in 12m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2afddbe19915406aa152e99a9bf03dd1 : SUCCESS in 13m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":3},{"id":"da4708221023ef15dfc1b1dadfde72f369117474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 14:47:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"00609990752d6db51f78bf0aeef9520f39d2e911","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 17:02:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdcf7eb6d3554d0eb886e4d768236e92 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ae6023e7419486fafa4770af62b246b : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1190f084a08c40a39f9c54507f78f102 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/502572efb7ed4c23b1f6fab805ad4670 : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f70266c87a54ac88304e9f22493e080 : SUCCESS in 5m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29f4705b4b58424d9c1c4b978683aa68 : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c245b3754e2b4977a69c3f4efb76d043 : SUCCESS in 9m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d149095da1ea4f1e9e0205df04622206 : SUCCESS in 17m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/880ad239a0c9454895c464672941c095 : SUCCESS in 59m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/455f22dfc2924f408faf3dbe74edabde : SUCCESS in 50m 19s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/26b7a9c55a3b41fd884774823288dd97 : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9f77efd120b440c6a66de4ee6b780191 : SUCCESS in 52m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d878d9ba34e94810b173b33a7682190c : SUCCESS in 17m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/29224ecff59b4917afbd2a7810753859 : SUCCESS in 13m 51s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3ff038e203d44904addf84566b69ec35 : SUCCESS in 2h 11m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a54caf06c73d4c2fac55e89330c103e8 : SUCCESS in 1h 49m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3f417bf2de0c4fedb257ee00f47abde3 : SUCCESS in 28m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5c501f23bc46406292448927d332f12b : SUCCESS in 25m 55s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c6c574f3100249ad82b1ee29eb3fb9d5 : SUCCESS in 33m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/68430c0506dc4cc18830d1fd2658d463 : SUCCESS in 27m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a661f9d02eb84b36a9fc02856ed6b988 : SUCCESS in 10m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/73a6bf331d7446f58b955341d1f4038a : SUCCESS in 11m 46s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":3},{"id":"2ddcec6270b8d232302a23a5a81d1c24b2e5640b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 17:02:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ab855a243e82da77d8ca19f35ba171864a781f32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 17:10:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deacb8dd2b404fd6ae53161da21a2fee : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bf62befec214489995ba8a459d68760 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7bf2ab99a2ab1b9c873436aef37d43917ecb7382","revisions":{"261b709c290e20474094c6375bf4f99601fe2733":{"kind":"REWORK","_number":1,"created":"2021-09-14 11:18:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/808931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/808931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/808931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/808931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/808931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/808931/1"}}},"commit":{"parents":[{"commit":"7c1366e85a917a9ee738cd0db574e79d5d8d03e4","subject":"Merge \"CI: Fix get_logs.sh docker logs output\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c1366e85a917a9ee738cd0db574e79d5d8d03e4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-14 11:14:05.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n(cherry picked from commit f639ad0b3514c3a6a485b2e75082ac0b3b850454)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/261b709c290e20474094c6375bf4f99601fe2733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/261b709c290e20474094c6375bf4f99601fe2733"}]},"branch":"refs/heads/stable/victoria"},"1f9373774f755b81e07eb1ed72b5d1aa7846b320":{"kind":"REWORK","_number":2,"created":"2021-09-14 13:25:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/808931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/808931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/808931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/808931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/808931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/808931/2"}}},"commit":{"parents":[{"commit":"7c1366e85a917a9ee738cd0db574e79d5d8d03e4","subject":"Merge \"CI: Fix get_logs.sh docker logs output\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c1366e85a917a9ee738cd0db574e79d5d8d03e4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-14 13:25:47.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n(cherry picked from commit f639ad0b3514c3a6a485b2e75082ac0b3b850454)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f9373774f755b81e07eb1ed72b5d1aa7846b320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f9373774f755b81e07eb1ed72b5d1aa7846b320"}]},"branch":"refs/heads/stable/victoria"},"7bf2ab99a2ab1b9c873436aef37d43917ecb7382":{"kind":"REWORK","_number":3,"created":"2021-09-14 16:14:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/808931/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/808931/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/808931/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/808931/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/808931/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/808931/3"}}},"commit":{"parents":[{"commit":"7c1366e85a917a9ee738cd0db574e79d5d8d03e4","subject":"Merge \"CI: Fix get_logs.sh docker logs output\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c1366e85a917a9ee738cd0db574e79d5d8d03e4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-14 16:14:13.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n(cherry picked from commit f639ad0b3514c3a6a485b2e75082ac0b3b850454)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bf2ab99a2ab1b9c873436aef37d43917ecb7382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bf2ab99a2ab1b9c873436aef37d43917ecb7382"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
