)]}'
{"id":"openstack%2Fkayobe~808996","triplet_id":"openstack%2Fkayobe~master~I4ed284173841def54953d832c3fcd288e1d2605a","project":"openstack/kayobe","branch":"master","topic":"typo-fix","hashtags":[],"change_id":"I4ed284173841def54953d832c3fcd288e1d2605a","subject":"Fix typo in inspector.yml","status":"MERGED","created":"2021-09-14 16:17:54.000000000","updated":"2021-09-15 09:57:33.000000000","submitted":"2021-09-15 09:55:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808996-typo-fix","meta_rev_id":"0843480ddcb4b62fc4295b420793c3d65e568905","_number":808996,"virtual_id_number":808996,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 09:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-15 08:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-15 08:10:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-14 18:12:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 08:10:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"f1015b948736b13b25ed23f949f9bfde2537f2ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-14 16:17:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b996bc7238b53296fa46f91a23e7edeb20d7401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 18:12:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2518a9fa3b9d4539a3f94b12877f87fc : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04fcc3a6eaf54bf9b55a8c0a56aaeabd : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af672e7b2c584e7db06ff0763a23034b : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/347bf63c9e83442a8fe05971f9bba577 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf7cc91215b74dd5b64d80692164e546 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd65f1371a54699b5c530ea1e4c78bf : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/38af3279d9a54784a7942ca3ffe79147 : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6a9066175af94765bb578057d93d8dd3 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/30a9de9921f0400bb2e5140815c90bfd : SUCCESS in 29m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2a6f65347a4d458aa5f4a215c0e3afa4 : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8551be89902a4a4e90ed36cc144357c2 : SUCCESS in 48m 30s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f91d1ff2536448b3a7549cc41067208d : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/21b2d8f9890f4c4a80eb6ce26ac282ce : SUCCESS in 19m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6fbd79fd9e441f3a2a3e3f8a4a6cd1c : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/18f0853455e04997bbd873f6fce5f037 : SUCCESS in 1h 52m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0350202f2cb94ebc8c4826f44f5d00ce : SUCCESS in 42m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca9de4fdcc054774a3811bc08ac0ec45 : SUCCESS in 29m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d0aecddefdf14cf7b505b49177d16f6c : SUCCESS in 34m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/db77c673fe9e443898a89a37ef583b0b : SUCCESS in 13m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/861834e752904220bee65c39ea3d8ed9 : SUCCESS in 12m 16s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"e8cf8d6a3996a44747f43d7e82707a82241559e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-15 08:10:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b3299c1da5223d0128037faf4eac29822525dca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 08:10:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"405730261c8c9709b6cb39358b7d99e33e2909a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:55:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ef525ca3939451c9b82522c87e24e8c : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e1a41972a534cdab1ef967a3724d85b : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2eb658382d64416a514eedd734fb803 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/496423406d52415185a25d93dd307aee : SUCCESS in 10m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1465038abb0a45138fcb0ee653035fff : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e02d429f9c6e43a9a1032db5300c5397 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c7039abcb61c4f91b8c05cf1dab0bf4e : SUCCESS in 29m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f0e7f3cd05d4473d82a2d95bcb173fcf : SUCCESS in 1h 07m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dabd27661b904e07b595d4b514256967 : SUCCESS in 44m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0a920e6d7bb74eb29e90256b451c6d20 : SUCCESS in 1h 13m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/92e73c4c63604fd0b2db9ac94a3d11cc : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3a69e6d47d14800a0c3895bd53bf749 : SUCCESS in 13m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/458e9ad642974d84869a3e3092547fe6 : SUCCESS in 1h 41m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bf2600ecd35044388762dfa53434d6dc : SUCCESS in 33m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc4e2a8c289c49a4a874821a8d3bf44b : SUCCESS in 25m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1d597635e4a64b319dd1b36d43813ff0 : SUCCESS in 26m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7469216dc10b4beca6abae72c0b783b9 : SUCCESS in 14m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef5e8ae7e507408cba19d13159053f65 : SUCCESS in 12m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"a526813cf662957fa540ccc6d4df18a3bfc8712e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:55:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0843480ddcb4b62fc4295b420793c3d65e568905","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:57:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6055e5abb0da40d990de2f7d2d670c2c : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc6ccba2e8d8333499b8240de05df4aeb2c9a3a0","revisions":{"cc6ccba2e8d8333499b8240de05df4aeb2c9a3a0":{"kind":"REWORK","_number":1,"created":"2021-09-14 16:17:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/96/808996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/808996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/808996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/808996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/808996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/808996/1"}}},"commit":{"parents":[{"commit":"65546beb0fc85c9c74ad0e877321230689b364a0","subject":"Merge \"CI: Fix get_logs.sh docker logs output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65546beb0fc85c9c74ad0e877321230689b364a0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-14 16:17:41.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-14 16:17:41.000000000","tz":120},"subject":"Fix typo in inspector.yml","message":"Fix typo in inspector.yml\n\nChange-Id: I4ed284173841def54953d832c3fcd288e1d2605a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc6ccba2e8d8333499b8240de05df4aeb2c9a3a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc6ccba2e8d8333499b8240de05df4aeb2c9a3a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
