)]}'
{"id":"openstack%2Fkayobe~809134","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I70d5f78871867f3a71150eacc3df1857949e6df9","project":"openstack/kayobe","branch":"stable/wallaby","topic":"network-allocation-stable/wallaby","hashtags":[],"change_id":"I70d5f78871867f3a71150eacc3df1857949e6df9","subject":"Fix typo in IP allocation filename","status":"MERGED","created":"2021-09-15 07:49:35.000000000","updated":"2021-09-16 12:52:42.000000000","submitted":"2021-09-16 12:50:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809134-network-allocation-stable/wallaby","cherry_pick_of_change":807157,"cherry_pick_of_patch_set":2,"meta_rev_id":"d6b35e4d665b99f14a0c57a5ea5695b7f25b0127","_number":809134,"virtual_id_number":809134,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-16 12:50:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-15 08:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-15 08:07:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 08:07:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-15 09:17:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac3ece152a44e4aabe222cc8b38b538ceb8a9b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 07:49:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4112d5dff6e87559b3d1f07c41fa77cd68937aa3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-15 08:07:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0480118c8d56c837194abe42bbc43efe6041bede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:17:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59286429967f4c56a2dbee6e519106ed : POST_FAILURE in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/956edf5c7e874cd78201047291845472 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9473fcdf28af49d693aa66a9c8ec402c : POST_FAILURE in 9m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db56755e644e4bf59af0454afc0c02c2 : POST_FAILURE in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8ad6290446548a48c26022d9d320f20 : POST_FAILURE in 10m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0045db412727405c9e85b0b9c6e13dda : POST_FAILURE in 12m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb232c8b376343679735f5fe1509c181 : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2a520f1b529141c2943bc99bf1c58bdf : SUCCESS in 12m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5bd3fd74f6684a75a710c83964a6cc6e : SUCCESS in 22m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/82f82f0f51594790ab74af2f75a99779 : SUCCESS in 1h 07m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7a9abda673d44659998bc4abf9ffe96 : SUCCESS in 50m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8c565eaf3b69484f8a409f89bee8aa04 : SUCCESS in 1h 10m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/79f124f2175447a5b5e2b84af6260c27 : SUCCESS in 21m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2897ac846504d3e870156bd7ccd95cd : SUCCESS in 16m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d6e6ec6c46454f7b9400064e167665d9 : FAILURE in 1h 16m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/88ec6cd853ca47ff859f414e019e00f3 : SUCCESS in 35m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9c0e19478784f408dddb1685c1b627c : SUCCESS in 33m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e2206027396d4c61ab27a48711718933 : SUCCESS in 27m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/aa7f7916ec16458786a99f787f6c9f9c : SUCCESS in 13m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc5632fc60fc49fca1cf6e0df19b37ff : SUCCESS in 14m 22s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"b5cdaa2a0217369819e5dd722c1b5a731365d2ea","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 09:34:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de9ab1f0f76b28c0ce6486803965ed7e9d1fb945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 11:31:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bbb4499d2a84ee0a89f3954a6b22199 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ce426c32dd74b4cb7da9a2fa69c30bc : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbc91f895b6b4182bf020cc5b6f41786 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7e933b12461431489dd9a00f928ed80 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/901c6cb3fd1447c5b411d9f88f36f086 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/226f13ab753e4878a4082e63ae850843 : SUCCESS in 8m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb990889572340fa923596878bf5ac89 : SUCCESS in 6m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0d7f58e732ca45be9bf54eb15e78e0ca : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/705f5efaa652480d8c6403e06fe6ee1c : SUCCESS in 20m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7263a4f621934bceb5df7cbcba59a14a : SUCCESS in 59m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e55cb50e0f3f4bb29af01ce70fe5aeac : SUCCESS in 50m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/806e3239d8184153a703cf9ce8284b25 : SUCCESS in 1h 09m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/dddc51f80127425288147db82ab2aed0 : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cb3c615865249c396c70c0c3c100f35 : SUCCESS in 13m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/03c59fb8905e40738a58a8dbb8c5eacf : SUCCESS in 1h 50m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4e6acd9bc28042e0841358b483c53f22 : SUCCESS in 32m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c2b7c3b3d3d4373b447d2d6e2f67dfc : SUCCESS in 32m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f24b5be466b344dc80e2ec3d40b32ad2 : SUCCESS in 26m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f070170d5e194b93acc1f73ec94d8465 : SUCCESS in 13m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1227bb4a849d4fcebbf61332a04c64f2 : SUCCESS in 11m 41s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"bda1da9020baf1ea87e048b521b62921fd3c2b9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 11:31:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69c95685a8956394c12bacda9eb1d930c12a80f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 12:51:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708f5f9c4aa04f8cac7e4c9cce389722 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29eb6373212349fcba758f149c343117 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e33ba36732e14056a6dd19a89895d237 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c3b84564224f6eab653e57fd0e5cd9 : SUCCESS in 8m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d204c4b2814c4071ba41f9ce85ff7e83 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/84152d288f764e3d909df2cea6468811 : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d541785d66534504abd0444211a732b1 : SUCCESS in 21m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2094676d533a4f6abe63796246bb9f6a : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de19c8e3709e4073a04ae5469db633d6 : SUCCESS in 46m 30s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6272d7febd84471d9cb43991d9433921 : SUCCESS in 1h 10m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7d575b4ae6a142b2ac1916b53b25d178 : SUCCESS in 17m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b996a797f02b4c43b09427645a2f4d46 : SUCCESS in 15m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/65674840f57540909765384aad19975c : FAILURE in 1h 16m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cab4f6ae45f04d44a2defed387660db8 : SUCCESS in 36m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/660322d48e14411ab0b066b2bf274475 : SUCCESS in 31m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/85e225f20322450db7279fce4aac85e8 : SUCCESS in 26m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5facaff6b39e4c6eb232940ba435c0d2 : SUCCESS in 13m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f740a036f7649288cff09bafbb083c0 : SUCCESS in 10m 56s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"5d1130fa183cbf76b197da0de8700c0db91fc3da","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-16 08:58:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6f87a4b27d8944992d15b78747e079d7c899db36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:52:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/569495b7bbfa4642b0e89e1d94f79132 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/571c8d1dee2c40b0a79774e3d0324dce : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a012b36226cb41f4901f2df873086baf : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46996892356a458fa2095e03067352f9 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7358e559f0dd43b79817aa072027e128 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c3a2e3165a49288ce4736d4d7962a3 : SUCCESS in 9m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a952de010bd40cb880abd5472983a16 : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dcb643055a1a4808b29a7552c8a800dd : SUCCESS in 11m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c5df2b7635c340bc870c1e1c41b1456d : SUCCESS in 21m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b662108d3fd546028f8ce0ff2af0ea5c : SUCCESS in 1h 04m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ea0c89b3b3e427e932e913fd5632521 : SUCCESS in 47m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0cb794e6de8b4495b0846b114c91478c : SUCCESS in 1h 09m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3251b0ddf99e45839686d907ac142dc2 : SUCCESS in 20m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f63fd5cba73c493e82db9bf8c507a056 : SUCCESS in 15m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/91b8d170c67d4d998161274d068810fb : SUCCESS in 1h 51m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ac4e9ad7a8cb44ed928f140a41705e40 : SUCCESS in 31m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccd64b1dca4c4ca38fbd152e9bf01af2 : SUCCESS in 28m 57s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/56e2af2cf6ad448c9a4b45ce12a0e39e : SUCCESS in 27m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/91f4cd0378dd4353bdd4848acd912361 : SUCCESS in 12m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cc28d36ea3746bea8425d546039245c : SUCCESS in 10m 54s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"be61840dbda93bb8f97051d2da4574c4cb673dac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:52:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2026c0e57dd554ab60f9e61c96aaa5a52f8f948b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 12:50:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd1eaad0090422a928a10e785c1b4c2 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/171ae6ca7f19454d82c6628d37554320 : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/896e38e1d876402ab4d322ef884fdc5a : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288ead3912a8455db0d118b5261dc4a3 : SUCCESS in 9m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3146547f21e4e49984f19afe83e4d58 : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f0800af1ef943819e1df4f341bf553c : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a65f9cffdcb46638f1316a076bfb47a : SUCCESS in 18m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6e3b0a2a838e40dcba22ce20a9c01db7 : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0f392f450c243388efc84daf25880f9 : SUCCESS in 46m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0c00e71e861e45739f6425302309ba7c : SUCCESS in 1h 06m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3ad0a524fd024d4f94ea486c95ab1704 : SUCCESS in 20m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd93ee3af311485a818a475ace7a9726 : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7bcb2c10753f4afba48235d9e4343505 : SUCCESS in 1h 55m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2f2c5505da674c7aa0609a12793c6150 : SUCCESS in 34m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/480923892f3649e487772d8e25cab370 : SUCCESS in 28m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f5a3d001512644a4a00ddd684531bc68 : SUCCESS in 24m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0dbafcf5bef443ab8e2280180bfdd886 : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac0f7301f3314e4f924a200b1b4f95ef : SUCCESS in 10m 51s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"ced25c2d5eb58cc4a918596dc5d09e8b7486356d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 12:50:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d6b35e4d665b99f14a0c57a5ea5695b7f25b0127","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 12:52:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5b6e17ed59f433fb00c5e7e5727144d : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e4806127f920778b16e9f8131fb88c8f581d5e6","revisions":{"6e4806127f920778b16e9f8131fb88c8f581d5e6":{"kind":"REWORK","_number":1,"created":"2021-09-15 07:49:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/809134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/809134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/809134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/809134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/809134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/809134/1"}}},"commit":{"parents":[{"commit":"42396caafe85c6686c5df3b2c99f6643888fa610","subject":"Merge \"CI: Fix get_logs.sh docker logs output\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42396caafe85c6686c5df3b2c99f6643888fa610"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 14:11:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-15 07:49:35.000000000","tz":0},"subject":"Fix typo in IP allocation filename","message":"Fix typo in IP allocation filename\n\nChange-Id: I70d5f78871867f3a71150eacc3df1857949e6df9\n(cherry picked from commit e8769373b6adace69b041a18a4342870a98eaabf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e4806127f920778b16e9f8131fb88c8f581d5e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e4806127f920778b16e9f8131fb88c8f581d5e6"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
