)]}'
{"id":"openstack%2Fkayobe~809135","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I70d5f78871867f3a71150eacc3df1857949e6df9","project":"openstack/kayobe","branch":"stable/victoria","topic":"network-allocation-stable/victoria","hashtags":[],"change_id":"I70d5f78871867f3a71150eacc3df1857949e6df9","subject":"Fix typo in IP allocation filename","status":"MERGED","created":"2021-09-15 07:49:48.000000000","updated":"2021-09-22 13:33:49.000000000","submitted":"2021-09-22 13:09:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809135-network-allocation-stable/victoria","cherry_pick_of_change":807157,"cherry_pick_of_patch_set":2,"meta_rev_id":"39946aea60846d1c23c98810bd464b0f1925d2ec","_number":809135,"virtual_id_number":809135,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-22 13:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-15 08:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-15 08:07:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 08:07:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-15 10:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65f1a90f4f9d2e1b41b72b76146c110f52009786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 07:49:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0bf085d4020e629e660a3a89997c0eaf60b1d989","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-15 08:07:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc2f83aac34c6057264a38215e6fedb3242c0ebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 10:12:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61fbe4135b254d08b95dbd48cbbc113e : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88726f5336434146aa4019d118da7bde : POST_FAILURE in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5da82b55fcc4d59ab52baf7665aada4 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7a441c730024b9e9a2404ce2da0f0db : POST_FAILURE in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7db9efd89a93412989b2232f6a4ee629 : SUCCESS in 8m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa82814411004dbd863b69573a1c0c09 : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c15fd38bac2c4f979569415a6d0b1fab : SUCCESS in 10m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/63266a72a52443628ca7b0a8ec576c60 : SUCCESS in 19m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/08e2cbf8acd04d03ab1d4e9276a697e5 : SUCCESS in 1h 03m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/143c62fec1eb4c3dbd0938ca0ea8a68a : SUCCESS in 1h 04m 49s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/59bb70ad9b854a9abe5fb9bfe4efa516 : SUCCESS in 1h 05m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b9be664ca2ad495b8f354abcb61d4390 : SUCCESS in 1h 00m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9207b9b78fce4b9486b1884d7c762227 : SUCCESS in 17m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/91113f813af24025968f32a3bed4eb10 : SUCCESS in 16m 18s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/102c8440cc174b7ab651876cdaee2ba2 : SUCCESS in 2h 10m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dc1148e81f784623b65aaded4440c699 : SUCCESS in 1h 57m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ae2b1561a07541c5910be054d563945c : SUCCESS in 30m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ea70e8d4e5584bcab50dc24b9c5da4c6 : SUCCESS in 29m 33s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fb21197fdcd749a2b7f4c80cdb4f7196 : SUCCESS in 27m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e43ebe59906540299779c210d8568897 : SUCCESS in 27m 08s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/77bed0f6881b453fac5cffbdac4606df : SUCCESS in 10m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5a0636859d854419a80c301e028d9f10 : SUCCESS in 12m 51s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"95c868642c0e7f3036e8c6ee06ef24a14e772f65","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 10:42:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"10f288956103e8163f05fb477a947538840be59c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 14:46:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/779d3156404a47679d8708337c6edda9 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153de686a39846fa97b4bbc64f21cc30 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e91049e970a422096ab82a867ae665e : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48242e65a2db4259aaf789264cd5b63b : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c487ba32ae6465b90d9015376ddfe31 : SUCCESS in 7m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/638261ff4e3a4471a5d77a6445bbcb52 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/96b8f50fc2a14f57ad7dd40195af5e9f : SUCCESS in 9m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75ba906d39e64ab694879ff00224c18a : SUCCESS in 24m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/95eee0c3afa04719843e9628987d334e : SUCCESS in 1h 00m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/15fcfcdeece84635bc0715defc1d21b9 : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d7bda0b0bc094d44a1b63f3497e797e2 : SUCCESS in 1h 00m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/605e6c1df8ae4a7da48dd1c3e84b135b : SUCCESS in 1h 06m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8c76123f701f4231b451e0beb5392f18 : SUCCESS in 14m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ab28723c454c488f9f9a2d99463d4297 : SUCCESS in 17m 14s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d9053efd5af7482194ee814597d3e312 : RETRY_LIMIT in 1h 25m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0e5dd071ce0341dc86bc8c07fee6303a : SUCCESS in 1h 49m 48s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f54d3b83d3e44c93a9f0344e81227cb2 : SUCCESS in 26m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3d78dd0ea1b940a38f212f67aaf0d6cd : SUCCESS in 27m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/47968cfebcf84021a84ef26a8975f758 : SUCCESS in 25m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4e77c0da16e2441d9f7138814b4392df : SUCCESS in 28m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d5694dced06e4b7aa00436be8b74d72b : SUCCESS in 10m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/beea9abfbdfb45ad905459d6e6033d10 : SUCCESS in 11m 36s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"f564e9b098d205d1626691532e3437828e431e69","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 15:08:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a2451082b7dd17bae188d04b524976caa262e5a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 17:14:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abe853a5851a472cb7651f55f553275b : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8d58f051f754685bc012d70a309672e : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bb99b3c673b4bc1b41a7842a2b9af80 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e887c151f0d4685b8a430c8a8348dab : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd33eb43021941918c81971bf463b7de : SUCCESS in 9m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c0d8766e69e4f758523515f688802d0 : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0f67e36cda834534b86dd2fba769aa22 : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71263277a61d4c9bbf79d6c3a16c46b6 : SUCCESS in 20m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5b539265f2874466ac35ff079ecbda0a : SUCCESS in 59m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5932ab73df784227b26ae40517e5588c : SUCCESS in 1h 06m 50s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e3581f13cf45482da12f7db4a4189a3e : SUCCESS in 1h 12m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e8a744b076af4c1c976250113b51f9a0 : SUCCESS in 1h 09m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/94940054837249d5a8ba22ead5ff31dd : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ddf1911aa44f4ca492c19b632dc62560 : SUCCESS in 19m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1b8dd620c3dd45e8aa74c3752a03bc4e : SUCCESS in 2h 02m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8a0f8d8315114af4b2d2d0d4271299b4 : SUCCESS in 2h 04m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/61728bf309bb49d08fb924af48757954 : SUCCESS in 30m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a6d0bc4ea4874f2ca1befcc094be3cef : SUCCESS in 31m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/da62ae60bc794e9f9646f3cd6219b28c : SUCCESS in 30m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2852bfde3eb846e9ab15df24f5f4bbcd : SUCCESS in 29m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b8d7203930524e67b034ccfaf642429a : SUCCESS in 12m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6d5a0634424d4b16a1cb418d111bd60e : SUCCESS in 11m 42s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"15111fc94e3f404d06a363f22427a5d3f21ce8ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 17:23:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbb6d1ea57073a8565cdfa7e0313260e0f55476f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 20:43:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14851457192443e5b2f2ac8adf96156b : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf3838d214c843e8b711c33208ee8c7a : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cf84f8bfe33440088126cd65c29b8bc : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c859c15fb0954469b86f723d243fcf4e : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/123c95b7e5d7428bab5c208ebfb9f4a3 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a8c4a9136d44b29a1e5bbbb3b80b8e6 : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/750f6ee4efef4fb2b02f25756c08dc38 : SUCCESS in 18m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d0319cb933834a7690c9da49e0c410f2 : SUCCESS in 1h 00m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5a8df572fa624d45b97496c1c837b127 : SUCCESS in 59m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/7b8c31bf5e1f4aceb200d136845bc696 : SUCCESS in 1h 00m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7e4962fa19984466a71f055d08d97958 : SUCCESS in 1h 04m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0ab4b3ce7fbe478288ba3d2a95bf199c : SUCCESS in 15m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/727f3b55b3234e719fa78517e9be1302 : SUCCESS in 15m 39s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/081bc8ff395d4b54b88d69f9225a8305 : SUCCESS in 1h 53m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6cf5da2eea2c45d8a8aac07134455509 : FAILURE in 26m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/959f5a73e8eb416d8c3a4c6d08b483fb : SUCCESS in 25m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bf204547989543cbb210ea97f2f1fcee : SUCCESS in 26m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fa04f8b59b0d49cea5b301808a3bc6a7 : SUCCESS in 25m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/081bd82ea94941fca0503ef0fbb827f5 : SUCCESS in 26m 00s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7ce1d19d2c3d4a99b22c6b695dd6c510 : SUCCESS in 12m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/403d9d05a69b4801a35c62d0fcda2f8b : SUCCESS in 11m 52s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"c5e1e0915d3c25ebff110715a002e5fa17edc575","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 21:02:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"79eb7913480983c26fd0e47fb9d0cfd72994d780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 01:49:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfb39429b08d42458198df767392c7ff : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d456c61a05431eae7de155f95f6d1e : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/686a1229ff50416a8ab7e20431e61e6c : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/917a51c50f1e499ba36725e1351bc7c6 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72604fff9d2f4d2e8c82bd257532c907 : SUCCESS in 7m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c639a92659e4b6fafa8cac927113b4d : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c75b725f041c466395004aa5d24fd3a9 : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/09718c6edb5142269f2470e468a2fe95 : SUCCESS in 18m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0f6ea67b3bd846779f0867b6bb32c561 : SUCCESS in 1h 01m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0db0f381ddef401f9f82c2d8de70d8ea : SUCCESS in 1h 00m 51s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/45b76c8f111a4c2886d4e0a5a3dc3772 : SUCCESS in 1h 04m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f70edba0f3cb4eab9dbf070a25cfcc56 : SUCCESS in 58m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a66de20e29a5409c889092e637219861 : SUCCESS in 17m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/49740398ac934eaa80917b90a1f97111 : SUCCESS in 15m 53s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f86040c6348847c9a51a3eb1cc244108 : SUCCESS in 1h 58m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7a275ab886594e1a8054b37c2c31331e : SUCCESS in 2h 09m 45s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d4fa8bbae3b94021a8bfaded73fa0e44 : SUCCESS in 25m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6e576f7f55084c2eb24040850fd84f93 : SUCCESS in 36m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/794d1672d5eb456797cf29a62f1c7146 : SUCCESS in 25m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f6caa1896638480bb84cadc69596df0f : SUCCESS in 25m 09s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2a1cb90a27c4445e95578b3a12387d47 : SUCCESS in 11m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5ea4865a3130415db42a43e4f61c36e9 : SUCCESS in 11m 35s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"82b68814d690b74e01b775413a567fa82ce99e84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 01:50:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"abe24462258cc573c4a22f059ae5e000241314fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 04:07:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da2cad41b4d14ffb8b98acda2ed80bed : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47d0f22e2af44e80b7b63fcc5af3990f : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1579704f4851492ea99b51135119e7dd : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e866594ff849d292a5bff91d8353c2 : SUCCESS in 7m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b856386ffcc4b3c975ba02194b2f760 : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/47f5e988939e4b1b92f02cd5cc47e186 : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/352240cc0b364cd28316ab87bce2dad2 : SUCCESS in 18m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/af334a9f813e40eaac956b429fc4f21d : SUCCESS in 1h 01m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/efddc0b5458e47eda8c2003aff905390 : SUCCESS in 59m 45s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d05de60df4af4295b9f902d529e6ceb5 : SUCCESS in 56m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/10eb348ea9a64203b5e70282b75a8a21 : SUCCESS in 1h 05m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fccbf07091094ecc833ab539025a637f : SUCCESS in 19m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/281e3f3ae4444d1b9056a3c03eeb82fc : SUCCESS in 16m 39s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/42fb782b34aa45deadad544a6f1cf088 : SUCCESS in 2h 15m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/919ee670252640268008903e41890e1d : FAILURE in 58m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4b68d6ab390e4a3bbf337fc8c4e6421f : SUCCESS in 27m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3b65be59b16740c5819fe10f0c8e9421 : SUCCESS in 46m 16s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a21abf9fc75a42e09937112e72551087 : SUCCESS in 27m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/064d3b3f18ce4d5fa45e6a5cf6366400 : SUCCESS in 27m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6a3a7773b14644b6860ee78151150cbd : SUCCESS in 11m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/932a4164d01641cab7798609d0db027c : SUCCESS in 11m 15s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"c742da65154d8e9e6dd22770b5b75a33240b1ff6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-16 08:38:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fd2eacece604d7402de9a74bd16fe4d63ee5fab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 12:45:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7260c351dcb41768be18c674c1b6cac : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6ef08dceb074916ad0850feaab5f706 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34d81b729b884609aa947248ebca7b08 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12224bcb580e4634a004d63c13fd13c7 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfcc2ccbd0064a44a5470ecf6e4bde81 : SUCCESS in 9m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5b2c173dc4f4706b37eca43d5162803 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2ce59e5e2206477faf41208808ff2d65 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/89c3ac7b43e047239c8189e6bcc5277c : SUCCESS in 19m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0339462f63dd4e0ab570a26fc4b05060 : SUCCESS in 57m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b868c146c27c450ca3f22063d56a0b90 : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/fa0e35b3be6f4d7385a39a364d4bf7b2 : SUCCESS in 1h 04m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/93f2af83f2d945c88d6cb095e5f60622 : SUCCESS in 1h 06m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9ac0ead2b3484fe9b09bf594e72a8bb1 : SUCCESS in 16m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a43ce1a1e05a4118bd39c56ba74a9536 : SUCCESS in 16m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/071dd00b45ec4e96838f0cd684475686 : RETRY_LIMIT in 1h 09m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9cd3791aaf54481daf6304416a8bf5ac : SUCCESS in 2h 10m 44s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9e38407405f94169830bb2e6253bfbf7 : SUCCESS in 35m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a926068c23ee42a6b7a1feae02d5cefe : SUCCESS in 54m 40s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5e035a19e12842a6ae834440f5ceff9f : SUCCESS in 29m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f7cede3acf2648498dde52d797ec6075 : SUCCESS in 23m 47s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/473f323255e6445b8be1e74e37400a5b : SUCCESS in 10m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/da9a2975d5154993aaf5c6cc6fa2031e : SUCCESS in 10m 34s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"a185ccdb1dd9662ac82e2f0114d1b5b35cb484e9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 12:28:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f5b000692f13ba242f38dea10fe5eddbf8fa8270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 18:26:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d2d7e397e6d4629a8cf822c110dc3b6 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad0e094129d044b899f636b5f6fa2cb0 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e9f7ec08f494cbeb677f29f36c87909 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/620ff1f7e6544ffe879f83a425edf2ef : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/442d66567e93465aa615f42cbf18e359 : SUCCESS in 6m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f0a21f7b1ffb49ae8bdf8781ffeea643 : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/341cf0a292d64b3abb287b92d8d5a017 : SUCCESS in 9m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/98ad34058b3c4cfdbf08309091f96c87 : SUCCESS in 19m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2030eb06c690427aa73d56e83baf431b : SUCCESS in 59m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9657634571d045eb967f8561c2eed323 : SUCCESS in 58m 17s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c244f9d98c464c479484c8780eddd388 : SUCCESS in 1h 14m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/873ea0f807f54e59a4ca4dd1a220dd22 : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1769ba26e35b47fbb7c046c3f28f37db : SUCCESS in 16m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/63d2004a8fb94f6194b5abcc8b01a446 : SUCCESS in 17m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5ee51caf551042beb844c0a2b803c0d5 : SUCCESS in 2h 15m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/45eba83fb29642c1add194e7ce5476e9 : TIMED_OUT in 2h 36m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2447b779daef4831962c4c3cef4722c0 : SUCCESS in 26m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ac41cd87d6134b818fc613f442c1d215 : SUCCESS in 30m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4ea5378dff9c48558d465bd92ac592f9 : SUCCESS in 27m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1975a42fbc3e4ff89781e53351fab5e9 : SUCCESS in 29m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e8871549741342778ca72f66d72eeac8 : SUCCESS in 11m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2fe811c7287749729385602758724b6e : SUCCESS in 11m 50s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"d28e31e7dd264fec4afab484eacdcd292b96f145","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 20:26:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3881d9211db1e6455eee1613f7f1cc521500f438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-18 01:07:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b331d785aea84c31bde35b3660624f77 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b32ac626f654366962c649e2e9f1f3f : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b7d9637575a4a70942c332f9df4f99c : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9db42e1ff69f494383d723f87f301822 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8edc8f61ac564a49b78b3464d5dca7b9 : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e931c9dfbe9343148bc07223c6f5aaa1 : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/065e8d8b92f94855ab82d8f7a001783f : SUCCESS in 8m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/efc7ac948102466cb17f2c7399c5c2a8 : SUCCESS in 18m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bdfb64e188934fc5bfca6d294e64e3f2 : SUCCESS in 1h 14m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/396464c314fe486ebc52c0d33487db6c : SUCCESS in 57m 27s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e590028564f1425ba58695ea884011d1 : SUCCESS in 59m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4dc8c772db384151ab248b023280bab1 : SUCCESS in 57m 49s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/46afab2d2d7d460da815a8ef4f2e8102 : SUCCESS in 15m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ac831c0de5ff43d5809de0c9843efb40 : SUCCESS in 15m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5782360e88234a4eaf6c15abf85a2b3a : SUCCESS in 1h 51m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/788dfefacedd4d548a59182e14eec0b2 : SUCCESS in 1h 46m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1bd179a5184f4fd8b3dd1eff3e062edb : SUCCESS in 28m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/82571d78a3ff4a77a56d0cc146c7d389 : SUCCESS in 27m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/93b3431a62744ca5b96201f120859eac : SUCCESS in 24m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e76ae10273ed4328b3820e97c612780b : SUCCESS in 25m 05s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4025b494feb24134abcce51cfa9b06e3 : SUCCESS in 10m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b8acf5e6b8474184ab2d0dbca8302eab : SUCCESS in 11m 52s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"d26af8e93e21f2553bbf920e3bc78b6a99d2dae2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-18 01:08:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6fda0680de3231390143689b97335f554a05a5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-18 05:21:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dffc813380154d1f80752b8805257533 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c811870d65ef41cfa68060d14af3d40b : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cbd53e19ae04d419eaa56c1048c9a25 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c8e0cfd5694cd882e290218d38e772 : SUCCESS in 11m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6bb91cf43f694c2cbb9d886ac6a614ed : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d8d05060ca443beb0ce644cf20e8401 : SUCCESS in 8m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ead7da4a73064ccf8153c30f7ae0eb53 : SUCCESS in 22m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/72a7c2c8ca844d26b1ec925685e207c1 : SUCCESS in 57m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0f9c4fae056240f0b1a0dbe4907a2852 : SUCCESS in 55m 35s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/aecf7f84cc8246a8bd262dcc04730626 : SUCCESS in 57m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/774dba8980df491690640229996e0f38 : SUCCESS in 1h 05m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2dd63d86fd674f5797b690a5aea658c8 : SUCCESS in 15m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c3d4d2b540734280aec4d43aa359d29f : SUCCESS in 14m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/64e2976c8ea544e389d2a6e9e0304dbc : RETRY_LIMIT in 1h 25m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3ea021dace8c498b84e718e5106b3f14 : SUCCESS in 2h 10m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7f61fb438d9d4abf947273b31f456513 : SUCCESS in 30m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7e7b579aab3d4ffca7199134af4d2f86 : SUCCESS in 23m 33s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8f69716137464e8e99817f8bba321524 : SUCCESS in 26m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5bd6100e03134c2f8a91170b492b282a : SUCCESS in 25m 28s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/063819665fb04a3cb47fdb457476b14c : SUCCESS in 11m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8798dfdf30fe4ef5b2eedd8158023d22 : SUCCESS in 10m 45s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"19a62d8ff2fb3db2c712a880470be70c6199cad5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-20 07:06:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ccd838a59edd59e5c11ca5f5e6b3a561098ae87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 09:31:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5f2554ca97f4b9593b4daa87a6b06f9 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c63a90b73d4423489f6319356132c23 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58a68bb22eae499fbc49129cae6f6925 : SUCCESS in 9m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/501cda78860042ac9e8bcefe9b8e74f7 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d7ed73ff68b4f8f89af1e761559e026 : SUCCESS in 11m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/68748e5875df4d6c9c643f277606e370 : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9430c03b8dbf4e2faeffe48487ab46b1 : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/29ad1d853337409b8403d77298b77357 : SUCCESS in 20m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/031907cb7d524d7fb06c31ee4b6256ad : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ee6effe8e4464af783653b0febf0a239 : SUCCESS in 1h 01m 44s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d97329faa7154dfe864270d398e0a27d : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/06d958c40362422eabba72001ad34d2e : SUCCESS in 1h 03m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/91c7bfcc1527431daf6508a4732b733a : SUCCESS in 15m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c0dadb7f3ece4312a504c5ad9ade159d : SUCCESS in 16m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a63219a559b24c7081a4a79f5f5c6bc3 : FAILURE in 27m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d41fddc0c2fe45ee9c1619b8e9c112f2 : FAILURE in 25m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c0f5add12ef543378f1a936ff3d17563 : SUCCESS in 26m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/76c2ef4e81e7493a8c274356d79ab889 : SUCCESS in 29m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/db677f1c045a4ea7b55a03fb2e543295 : SUCCESS in 35m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fb9c408d2ab54a7080b253656bcd778c : SUCCESS in 31m 22s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2dbecca78af6402da24c6fe11622d2c8 : SUCCESS in 12m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/85563cd7dea547cb904ff8d582855cdf : SUCCESS in 11m 47s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"cab5b90d73a734a020cf3b9ae0626453e8f575f2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-21 08:38:13.000000000","message":"Patch Set 1:\n\nrecheck\n\ndocker.errors.APIError: 500 Server Error: Internal Server Error (\\\"Ge\nt \\\"https://mirror.bhs1.ovh.opendev.org:4447/v2/\\\": net/http: request canceled while waiting for connection (Client.Timeout exceeded while awaiting headers)","accounts_in_message":[],"_revision_number":1},{"id":"3960215dc65bbb4c4969bad816e3135099a30ec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 12:41:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6355dbac2eb548008a5a3de513c4bed4 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a62a206aed462ca00365986488a382 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76f584e51de14e958c5eee19a4d3e5c6 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39eb0faf6ddf495aba34d8edc4112efa : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a352d0b974464eea9c59af18d815f8b3 : SUCCESS in 8m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/433a93fafb6c46668860c8450dea052c : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b625554123a44a5864e66b8e37977b7 : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9c677efb96654de2977a0771970620f8 : SUCCESS in 19m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7faaf47e799c4b25a6b37c97a7f0f91a : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a1ecc6bf897d478cb9e353fa15ae8b81 : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/bedfcaa84bb94d4e874572fb214dafdc : SUCCESS in 1h 11m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d5f46da4c2444bb688d573d95693c872 : SUCCESS in 1h 10m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2809cd70bc86471f8b165140c561a120 : SUCCESS in 20m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/98007f8f5dfb433e851d47327d5f997e : SUCCESS in 18m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/06df0ef8bd634604a506099702e0189a : SUCCESS in 2h 20m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cb90a680e5aa4ee2b4f41b7b1d149b0c : TIMED_OUT in 2h 33m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/07c18e7f9a9e481ca57583a2a6258290 : SUCCESS in 26m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/684d231707584faea7f0433a5cecd79f : SUCCESS in 25m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f8773a8933a94c04a131c7984a18f707 : SUCCESS in 24m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/64ce2183e4414362b103cf8121430d3d : SUCCESS in 26m 38s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/45ff045931054f54804f31543181adcf : SUCCESS in 10m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d8f2976b83514c6abcc9c051841eb836 : SUCCESS in 11m 28s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"bfb614b345d28d6d13d41fae89bd9a96ddde3a78","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-22 08:31:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a97d0ee61756b9d11f248235d5ade8f18e706036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 10:45:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34be6b32c8004d80808f0759dfb0b16e : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/972bc11d9f0b4259ba327c93e34be193 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/797ffcb648b1495e8e88a434f84afb67 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afab2bad9643458593166333ff37b98a : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29aadccfdc1541739be6206a63c8c8aa : SUCCESS in 8m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d366928c46664d0582cdbe4810cf87f7 : SUCCESS in 6m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f90817490106448fb6e4ad164f8f67ce : SUCCESS in 9m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/880242908ffe4b2d890a16c4f1d3ef4e : SUCCESS in 19m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/79f4a61d131a4fab900d1bcfca3140d2 : SUCCESS in 1h 01m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b997100faeb94a3a851d8384b3053bde : SUCCESS in 58m 31s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e3c8748bffc14fe7b80a9720dd4bbe84 : SUCCESS in 1h 07m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/03242db7a113429580eef13265bec243 : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0f96795db831464fb8a32355a258f21f : SUCCESS in 15m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a6c7925743a34d449dd492d60f944b7b : SUCCESS in 19m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a6802f8e262b4c759fb15fb5832acf0b : SUCCESS in 1h 56m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1c141ab89f394ad282e5408c926c6bff : SUCCESS in 2h 06m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/68c9269fd2ec4b86bde2fff333804745 : SUCCESS in 30m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/56675d2421f94cd5a8f30703edddce79 : SUCCESS in 38m 16s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e25d9cbf8da24b379ac6ffffb4c502b4 : SUCCESS in 26m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/952b96aca39842f19b9e2a5d6da74369 : SUCCESS in 27m 21s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/70b3b3b3c1514699a21fcb5506f7fc00 : SUCCESS in 11m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ff563c7725454c788ac0cfb80d8c5999 : SUCCESS in 11m 35s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"6697e0637a699af33f6c3667c0985f662adb1dc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 11:07:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33e7e1bdcd26e7a93b2e8eea13e6c18d0579369a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 13:09:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa507738af6f4587a736a309ffc82277 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/125bf81c29ee454aafba8f29d49ca4d4 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a5998a1c92544d08a586b031038aad0 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9374d9ea5b2143da9e3e5d2db9158066 : SUCCESS in 8m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0aaac4f907aa4745a78258ec8d552180 : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c2c3ceb6c4694606b743d8a6b7a72a3d : SUCCESS in 8m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a72ca9750bef4ec09c8cdc51b06396cf : SUCCESS in 19m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8aa968f69b464a1ba7bddbadbd789095 : SUCCESS in 59m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/33cdd3b8d9db4c289b2cc6ad413ab3df : SUCCESS in 1h 06m 20s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/872a7b2e6036474abde147c443c14e33 : SUCCESS in 1h 03m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c4ea378ecd774d2491f8ed9297064848 : SUCCESS in 1h 11m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b992042787fe407fbdfb0ce5bba088aa : SUCCESS in 16m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f6e23bff8f7145e8b92c6a88f8c20631 : SUCCESS in 18m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/68844c7d0b4b471699f4f0ad81564cac : SUCCESS in 1h 59m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6bea9406c31447b4ae5bc09a1405ecda : SUCCESS in 1h 57m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/733a2bbc2d8746beb346ea426edfe890 : SUCCESS in 28m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ea7f19c505684f07986a91037f0ae267 : SUCCESS in 28m 37s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/30683c7e083f4b71b154bd540eaa8590 : SUCCESS in 24m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b009e553bf164045ac9baddd194f6b0a : SUCCESS in 26m 46s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c92c9619fcdd4c8bade564e2f8f627f6 : SUCCESS in 11m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b58849eb3d5c48efb55dc88474fec9d4 : SUCCESS in 12m 57s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"4835042d5939f178c4c9ea4c825a78da6e72509e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 13:09:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"39946aea60846d1c23c98810bd464b0f1925d2ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 13:33:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77add9fb31244f159a2de8c2d21bf575 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0978e77aac35913c8e69ab227f82ba1fea4d31e","revisions":{"b0978e77aac35913c8e69ab227f82ba1fea4d31e":{"kind":"REWORK","_number":1,"created":"2021-09-15 07:49:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/35/809135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/35/809135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/809135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/809135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/809135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/35/809135/1"}}},"commit":{"parents":[{"commit":"7c1366e85a917a9ee738cd0db574e79d5d8d03e4","subject":"Merge \"CI: Fix get_logs.sh docker logs output\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c1366e85a917a9ee738cd0db574e79d5d8d03e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 14:11:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-15 07:49:48.000000000","tz":0},"subject":"Fix typo in IP allocation filename","message":"Fix typo in IP allocation filename\n\nChange-Id: I70d5f78871867f3a71150eacc3df1857949e6df9\n(cherry picked from commit e8769373b6adace69b041a18a4342870a98eaabf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0978e77aac35913c8e69ab227f82ba1fea4d31e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0978e77aac35913c8e69ab227f82ba1fea4d31e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
