)]}'
{"id":"openstack%2Fkayobe~809137","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I70d5f78871867f3a71150eacc3df1857949e6df9","project":"openstack/kayobe","branch":"stable/train","topic":"network-allocation-stable/train","hashtags":[],"change_id":"I70d5f78871867f3a71150eacc3df1857949e6df9","subject":"Fix typo in IP allocation filename","status":"MERGED","created":"2021-09-15 07:50:12.000000000","updated":"2021-09-15 14:34:01.000000000","submitted":"2021-09-15 14:32:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809137-network-allocation-stable/train","cherry_pick_of_change":807157,"cherry_pick_of_patch_set":2,"meta_rev_id":"d599843f877fbdda97a7850c662b7412120498a2","_number":809137,"virtual_id_number":809137,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 14:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-15 08:08:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-15 08:08:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 08:08:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-15 09:49:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db132e23e3c30b9bec2aaf2a77450c7f809d9751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 07:50:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"faf498d558e6ed223de73a0a177c9a039f32eae9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-15 08:08:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35d6026bfc3e853fc25ff1e7e32c20f8a8326c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:49:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e163ff00d2684d6ea278cd020cc91285 : SUCCESS in 5m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/58906a93f0ca4ae190eacea7718c76a2 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68d46159b30d4173a1cce80b47800fe3 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b318aede1cf94dbc98f40e7eecc43e18 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/947b29f649be43a8992f3a51256d5821 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d3552ffa228437b8858166c25e58ab4 : SUCCESS in 7m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a7baeda9b524516ab92b157415c34ac : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/66f42782a9d24fe9b92ca48cbe2ebda7 : SUCCESS in 10m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/33413d2712e34bbaa25c9299c4067197 : SUCCESS in 17m 43s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/bb5baeb13e2e49989da240c6c6a38857 : SUCCESS in 1h 05m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a6b3dc12149147b1a38252b681553f9b : FAILURE in 58m 01s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/047a823e18a24431995e3848ca137d68 : SUCCESS in 15m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6ab2db90efb94cd599f74da78f443d04 : SUCCESS in 18m 45s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b479191c1372472baee98b9132bf3d7e : SUCCESS in 1h 40m 49s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/e4004158440a452e8d55c0247e112c90 : SUCCESS in 24m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d2541a0dd1cb4f4b9975e1a7237a20ba : SUCCESS in 30m 06s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/355e8e6d638f42749eee705163758263 : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":1},{"id":"465794183c728a87250a8ad7547e54f98f0741bf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 11:01:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"78adfab5a4c001482b2173f250868dd59296f6f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 12:28:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ced4015e1d64402ab03434bb4f4a4f1 : SUCCESS in 4m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/966c8722c7bd443e9bbcdd71e2f4fdf4 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/397a9939db4b44fa9e7d59ff81f0e1c7 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7839ffbaee204f109f9c4089f42a66f1 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3c12598f372472d8b92940070d8d2de : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caedc032225a4a3cb6b067da95f4506c : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b93e81bd1ec54da5a401fc247375f822 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2297b3b6f6ed4f4d893178f903d2434d : SUCCESS in 8m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/43e1e3d10a03413899abe0831e0f6665 : SUCCESS in 16m 42s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/9797735f8bfa446ebdccd9eac4e836e1 : SUCCESS in 56m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0cdb1509d2a14d468ed83a9c002f2ab9 : SUCCESS in 55m 20s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/9dc0b2238ce1446086a222a1a0cca0f0 : SUCCESS in 13m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/121ee9fd5ca84aa9a827b5e486a38c45 : SUCCESS in 14m 26s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/77f65ff5d3814d869f86a795e78cde77 : SUCCESS in 1h 23m 42s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/09433943e4b446f1afb07c9518b8de76 : SUCCESS in 19m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/30a8a0a64a4f44c4812b38e69e340ba6 : SUCCESS in 26m 03s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b57b7fb15ae04977abe6a3269765968b : SUCCESS in 31m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3fea8d5c7248d0615d49aeb930d59ee77cb38b05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 12:29:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1865923f37f5d02645ccac9d0922a93077baaf05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 14:32:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/32325f0d047e4c31bdada9376feb5970 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c5684c180934fa09c09e8b9467532de : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6abfe54e1e4c46df924254acdf35953d : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df5cca1292fa4b8298447f21def491fb : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ba65d5c3e74a7f996216437ff95fe8 : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29ec9765884448cebc64b8c855879cda : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/798916d4194e434e980a13ff2d018630 : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eeee2f3336e24bf59573308d9557a2bf : SUCCESS in 17m 14s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/88eaed9959b64873ad817b3bb5fd798a : SUCCESS in 1h 01m 47s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4b143870ac524133a612b2454dd61547 : SUCCESS in 1h 01m 01s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/0494edf2dc9d41a4bd7457f0d9361aa5 : SUCCESS in 15m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/07cbdb491a8e443dbab4f507776d04a5 : SUCCESS in 16m 47s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b75edef58944457d9cab2cdc02aa4857 : SUCCESS in 1h 35m 06s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/1053802837484f05815f7813eb1c1027 : SUCCESS in 24m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0c7870da374f460bb98fd51f5f48a410 : SUCCESS in 30m 05s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/136303aaae2a41c1a035941d302c114a : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":1},{"id":"216842ec462306a16674747b9603eaccc6fe3a7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 14:32:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d599843f877fbdda97a7850c662b7412120498a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 14:34:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e2db6d8ef44095a4e481f82ffe5c29 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c28d5038f00b9bb81a92a3952ed71ada409cb08","revisions":{"4c28d5038f00b9bb81a92a3952ed71ada409cb08":{"kind":"REWORK","_number":1,"created":"2021-09-15 07:50:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/37/809137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/809137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/809137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/809137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/809137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/809137/1"}}},"commit":{"parents":[{"commit":"a40ffc4a7f816af13f97516f7387a141d0f6101a","subject":"CI: Fix get_logs.sh docker logs output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a40ffc4a7f816af13f97516f7387a141d0f6101a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 14:11:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-15 07:50:12.000000000","tz":0},"subject":"Fix typo in IP allocation filename","message":"Fix typo in IP allocation filename\n\nChange-Id: I70d5f78871867f3a71150eacc3df1857949e6df9\n(cherry picked from commit e8769373b6adace69b041a18a4342870a98eaabf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c28d5038f00b9bb81a92a3952ed71ada409cb08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c28d5038f00b9bb81a92a3952ed71ada409cb08"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
