)]}'
{"id":"openstack%2Fkayobe~809164","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I136adf6ba9d5129343c64c60c7fb21e4ac6e2a19","project":"openstack/kayobe","branch":"stable/victoria","topic":"ipa-centos-stream","hashtags":[],"change_id":"I136adf6ba9d5129343c64c60c7fb21e4ac6e2a19","subject":"CI: build CentOS Stream deployment images","status":"MERGED","created":"2021-09-15 09:56:29.000000000","updated":"2021-09-23 11:03:17.000000000","submitted":"2021-09-23 10:59:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"809164-ipa-centos-stream","meta_rev_id":"33b3e093de5e2e3d6d81007558000aade5023253","_number":809164,"virtual_id_number":809164,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-23 10:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-23 08:51:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-16 08:29:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-09-23 08:51:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 15:16:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 15:18:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-09-16 08:29:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-23 08:51:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"fa80ec06d896433ca2e33972d1622212da427606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 09:56:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ee544f12b410cb6ff86acd8d60a465299b4a4bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 11:10:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"71ec0e91d1dcb5d788dc3544188e360c91207278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 15:16:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f30007d39dc240ffa5727989781e515e : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b91f3981aad74971b1a086366d870224 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fd0b74d79414e8e97f03b892d4e9321 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c37d66deb0c0418f96bea5604694df82 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd1e39b0db04a6095ee5ef4f3b88d7a : SUCCESS in 7m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b81cf376970408497748c7a60def231 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0da4f2e81068458c9cf35edf50e59a33 : SUCCESS in 8m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2edf0000322d45b7ad012934ecb7fd8f : SUCCESS in 17m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/be6740a62e5c474c8575a6f240774756 : SUCCESS in 56m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6fb32c7a2bdf40e7957050d254ee0047 : SUCCESS in 1h 04m 20s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/409fee233bb044cb8625deff27904e1b : SUCCESS in 57m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/148ddf79e4434d0e8217626debeded09 : SUCCESS in 58m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9e0ca27c3a7747ab8ba2b69105786092 : SUCCESS in 16m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/09678ebcf1c14ed787f315278e9e179f : SUCCESS in 16m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6d205bffc50a4c26b197349b61642c01 : RETRY_LIMIT in 1h 17m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ae497dde8f4d456cb38d42e57a999009 : SUCCESS in 2h 10m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9b289d094fad4713908f4c036476440c : RETRY_LIMIT in 3m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3f7c185e65d542e8a1c71ca2a721e4c7 : SUCCESS in 29m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f605bdc82a9d41e79e373352e4f7cdcd : SUCCESS in 22m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5784931300344531a892206681a247f3 : SUCCESS in 24m 23s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/995948c9d3ed41ea8b6836e063a4ba58 : SUCCESS in 10m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9bf56edef47b41438f8824bdae070e93 : SUCCESS in 11m 01s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"47d177e7242e4fdc4529d896321e81bd7327494f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-15 15:18:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da14b39666d868977f871258dd9a61b8dc2cbc49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 15:25:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fb2f15e6634da5dcbef8fc067ccc4f0e10e9c771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 17:30:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db94596f15834a0cb45b8ee64613768f : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/977debcfc8d941a8a8ec9039077f8b35 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54d7e2d679ad442f9a18a6359cc11508 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0072bced12fc48e4937491dcc9010c99 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0885c93986a940359f39c4218f409c00 : SUCCESS in 9m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8003eea7f4c94fb5bc77e6d4ef4f6b01 : SUCCESS in 7m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9dace7e12c6c43249c1962d0752e0859 : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3c6862509b034085a22f8db290a06894 : SUCCESS in 19m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/90398bf2fe2a406d8af8f5b13bb8fa78 : SUCCESS in 54m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f862e0842ecd491f9dda6a79e551c9dd : SUCCESS in 1h 02m 53s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ca5e7b6289a744acb76360808cf82d22 : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6fe39ec1e4c2424788da3d9c81e6f928 : SUCCESS in 1h 05m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a925b93f21d44f2abbb28d8f50a482ff : SUCCESS in 17m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/de65904506fa4eb19624d683d8ea8ad3 : SUCCESS in 17m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/895d10de3d9c4b4b85745c03d8c57c97 : SUCCESS in 1h 55m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c391d9b3faa1476ea338217104d3b5b5 : SUCCESS in 1h 58m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ba2ac688b38140fe94e4f49e8fc8e2a4 : SUCCESS in 29m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7a76ea3b68ba49e0964348474e69033c : SUCCESS in 37m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/53057eced3064b1481daa2b26442ac6c : SUCCESS in 26m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/10665099ed9442949a667a0bd9f3bd5a : SUCCESS in 26m 00s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/34370f81fafd4e699ee5650fbade1566 : SUCCESS in 12m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d13531af515d489d9f48a505d809aa4f : SUCCESS in 11m 15s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"b4062e86d16fd65921b3d0a806ca308d557f07e7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-16 08:29:18.000000000","message":"Patch Set 3: Code-Review+2\n\nHave we added docs for CentOS stream on victoria in kayobe yet? This config ought to be in them","accounts_in_message":[],"_revision_number":3},{"id":"e177be30b4d669d9cde9b2511dc8dfa56433c71d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-16 08:38:14.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review+2\n\u003e \n\u003e Have we added docs for CentOS stream on victoria in kayobe yet? This config ought to be in them\n\nCan\u0027t see any yet. We have CentOS Stream notes in Wallaby and docs have been updated in Kolla-Ansible.","accounts_in_message":[],"_revision_number":3},{"id":"f87e0cb334b32f4587ffacc01e7bc0b7c06cb151","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-23 08:51:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6cad935d3ca561caccfece42f52b6d2f2f9d366a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 08:52:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d2bfcf89205ff2f7d4dfb9bb2efa54365df162fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 10:59:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/927e48a83f734b038e49e9b4cc8c1043 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6ff6fb7ad1942df9a3a7a4feed57d90 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/625150a6c9bd41d6ba2b08ff25512a11 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49c9040b46c04b84bc99eeb5deae459a : SUCCESS in 7m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d45257f7789d4cebabe65a1d341f0cad : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ed9f4af2881343d894db80032fbf22c9 : SUCCESS in 8m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ab13c6eab4c34c78a89942d3a886ccf8 : SUCCESS in 20m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/255f560446af44ba84b9a3819957ad69 : SUCCESS in 1h 05m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f9fa926b62ed4751a82a9aedbeccd1bd : SUCCESS in 57m 00s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f27d1efbed85448fb42848184f778a1a : SUCCESS in 1h 03m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7ff37c64eff34a82911d470c31889069 : SUCCESS in 57m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/92a1515831e6483a819123aa04dacf32 : SUCCESS in 17m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/53ebc332702a431ea750afa6288fce6b : SUCCESS in 16m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d5ec5f64c0ae44d48da3f5f32dc50c50 : SUCCESS in 2h 05m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5d55dadb51a04b1a917f5c0e46236449 : SUCCESS in 1h 51m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6661b1c8bc0f4089956122aec1143833 : SUCCESS in 26m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/157121210a0a42faaaf32ae3f0dbe1b4 : SUCCESS in 28m 16s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1ccf43e192854d35a3d4b5a05bde2a4b : SUCCESS in 31m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0148d341710d4d008d944be600bf1a2d : SUCCESS in 29m 00s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e6c8c55e2b43470ca589b66acde0781d : SUCCESS in 12m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/adfc0f9570414be39b9bd4a937cda632 : SUCCESS in 10m 33s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"fb4b52d7af9c286551a2d22c8496f5600ae11d1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 10:59:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"33b3e093de5e2e3d6d81007558000aade5023253","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 11:03:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa4b0df7e4943d78262e095e89a848b : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"46bbe36c4aaa6185456529f54d998ec1e1eee1e5","revisions":{"86615ca6c606e3dea600aa61c934c2da9e57cc72":{"kind":"REWORK","_number":1,"created":"2021-09-15 09:56:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/64/809164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/809164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/809164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/809164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/809164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/809164/1"}}},"commit":{"parents":[{"commit":"7c1366e85a917a9ee738cd0db574e79d5d8d03e4","subject":"Merge \"CI: Fix get_logs.sh docker logs output\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c1366e85a917a9ee738cd0db574e79d5d8d03e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-15 09:55:44.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-15 09:55:44.000000000","tz":120},"subject":"CI: build CentOS Stream deployment images","message":"CI: build CentOS Stream deployment images\n\nChange-Id: I136adf6ba9d5129343c64c60c7fb21e4ac6e2a19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86615ca6c606e3dea600aa61c934c2da9e57cc72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86615ca6c606e3dea600aa61c934c2da9e57cc72"}]},"branch":"refs/heads/stable/victoria"},"c4d613d66a0f3749de6cb4a4379261bbb189ddb8":{"kind":"REWORK","_number":2,"created":"2021-09-15 11:10:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/64/809164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/809164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/809164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/809164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/809164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/809164/2"}}},"commit":{"parents":[{"commit":"7c1366e85a917a9ee738cd0db574e79d5d8d03e4","subject":"Merge \"CI: Fix get_logs.sh docker logs output\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c1366e85a917a9ee738cd0db574e79d5d8d03e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-15 11:10:04.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-15 11:10:04.000000000","tz":120},"subject":"CI: build CentOS Stream deployment images","message":"CI: build CentOS Stream deployment images\n\nChange-Id: I136adf6ba9d5129343c64c60c7fb21e4ac6e2a19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4d613d66a0f3749de6cb4a4379261bbb189ddb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4d613d66a0f3749de6cb4a4379261bbb189ddb8"}]},"branch":"refs/heads/stable/victoria"},"46bbe36c4aaa6185456529f54d998ec1e1eee1e5":{"kind":"REWORK","_number":3,"created":"2021-09-15 15:25:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/64/809164/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/809164/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/809164/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/809164/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/809164/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/809164/3"}}},"commit":{"parents":[{"commit":"7c1366e85a917a9ee738cd0db574e79d5d8d03e4","subject":"Merge \"CI: Fix get_logs.sh docker logs output\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c1366e85a917a9ee738cd0db574e79d5d8d03e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-15 15:25:43.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-15 15:25:43.000000000","tz":120},"subject":"CI: build CentOS Stream deployment images","message":"CI: build CentOS Stream deployment images\n\nChange-Id: I136adf6ba9d5129343c64c60c7fb21e4ac6e2a19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bbe36c4aaa6185456529f54d998ec1e1eee1e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bbe36c4aaa6185456529f54d998ec1e1eee1e5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
