)]}'
{"id":"openstack%2Fkayobe~809215","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I300e37106ab4dc2ebfa70371391c99cd43c7b14c","project":"openstack/kayobe","branch":"stable/victoria","topic":"ci-du-stable/victoria","hashtags":[],"change_id":"I300e37106ab4dc2ebfa70371391c99cd43c7b14c","subject":"CI: Log disk usage details","status":"MERGED","created":"2021-09-15 21:18:06.000000000","updated":"2021-09-27 11:05:00.000000000","submitted":"2021-09-27 11:03:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809215-ci-du-stable/victoria","cherry_pick_of_change":809209,"cherry_pick_of_patch_set":2,"meta_rev_id":"dab3129d2e530bc7bb98c693e5aed061a6e63001","_number":809215,"virtual_id_number":809215,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-27 11:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-20 10:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-27 08:35:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-09-27 08:35:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-16 00:46:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-20 10:24:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-09-27 08:35:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"3b4298c0ff379643232b0d117c800fb3dc45891e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 21:18:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"178fa2c6737fb01254c4195eec5f165a4b25fb85","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-15 21:18:45.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"2922ecfe566657e24cf0c9e1d828b14a0a2c8561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 00:46:45.000000000","message":"Patch Set 1: Verified+1\n\n(18 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bd7a3db098e4998a8989fc1d193586b : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c9f98c9b454dc1ba93791be8e5b32d : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b446dfc692594960bf9595bcedb58622 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49435a4a89974f64bbe31b0fecbcc113 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08dbb2e8580846d88e0c7d45b28c6446 : SUCCESS in 7m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6e435e7025ef4c97942ab708679fc4e8 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f24b86e7ba1e4c2db619f170d6661316 : SUCCESS in 9m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f22454b45d74bfdbbbecbd2074b6785 : SUCCESS in 21m 42s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/05822e97b0334582bb9e69adfe059c5f : SUCCESS in 1h 05m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/85e966cd9bc74473b7723da2e6cde0c9 : SUCCESS in 1h 01m 58s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2eff3ffa15f1466e921b5f6a308091e9 : SUCCESS in 1h 11m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fcd9869c067241958ebbd654e24175d3 : SUCCESS in 1h 03m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5f84aa405c4d47a48b62c4a20e2c1462 : SUCCESS in 15m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/aea610ca330f442c91a1a27969aac620 : SUCCESS in 18m 14s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8988ca2234444509b37db21b780549e3 : SUCCESS in 2h 13m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/adc6c7e8b24f42dca1115d69869a80f9 : SUCCESS in 2h 02m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e966c76e8d3c4b2e9a97c37112756741 : SUCCESS in 33m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/746077243e8346d89587061573ab8872 : SUCCESS in 28m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/72cfc87f4a13403d97487ee8ddce1081 : SUCCESS in 23m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fb58bf23e19f430bab75c31e7e65bcf9 : SUCCESS in 26m 57s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8d1cde00d2a24eec98e0a76cbf0cead7 : SUCCESS in 11m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e9d6500ffdcc4bd88da2dc454c7e45b5 : SUCCESS in 13m 04s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"68cbfbde15d65277335a227138b4d7feb2c4de17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 12:25:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad820322b9e30e0edcbc3a6444ae77da12a56d14","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 12:26:20.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"b8dcc2ea1636e5801c047f46c899be4f154645d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 14:59:44.000000000","message":"Patch Set 2: Verified+1\n\n(19 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46b15c2280394ed7a7a928fe409f124b : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/846c3e05db08421790895867f9f1a053 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3878395a8d4475fb3da1498f75fe301 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3c34ef9a6354509a6b0e3e51b2f50dc : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97a4b79030de46d4883e0eeb4bed4f2f : SUCCESS in 8m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bce800578741488eb6e7408d3ffa03e3 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5506be3c79ca4f418cb74e86b37fa82b : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9eca4dbf009c4443b2bb9dfd2a7087b9 : SUCCESS in 18m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0d64c898a8be4bd6bfae4bf66d0e54d8 : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2552e1b6852c4d03986fb8c14f9976a1 : SUCCESS in 1h 13m 20s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1b20db09418e442aa9262b12d02f9865 : SUCCESS in 1h 38m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d24110b6b8b94528858467d34bc43af8 : SUCCESS in 1h 04m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e71735fc1b9e49689a61a79a73a0aa25 : SUCCESS in 17m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3ef0276f5f7943109b76bc96e6e429c4 : SUCCESS in 18m 11s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ce84d9dec14f40758bc84444f84fec8b : SUCCESS in 2h 13m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/00ad0a74aa3a4e08a8e708c18b2ff4b2 : SUCCESS in 2h 01m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/51a951905d52497ca378453ba04356ef : SUCCESS in 31m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1f53a3ac7bd749ca859d8f572d556489 : SUCCESS in 29m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a1402041150f484397703a03ed19cc22 : SUCCESS in 28m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/040d68853c634bf2951be1fee7580c1a : SUCCESS in 27m 57s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/573d83995bd64ea69ae4e3190ee9fd39 : SUCCESS in 13m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8c1e50314d1646dea161a84b015604b5 : SUCCESS in 12m 15s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"083c6cc088f3f5e59d025ec021aa1309f6b7a453","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-20 10:24:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"40610c4a92737a030dd56c956ec26e8252c17dd3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-27 08:35:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d22bb3c16a55393ae3710e228d600356e8d33a49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 08:35:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bba65c25fa4f4c635093cf5c0a27b7e8b20a8fd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:03:13.000000000","message":"Patch Set 2: Verified+2\n\n(19 comments)\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf2a0ffce7ce4e87bb11842fadc658db : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48ff369a2ff44c758d1452ecc5a2dc83 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2652dc73e3d48b1b3ca2f0ea4f178af : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbcf48eee564fdf94c7687cc877dd26 : SUCCESS in 7m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/12c73742451e4e81bf6790a264354f64 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ef6ebeef8a3c437cbcc3a773b8557cc4 : SUCCESS in 9m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81400073ec974b18bb4305b6af30d42f : SUCCESS in 19m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/581684bb288a4363b87f99d2317cc1b0 : SUCCESS in 1h 02m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ba9b088368c544efa97d0e6def6599ac : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/345c84f54ef74aa79dd7c008ba2d63a4 : SUCCESS in 1h 07m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5b82e148ccc74e598e2812127fc5808c : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f28c1719427f4ba6920cb390d9bc0b7a : SUCCESS in 17m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bc997b80d18a4954b3b33aa6be453cf4 : SUCCESS in 19m 54s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e149c4c5dc62434f863b119746bec34b : SUCCESS in 2h 02m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8db539638c484abb96c5c500779ee18b : SUCCESS in 2h 25m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/768fa90e6c9b4240aac24473ea138dea : SUCCESS in 29m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/aafbbf1db5094b9fbaa6222c3a38ed23 : SUCCESS in 34m 10s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/653700dfc3d54878a5dbc8d8a3d622b4 : SUCCESS in 28m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/99f42b37711d4fcea3d4604a15297001 : SUCCESS in 28m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4412137251934a3699f70cc312314952 : SUCCESS in 12m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dd5538ef56534569b3e33fb0a2ded362 : SUCCESS in 12m 40s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":2},{"id":"63d35a7985d6cfecba06f9efeb511daaa656d24b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:03:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dab3129d2e530bc7bb98c693e5aed061a6e63001","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:05:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3276f78c05448fb7ebcf71cceda9f5 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aabb6abf3ec33b57074b0eb0f9bfea6aa74b16be","revisions":{"4e59734d977c68c144506795c51c163294bce819":{"kind":"REWORK","_number":1,"created":"2021-09-15 21:18:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/809215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/809215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/809215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/809215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/809215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/809215/1"}}},"commit":{"parents":[{"commit":"7c1366e85a917a9ee738cd0db574e79d5d8d03e4","subject":"Merge \"CI: Fix get_logs.sh docker logs output\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c1366e85a917a9ee738cd0db574e79d5d8d03e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-15 21:15:36.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-15 21:18:06.000000000","tz":0},"subject":"CI: Log disk usage details","message":"CI: Log disk usage details\n\nChange-Id: I300e37106ab4dc2ebfa70371391c99cd43c7b14c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e59734d977c68c144506795c51c163294bce819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e59734d977c68c144506795c51c163294bce819"}]},"branch":"refs/heads/stable/victoria"},"aabb6abf3ec33b57074b0eb0f9bfea6aa74b16be":{"kind":"REWORK","_number":2,"created":"2021-09-17 12:25:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/809215/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/809215/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/809215/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/809215/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/809215/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/809215/2"}}},"commit":{"parents":[{"commit":"7c1366e85a917a9ee738cd0db574e79d5d8d03e4","subject":"Merge \"CI: Fix get_logs.sh docker logs output\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c1366e85a917a9ee738cd0db574e79d5d8d03e4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 11:01:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-17 12:25:17.000000000","tz":120},"subject":"CI: Log disk usage details","message":"CI: Log disk usage details\n\nFive levels of depth gives us details about each Docker volume. Remove\nfiles or directories smaller than 1MB from the output to keep it short.\n\nChange-Id: I300e37106ab4dc2ebfa70371391c99cd43c7b14c\n(cherry picked from commit 545f36106cc03fa00cb206077682bcd5dc23c683)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aabb6abf3ec33b57074b0eb0f9bfea6aa74b16be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aabb6abf3ec33b57074b0eb0f9bfea6aa74b16be"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
