)]}'
{"id":"openstack%2Fkayobe~809322","triplet_id":"openstack%2Fkayobe~master~I114fbb8893817b271e330de92281c9a22d70b5da","project":"openstack/kayobe","branch":"master","topic":"docker-uc","hashtags":[],"change_id":"I114fbb8893817b271e330de92281c9a22d70b5da","subject":"Revert to installing Docker SDK with upper constraints","status":"MERGED","created":"2021-09-16 07:19:40.000000000","updated":"2021-10-07 11:06:22.000000000","submitted":"2021-10-07 11:04:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809322-docker-uc","meta_rev_id":"858578a44b5bba380587212fd9faa8fa25cd46a6","_number":809322,"virtual_id_number":809322,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-07 11:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-07 08:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-04 10:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-10-07 08:55:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-07 08:44:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-01 10:48:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-16 07:53:04.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-09-16 09:32:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-04 10:05:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-07 08:44:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"5c79c62c4222c637f352f915de72c671f4065f86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-16 07:19:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9e8e6fd036115fefd08ab3caf9178ba03ea02d9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-16 07:53:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3f8c1280266c29e3121be61b1113c083e71eb2a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 09:32:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e352f65a34924769831afcfbe1f19a8c : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac72869b7c7c42cc827a422bf034b5d6 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/334aaaba9cde465e93e345904d1b8d4c : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea42d6dc32314172a368707e829edb05 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7c978beb2604f6d92db651d964b0050 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd48ca6a6ed246dd9a32209f6ec4fd05 : SUCCESS in 9m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f68fa68a72a94529b0d0ba00961dd07c : SUCCESS in 7m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48c7387d231c4b40b0604c6cbbdf91f3 : SUCCESS in 12m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/136de94693e74b679e303226c9d226e2 : SUCCESS in 29m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c1a6a0c8f07a4ead960f8a829642b6b3 : SUCCESS in 1h 03m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/356e6b7a27fb41df83b045a452cbb92d : SUCCESS in 48m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ec1f2078908b460ebc1fd1090d6684bb : SUCCESS in 1h 06m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ff070e16e1e142029d7a35063509f922 : SUCCESS in 17m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac9250d73f2d4675aa616487c1197637 : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/57792a86218d479b8cabd0df7f09b7bd : SUCCESS in 2h 06m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6c377fa9296147ac99b86b024563d86e : SUCCESS in 38m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec724d5258d448ed839a96d7e3c4524f : SUCCESS in 27m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b47d144e8fb945ad830fd6b38653b784 : SUCCESS in 25m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b92752652f354e74a42f629f7fec22f8 : SUCCESS in 15m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c21a2240bd944cbe855ac3c5d5f09f34 : SUCCESS in 11m 08s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"a8b697cfe68e1acd055ad2261d8e7f2fb3098e0d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 13:01:11.000000000","message":"Patch Set 1: Workflow-1\n\nRequirements change was abandoned.","accounts_in_message":[],"_revision_number":1},{"id":"201abea738e914db9a26cb56cef724557de42b60","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-24 11:13:39.000000000","message":"Patch Set 1:\n\nI requested u-c bump at http://lists.openstack.org/pipermail/openstack-discuss/2021-September/025042.html","accounts_in_message":[],"_revision_number":1},{"id":"a6be6253c6209a09b9ea2e9f9bad96cf6e0cc95e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 10:48:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c412c4ece2d89aaefd792920b13e40e61cca4c11","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-01 10:48:28.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"eb420132b5496f3305c6ab1e707d886d01abc569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 12:36:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d890247c3835443483151d976a313935 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a5195d23f5745d2b72eac21ca540946 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f08285716864c509322ccfda92193bd : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c179fc57b78467cb387d7530de0837d : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5a4a051addb4955b8fa7a804427e4b4 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c5b74841694590a856048c2e1ccb5e : SUCCESS in 11m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/68dc266b82174f0a8a75e4762c8edf1a : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c6e2968df36742d591f719f1bf36d288 : SUCCESS in 12m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a1167755df4b4aabbe0e49d0d96ca86f : SUCCESS in 32m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8f09dd241a0344c089adc11468bf31a9 : SUCCESS in 1h 02m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf3cf4e84aa746d2b0fcab06676468ce : SUCCESS in 46m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/582199de153d4f7daa7c1c1d340eb450 : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/340f5740a6f741b0bb7b6e5c0dc9e02c : SUCCESS in 19m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/740f8c6da38a40208768ceaa3ee0bb17 : SUCCESS in 18m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f983012e0dce420c90f9b80988033ab3 : SUCCESS in 1h 46m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7cfcfb6385d34b968148ffc145f99520 : SUCCESS in 38m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c84428946bf1459790e1c0df98a2c720 : SUCCESS in 28m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b66b36c655c742a0bd15a5b0e33ffdab : SUCCESS in 33m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6853899f9ac84dccbc7c7b81bc7794d7 : SUCCESS in 15m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02fbcd08af0d42a1914ca23453bc38bb : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":2},{"id":"36db46648daeddc4705a4b6eedd04109ad17a9f3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 10:05:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6462fbfc80eabe79a0eb56b35175df421ba22116","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-07 08:44:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"70aa26596737a4103cc0c6dece793e1f097e143e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 08:46:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7990fef73b010c97e9a2a714795677f85d2c1309","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-07 08:55:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8264b319548da3df88548801e93dc626caefca39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 11:04:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15e66a82bf764ed79e04bc53cb79f338 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b16bf6c7eb44e279311299b9171d60f : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8dc76788cda41c99cec26aa462538cd : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d49a7c12eb41a6a76d80ce89236221 : SUCCESS in 11m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be40159c92c34756a4446516df18c018 : SUCCESS in 8m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ec33a779ce4440d817c6e5c6d93229d : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/79ba67fe1b0141b4abe520a09c44058f : SUCCESS in 35m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f44c8cf5e6c246b28493f4bfbbc0c80a : SUCCESS in 1h 18m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a95f6ddc89714851b2fc99c2ee8e747a : SUCCESS in 52m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d78f82a576c64a888d4192af68b442db : SUCCESS in 1h 18m 37s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/30453be81541414283b30639746f06c7 : SUCCESS in 23m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34b131b8064b49158adf0bd32a1a1cba : SUCCESS in 18m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9c906d3377184019927879fc3b25a72b : SUCCESS in 2h 10m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3984acd97781404a9f0cb0032ee71066 : SUCCESS in 43m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c78b06789ac42e2b313f21c1ad7a8f5 : SUCCESS in 34m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a0dd61733186407db41db4065a1a7b18 : SUCCESS in 40m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/49ef39b87b474317b91fce4d0798deda : SUCCESS in 16m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7808262298ba402eb00f8e882147295a : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":2},{"id":"ad01dbb9a75f35c42d89fee205ff73f8b19bdb00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 11:04:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"858578a44b5bba380587212fd9faa8fa25cd46a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 11:06:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea99d69ad12949faa35008edcb76f14f : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f50e9b0287e1e5721569a5f0ccafc5ab31e1d37a","revisions":{"cdcd342e57e9b8004c6bdf6aec28530b164ebbe4":{"kind":"REWORK","_number":1,"created":"2021-09-16 07:19:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/809322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/809322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/809322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/809322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/809322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/809322/1"}}},"commit":{"parents":[{"commit":"480ba15e7efe6e174a5a6c91e411f16f58bd69b3","subject":"Merge \"Sync documentation with kolla_openstack_custom_config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/480ba15e7efe6e174a5a6c91e411f16f58bd69b3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 07:18:20.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 07:18:20.000000000","tz":120},"subject":"Revert to installing Docker SDK with upper constraints","message":"Revert to installing Docker SDK with upper constraints\n\ndocker 5.0.2 has been proposed to the requirements repository.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/809319\nChange-Id: I114fbb8893817b271e330de92281c9a22d70b5da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdcd342e57e9b8004c6bdf6aec28530b164ebbe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdcd342e57e9b8004c6bdf6aec28530b164ebbe4"}]},"branch":"refs/heads/master"},"f50e9b0287e1e5721569a5f0ccafc5ab31e1d37a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2021-10-01 10:48:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/809322/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/809322/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/809322/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/809322/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/809322/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/809322/2"}}},"commit":{"parents":[{"commit":"00f422c7850fde1166bae0a5f99a88494b264ee4","subject":"Merge \"Add support for apt proxy setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00f422c7850fde1166bae0a5f99a88494b264ee4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-01 10:47:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-01 10:47:45.000000000","tz":120},"subject":"Revert to installing Docker SDK with upper constraints","message":"Revert to installing Docker SDK with upper constraints\n\ndocker 5.0.2 is now allowed in requirements upper constraints [1].\n\n[1] https://review.opendev.org/c/openstack/requirements/+/811522\n\nChange-Id: I114fbb8893817b271e330de92281c9a22d70b5da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f50e9b0287e1e5721569a5f0ccafc5ab31e1d37a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f50e9b0287e1e5721569a5f0ccafc5ab31e1d37a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
