)]}'
{"id":"openstack%2Fkayobe~809329","triplet_id":"openstack%2Fkayobe~master~Icb3569bc764435127a00ee15583c6f3e4e598eca","project":"openstack/kayobe","branch":"master","topic":"reduce-log-size","hashtags":[],"change_id":"Icb3569bc764435127a00ee15583c6f3e4e598eca","subject":"CI: Remove IPA images from saved /etc/kolla","status":"MERGED","created":"2021-09-16 07:38:08.000000000","updated":"2021-09-20 18:44:01.000000000","submitted":"2021-09-20 18:41:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809329-reduce-log-size","meta_rev_id":"14815f8f68f8a95d91219abbd4e206e4b6ab229a","_number":809329,"virtual_id_number":809329,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 18:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-17 12:30:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-20 18:44:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-09-17 12:30:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-16 08:17:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-16 10:39:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-17 12:30:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bce0de64189c11663a0ac7680dc1f021a1f312e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-16 07:38:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fdee4a61fc9f4fa2cf626e2145218ddce33f47f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-16 07:44:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e17ad40310983874f223786be166b717d3eab518","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-16 08:17:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a74fa5cb7ca871fbcd2bf35bbda424c8d7bf5d71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-16 08:44:53.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ca86297fdbc5a49ce7b42c8880caac57e84d7236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:39:20.000000000","message":"Patch Set 3: Verified-1\n\n(21 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d750ebd0fc8842328d16dc109a708f54 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f08e0b2dc2014b5c85bc2d1dcd790cae : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5c7f2d5a1494c1fa0c0e0a74780a04c : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1072a27aae7f451987b5d217e569c508 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61c580801d4d489897874a87f85ff31c : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2524f68124c64145bf6e19ef4b288f7b : SUCCESS in 10m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c28d4c598c864508b3ef9c20f11722ed : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0c938271ef164a2ebc90e227418f8609 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77ba10ba7a9f44c3a7a32ef595d7f362 : SUCCESS in 30m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6af44e943d434de7982f4e2d52f43bcb : SUCCESS in 1h 10m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22a6f4b3584c4f92b5decbcae467c95d : SUCCESS in 48m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/54b4bae59b7645c08bad7edb116b9961 : SUCCESS in 1h 13m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1a9c5ceec35944448be6376259d1fba9 : SUCCESS in 18m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79a94f4602d748ddab8c475058b85386 : SUCCESS in 15m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/75a0993bbf67426b9bd26e7d2b6181ce : SUCCESS in 1h 48m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7b2e9883ae774ab68da984f18fd17df8 : FAILURE in 16m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40e3ac978c064524b6da5876def66b8c : SUCCESS in 28m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1794f666dc994faf90ae93724ce48882 : SUCCESS in 30m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d5fc98efdcdf423696055139b19d539d : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f9e9bdc7185476fac1e6f6bb7508234 : SUCCESS in 12m 11s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"df7fb9a749dfdd98476fc7e4123373afb880d95c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-16 10:50:05.000000000","message":"Patch Set 3:\n\nrecheck\n\ndocker.errors.APIError: 500 Server Error for http+docker://localhost/v1.41/images/create?tag\u003dmaster\u0026fromImage\u003dmirror.gra1.ovh.opendev.org%3A4447%2Fopenstack.kolla%2Fcentos-source-bifrost-deploy: Internal Server Error (\\\"Get \\\"https://mirror.gra1.ovh.opendev.org:4447/v2/\\\": net/http: request canceled (Client.Timeout exceeded while awaiting headers)\\\")","accounts_in_message":[],"_revision_number":3},{"id":"18086ed4298386ee39089dd4d5d57f326cefbe47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 12:45:59.000000000","message":"Patch Set 3: Verified+1\n\n(21 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e391641aa470467cba9f98607b1ac45c : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa07b84c5a746dc8d61f0a286803720 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab6abb1ec327409c9819264432387332 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db0e601b989e4922ba47c54dcde31f5b : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8d26fafe652404ea377e74d6f855cad : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db5af7a5bf984747b5bc1773932299dd : SUCCESS in 9m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0e79f4e780374ecaa053beebe5014a51 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6f0c052357f482fa764def845cf2f75 : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a6d24c7f622d4626b2f65cd45e9cb2c5 : SUCCESS in 30m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0133cebbfc854f85b162f1e814d4d51f : SUCCESS in 1h 04m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bede14234de74e8880e8c2d6bd6d1604 : SUCCESS in 56m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e2805bd1f3b94e389feab489692368c4 : SUCCESS in 1h 05m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bf4261ea2c5a4a68a8cfd0aae549b395 : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e9a400e26974e369895bad2dea9eda5 : SUCCESS in 14m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/db21ca400f394198b16e678eb40c628a : SUCCESS in 1h 47m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8d0db3b813c74f38bddb740f17e3e9f5 : SUCCESS in 34m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c88f0fd38f864b8e96d2e65c737eb5f7 : SUCCESS in 28m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/40f2a3aa1df441778b4f2e2bd8f2212a : SUCCESS in 29m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/57c93430c0be4b189feb1df166a4ae36 : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8557f4293d65447fb2f7b80e992b33c6 : SUCCESS in 10m 58s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"e2cb86129e5046c50fa5b31df58cd9671e50df94","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-17 08:05:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6916229efc139e53528f4af86063cd616b7e8d84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 12:28:41.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"37aa24937fe05fe09a000037903d31478f29f4f6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-17 12:30:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"750415532bd80f2d89d83a40a2ecc3b72775c56a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 14:41:27.000000000","message":"Patch Set 4: Verified-1\n\n(22 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f5e0b91973244f3b4e465cc9abd2a6b : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc8278bc329742cc87a7afd20812ffb9 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/720eccb9403e496a890e263c3e84ab77 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/376fa07b8a96484db4db7faa38659816 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/697a3de55efe46ae8cb36c1f7b7af98c : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b50b995ed90e4b7fb69c23dcb409db01 : SUCCESS in 10m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/edb46b42e4ef433e9aaeac899063a7b3 : SUCCESS in 7m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6210a90a6f4240d5bc0b2ae19efa7fef : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5d0facd2b46c4187a1cd5ad99b8b3c80 : FAILURE in 7m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/94f3dbda80da49f1b30669041209d177 : SUCCESS in 1h 12m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7a1149284754c7697a5072411338f49 : SUCCESS in 50m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5c8c67ea7ada4ad498c466b3948ea7a1 : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/58b7ecc040234c9cb1d53febc20f9bbb : SUCCESS in 21m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/665ef1cad53c4d8dbf1c6238748fdd97 : SUCCESS in 14m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4d20f0d5026546ac9fc2523077c1f3bf : SUCCESS in 1h 56m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/07b20ee5036244a0b353da79745210c6 : SUCCESS in 33m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a6360dad0d54241b5db0b49755cd815 : SUCCESS in 28m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/13453669328a4adea4a9587e54ad6481 : SUCCESS in 25m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d8bc47568c494e0fbeb5b42ef3e5ba05 : SUCCESS in 16m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e996e643979047d998115aa91ecbc689 : SUCCESS in 12m 52s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":4},{"id":"98f77fa5ac52c83b43139a63c0b7ff732b601636","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-20 07:07:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b4898dace578a338ae03e388865a1cd10356122a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 14:36:25.000000000","message":"Patch Set 4:\n\n(22 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2af1fe12cfca4516bbf5a59c3d7f88d7 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1812fb8807884c29a41caaa510722160 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b62a9cb043d447c18001b3be44f0c359 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea4a411700d540f69e2ea7dfba13c7b3 : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af8961af00e14e9a872459b680b93098 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46f8626ccd5544569cbb1c4ec4bb3e39 : SUCCESS in 8m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8e00006224b41f98da80f9b2251d8e7 : SUCCESS in 6m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e0b534d224043c7b07b302937b51388 : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e4567cb50914e0a9943a1e6753d9746 : SUCCESS in 29m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b2738cb42dbb492f9f21f9845d7da61a : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a32394ffd8044419ef18e823dc20395 : SUCCESS in 42m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b6ff4984c0c54b08a77547002ed0d74f : FAILURE in 27m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2bdc749d3f844db591ffeb64e5366ad3 : SUCCESS in 21m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05c3bbd96e7e4e97968564d809127ef3 : SUCCESS in 16m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/01391b7e634a48b1a2fcc307216dd27d : SUCCESS in 1h 41m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/32734841ccdf4358a240e0a46f50da3b : SUCCESS in 35m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6929bd9fcd0b4a088f2a6ed8e64f4e9d : SUCCESS in 33m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ee481e5c00f54926952893b0eb9fa5c1 : TIMED_OUT in 6h 22m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/835dffb2155643f9a41bfbe3e2da8d13 : SUCCESS in 16m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f29cf01974ae4a5c8fc920184336ff57 : SUCCESS in 15m 00s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":4},{"id":"6a63e1590ced23431785f3c6491052ea3418ad6d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-20 14:56:57.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"af2e20ff1739ced430bd960f164486423d8d0fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 16:43:37.000000000","message":"Patch Set 4: Verified+1\n\n(22 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f82e31b3f82d4cbdbaaae427cad74955 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5597623987d499281a0d1f3eecf2827 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1931640ac8e44eb3b3f15bade26a89a2 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d697e1b1309d4b1e875ebf0f1e224876 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/646ab6c95d0544349796725287b4c63f : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/319ee2dd70fb4f34888ee30b782c7896 : SUCCESS in 9m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e542330413104972824d89a8bdd223c0 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/78ec44f2b18f45f9876f9aaebc32bd5c : SUCCESS in 10m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/340aa325c9e745dea4e571e598bfd054 : SUCCESS in 29m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/30ec1f7f9932467bb49dcac39b3efdcf : SUCCESS in 1h 05m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36cddedddaa14880b6b2c0f0310e6832 : SUCCESS in 45m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/52676bc78d664992816fa8ddc0207df4 : SUCCESS in 1h 04m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5ccfe39c89f94f9a9e40f11e77662344 : SUCCESS in 20m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f31851dc5288440a8a1a6a8afeb06d53 : SUCCESS in 18m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4f362d689cc44c38ada60ce8583582dd : SUCCESS in 1h 42m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/84cdd3244ed64286ae2658cab02ffb13 : SUCCESS in 37m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b98e01a368344cd6aab75ed221e10bf9 : SUCCESS in 34m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d0c16667b42045c99c3416a70f290ccd : SUCCESS in 25m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/593538fbce4f49278507e59ea5ae10dc : SUCCESS in 15m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e5ad8c9b3e94515b1435ca3ee044375 : SUCCESS in 11m 23s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":4},{"id":"0264eccc53c1964c60be9e8bbe54618acab07025","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 16:44:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ec4d8d6addeced3eb32567f1b344398d8ce667d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 18:41:52.000000000","message":"Patch Set 4: Verified+2\n\n(22 comments)\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e06215daa09b4ddaa29dd578f8c7dff4 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10781df3925f401c9b33c9b2b480a79e : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fd784a161e641b989073ca459b86403 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a8169948269469f8f9cbc51e924c183 : SUCCESS in 9m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61f61929750f43a5bb115a854ff06841 : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e42fd9d4dc0c44b1a4f9f26afe40df1b : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/98554a632bec4413b685a930963d3596 : SUCCESS in 29m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2ee4fbd21a3641fbbfe3f40ab8d35458 : SUCCESS in 1h 06m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8158dd9858c74358b2ceb3dfad5a276b : SUCCESS in 50m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1252bcf8f7b1490294f244e665d3c73b : SUCCESS in 1h 10m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/82533ec93cf6443bb8c43d35a5034f47 : SUCCESS in 20m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa589f76fa534e259770c8baac8c0207 : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/95cb1b477e904b029f18b3efde93bff2 : SUCCESS in 1h 54m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0bdb54cfac564751aa09ac8e4625126a : SUCCESS in 36m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ef34acad91b429d920a7957dc76b072 : SUCCESS in 31m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6d5368701173490c92087d4f31b99087 : SUCCESS in 28m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2fd8ad4778f04e53ab2b9131ecf4f793 : SUCCESS in 14m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99420247199643c38e481f059fe1a695 : SUCCESS in 12m 08s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":4},{"id":"e03fa361b2fd88980a7d028bee22a352446046ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 18:41:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b8434763a27268a9ed96b64105ec27ff26bb06be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 18:44:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26785bb73ea34c278545b34880b804e6 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4ef6e59ef582e3e492ba0aa16d6f9698c8be1d5b","revisions":{"f37e09d37b8a1b05cc0fa8be1064e2e1bc1286ef":{"kind":"REWORK","_number":1,"created":"2021-09-16 07:38:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/29/809329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/809329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/809329/1"}}},"commit":{"parents":[{"commit":"480ba15e7efe6e174a5a6c91e411f16f58bd69b3","subject":"Merge \"Sync documentation with kolla_openstack_custom_config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/480ba15e7efe6e174a5a6c91e411f16f58bd69b3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 07:34:10.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 07:34:10.000000000","tz":120},"subject":"Remove IPA images from saved /etc/kolla","message":"Remove IPA images from saved /etc/kolla\n\nThis saves 156MB of storage in each CentOS job on the master branch.\n\nBackport note: ironic-ipxe is Xena only. Storage space saved on Wallaby\nand older will be about half (varies depending on Linux distribution and\nOpenStack release).\n\nChange-Id: Icb3569bc764435127a00ee15583c6f3e4e598eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f37e09d37b8a1b05cc0fa8be1064e2e1bc1286ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f37e09d37b8a1b05cc0fa8be1064e2e1bc1286ef"}]},"branch":"refs/heads/master"},"a96fa6055c36f0833776da8f46699034ea046c62":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2021-09-16 07:44:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/29/809329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/809329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/809329/2"}}},"commit":{"parents":[{"commit":"be80bab40a85a4f652ea0c56531492255f540a88","subject":"CI: Log disk usage details","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be80bab40a85a4f652ea0c56531492255f540a88"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 07:44:16.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 07:44:16.000000000","tz":120},"subject":"CI: Remove IPA images from saved /etc/kolla","message":"CI: Remove IPA images from saved /etc/kolla\n\nThis saves 156MB of storage in each CentOS job on the master branch.\n\nBackport note: ironic-ipxe is Xena only. Storage space saved on Wallaby\nand older will be about half (varies depending on Linux distribution and\nOpenStack release).\n\nChange-Id: Icb3569bc764435127a00ee15583c6f3e4e598eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a96fa6055c36f0833776da8f46699034ea046c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a96fa6055c36f0833776da8f46699034ea046c62"}]},"branch":"refs/heads/master"},"66453b02af12b77d7c54587134b9f1f64fd9134b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-09-16 08:44:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/29/809329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/809329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/809329/3"}}},"commit":{"parents":[{"commit":"adea8ec8cf36d2eca0b623845289317c678f793d","subject":"CI: Log disk usage details","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adea8ec8cf36d2eca0b623845289317c678f793d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 08:44:43.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 08:44:43.000000000","tz":120},"subject":"CI: Remove IPA images from saved /etc/kolla","message":"CI: Remove IPA images from saved /etc/kolla\n\nThis saves 156MB of storage in each CentOS job on the master branch.\n\nBackport note: ironic-ipxe is Xena only. Storage space saved on Wallaby\nand older will be about half (varies depending on Linux distribution and\nOpenStack release).\n\nChange-Id: Icb3569bc764435127a00ee15583c6f3e4e598eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66453b02af12b77d7c54587134b9f1f64fd9134b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66453b02af12b77d7c54587134b9f1f64fd9134b"}]},"branch":"refs/heads/master"},"4ef6e59ef582e3e492ba0aa16d6f9698c8be1d5b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-09-17 12:28:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/29/809329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/809329/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/809329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/809329/4"}}},"commit":{"parents":[{"commit":"545f36106cc03fa00cb206077682bcd5dc23c683","subject":"CI: Log disk usage details","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/545f36106cc03fa00cb206077682bcd5dc23c683"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 08:44:43.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-17 12:28:02.000000000","tz":120},"subject":"CI: Remove IPA images from saved /etc/kolla","message":"CI: Remove IPA images from saved /etc/kolla\n\nThis saves 156MB of storage in each CentOS job on the master branch.\n\nBackport note: ironic-ipxe is Xena only. Storage space saved on Wallaby\nand older will be about half (varies depending on Linux distribution and\nOpenStack release).\n\nChange-Id: Icb3569bc764435127a00ee15583c6f3e4e598eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ef6e59ef582e3e492ba0aa16d6f9698c8be1d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ef6e59ef582e3e492ba0aa16d6f9698c8be1d5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
