)]}'
{"id":"openstack%2Fkayobe~809660","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I300e37106ab4dc2ebfa70371391c99cd43c7b14c","project":"openstack/kayobe","branch":"stable/train","topic":"ci-du-stable/train","hashtags":[],"change_id":"I300e37106ab4dc2ebfa70371391c99cd43c7b14c","subject":"CI: Log disk usage details","status":"MERGED","created":"2021-09-20 10:22:49.000000000","updated":"2021-09-27 13:27:43.000000000","submitted":"2021-09-27 13:25:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809660-ci-du-stable/train","cherry_pick_of_change":809209,"cherry_pick_of_patch_set":5,"meta_rev_id":"2159d3a9bc81cf5e790b64caa02e8573642c24d9","_number":809660,"virtual_id_number":809660,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-27 13:25:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-20 10:24:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-27 08:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-09-27 08:35:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 10:24:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-09-20 12:07:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-27 08:35:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"ab27bcb1edfe750500be267f9e0a828bf7adcbb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-20 10:22:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9518782d6088b53c38244a8c1ef95fb8c01c43e6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-20 10:24:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aaa780499d3cc8b17aff43dc264ac39560c19f8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 12:07:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13670550b2f149dbb8b0169ffec64f36 : SUCCESS in 4m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a8fa2d92a8e94174aa60c80a457a78f9 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/902910d2afd14bffbe637a0286188098 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67e621190bcb47ebaa2ea5b4a4f747f2 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ddae42a1811e4d8d881a07e3eda185ee : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ebdf1b23e3d4d5aa9c7a8df7fc54054 : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3d22400edbd640c3987d8f858c849722 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f1c51be1b3bf4ee88db72f5ae4a934ab : SUCCESS in 8m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e0fd4ceb87e34e05b1ac5bb4a7b073c0 : SUCCESS in 17m 12s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/3c666352598a4bc593ee0f19d43352d3 : SUCCESS in 58m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fdeaaba4ac6e4bccb77bab48aef16032 : SUCCESS in 57m 21s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/5eafeec86baf4852bc44616a9ff1281e : SUCCESS in 17m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2f57cd557c3e40108d310cec86f1d1d1 : SUCCESS in 16m 13s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/4d7ed041bee64737961b20c75df0f2b0 : SUCCESS in 1h 38m 28s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/9f9df4b6b6454dc3a3afdd7e622334bc : SUCCESS in 22m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c445d7902732401a978ea414c3290368 : SUCCESS in 46m 33s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0fd734a4f0fb4ded86e766bcdd6f073b : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":1},{"id":"90c6f7e1f26e7adb49c1ff4f6775d7390ade9592","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-27 08:35:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1773a919bba6e1ed7ad547b3f54459e0a4cbbc94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 08:36:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9a57eb5c09e0a311c554cab2bb7b7e2c59e0766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 13:25:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/47962d1fc13e44cc99f4089d2dfb41eb : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d587d74c0ea54a699c3352f0b0df5140 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6c23f99d8a47f1834742a2978ab271 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef9d48f341fe4fed909d672ed6a5d268 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73442d0406914833bafe627afdc0fc27 : SUCCESS in 8m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29b0dcfc97a14cf5bac80b8b809f958a : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4533792520544ddbff74e4806c7f8e8 : SUCCESS in 8m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/09210fa1d70d4e90a39479284c1a501c : SUCCESS in 17m 22s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/297854bedb524bf1b8c2beec9fc7411c : SUCCESS in 1h 07m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6cf138d032e24eba94237a0e7d6cc017 : SUCCESS in 56m 40s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/24bbc3b84d414d7cbbfd422d3b016499 : SUCCESS in 15m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a90c19f6502744e7977f73b9eb2d9322 : SUCCESS in 20m 22s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f122e33511fa44c09a9d4c2eaacef46b : SUCCESS in 1h 44m 49s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/003991c1c66949c7a12cc3c425a9c32a : SUCCESS in 23m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cbb99088e6f445bb80e52fa7780bc5c5 : SUCCESS in 34m 42s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/eeb621b2299846c2bc42557bceff8163 : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":1},{"id":"1c23f2e4fff3afa2c78f9e969704ffbaa7631ccd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 13:25:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2159d3a9bc81cf5e790b64caa02e8573642c24d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 13:27:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1915e85bbe6b47d0ac693ad3019775bb : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdc78986c72eb1603204e1001e9787c35cbb4e06","revisions":{"fdc78986c72eb1603204e1001e9787c35cbb4e06":{"kind":"REWORK","_number":1,"created":"2021-09-20 10:22:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/809660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/809660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/809660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/809660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/809660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/809660/1"}}},"commit":{"parents":[{"commit":"4c28d5038f00b9bb81a92a3952ed71ada409cb08","subject":"Fix typo in IP allocation filename","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c28d5038f00b9bb81a92a3952ed71ada409cb08"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-16 11:01:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-20 10:22:49.000000000","tz":0},"subject":"CI: Log disk usage details","message":"CI: Log disk usage details\n\nFive levels of depth gives us details about each Docker volume. Remove\nfiles or directories smaller than 1MB from the output to keep it short.\n\nChange-Id: I300e37106ab4dc2ebfa70371391c99cd43c7b14c\n(cherry picked from commit 545f36106cc03fa00cb206077682bcd5dc23c683)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fdc78986c72eb1603204e1001e9787c35cbb4e06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fdc78986c72eb1603204e1001e9787c35cbb4e06"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
