)]}'
{"id":"openstack%2Fkayobe~809769","triplet_id":"openstack%2Fkayobe~master~Ifd75f704177751ec740475bcce8a9f7b8a1578f3","project":"openstack/kayobe","branch":"master","topic":"ci-molecule","hashtags":[],"change_id":"Ifd75f704177751ec740475bcce8a9f7b8a1578f3","subject":"CI: Avoid molecule 3.5.x","status":"MERGED","created":"2021-09-17 15:24:56.000000000","updated":"2021-09-18 01:24:26.000000000","submitted":"2021-09-18 01:22:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809769-ci-molecule","meta_rev_id":"02f5f2bcb18127b5ce06d1e0e2442a366dfd8063","_number":809769,"virtual_id_number":809769,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-18 01:22:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-17 16:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-17 16:31:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-17 16:08:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 15:59:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-17 16:31:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-17 18:30:32.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"}],"messages":[{"id":"91c9f6552ec81f34ea45c601802ce41190c5e8bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 15:24:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03afe9396c4d764e7aa568f940cbf5428eef221b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 15:59:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a786958dd49b46188113fefe18d3a03f : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f85a1cd4a2f43be87de658f520d15bb : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4076c33948de40a492cbf6a58083aa21 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbae053f9ed44079b20a033c43f79ad6 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77132956f1e74e90b880e842aca898c1 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a5d867358d8412b86e571ac8b24d3c0 : SUCCESS in 9m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bbead1365f6d435fba484850fceacf7e : SUCCESS in 29m 55s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"6eb31c4563e68a8a84483afcd676711f6aed85c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 16:04:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cda2e63de08193e841da02915f000db20953ce38","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 16:08:12.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"4fc6c58e5716c26a1ac9d2475bb1e3b9d2b5dbad","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-17 16:31:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4c5065bfd82bce3b2d400cb7a3160d183ab49d40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 17:59:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b347cf2fbe1d473b977a02f87fe5ee4b : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2de5c892841480098e6e72edc9829d0 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7032ab41b5944df187b9800ea46fea6c : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95d254557adc4f8c8270487813d79095 : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f151e23e12b45aca7205b5b35a8fb01 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/215f98e67c8a4241bb1e71a32c769cd5 : SUCCESS in 11m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/717823cd998e45f58a39e80f5d5f26dc : SUCCESS in 7m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/21936cbabead4cab86f83d05a23a5c2d : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bf20f457d8f04a679873ff91c9520817 : SUCCESS in 31m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fe5147ef395e46a8964f4c3a5e61890a : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/413cb9a090754eb9b10b705367e42511 : SUCCESS in 44m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/597fc2de7f9444f18ee4e045faaa502a : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2f66ffbd878d4150aff4dc62e8a6e58e : SUCCESS in 19m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/321493503f2244a8a9debe61b72ce2e9 : SUCCESS in 15m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e97cebef21e40b79c11e27e628796e5 : SUCCESS in 1h 51m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a06421c0acb24b17a8ee035608901f14 : SUCCESS in 36m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc16a4d1ca4142cc85891737aef15f98 : TIMED_OUT in 1h 31m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/04401dfa5ca640368483e4b8cb5a1368 : SUCCESS in 31m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/51d9999a8082469390389cda8d19a06f : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69f107903234462eb81ab2f7b0382716 : SUCCESS in 13m 12s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"cc5d5470b4503019e8a74cc6d8aed9bd11a9ff5c","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-09-17 18:30:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cdfbf85caed1aad68bcac2b7363365108428bbab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 21:05:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c599ef618cc94e80ad3a1f1767b2034e : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d79dcd81ba5d41fbab8cdd0f9cf21649 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/def3a77ce853493780c676c182d2f6c6 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d41cfb25bd7d4593b78c4a96b07ed2ee : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95dc487db6724fd4bad5a6f166e669f5 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b8f9cbb54d43c8a92e7284dc6101ae : SUCCESS in 8m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1cd6da314ace40c2a27270433cdf37f3 : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5baec1f026974e34875d8998a9405eb0 : SUCCESS in 13m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/941ba6df25634654be0e617d9db1adb8 : SUCCESS in 31m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ea5bad6061284921b326aa94e3971588 : SUCCESS in 1h 04m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bd99c47209a471ab4060fd5ace5f2a4 : SUCCESS in 45m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4c1969a251f34c99aec8e6ddfe2ba644 : SUCCESS in 1h 01m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ada93f7a86924fe3b9b8341957cd245a : SUCCESS in 19m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0dffd3713864ce99d5893803ce407eb : SUCCESS in 12m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7fc0314dada649f8948005ad4768ffdb : FAILURE in 2h 30m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0addefbb6e14427098cccf5970bdd58a : SUCCESS in 35m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/470470f04a604810ba85609b0da868d2 : SUCCESS in 33m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6d5ff975c38b4ef9b6d630428dac9447 : SUCCESS in 26m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e65ab5c91436418cb1311548943efa38 : SUCCESS in 13m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/512a82f1cb4c4323ab1613acff337fef : SUCCESS in 17m 58s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"1ea8a53ea0f8f4a9da380b7638a051acc72bbc8a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 21:27:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4d3fb22189f3ebeafc14cc3a19a458f53c7fe47c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 23:17:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20e86ce857b34089899f90e51ffe59b3 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b8db943cf344179a868f4f9ea163600 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0504b9bfdffd4b3a8517cae5da51ab6b : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a548cd85d8c4c22b4207cf980a2d06d : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2561e9a8a3de47bb82cb29453bcb35e3 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787abd2f610b4609b672df52bea5226b : SUCCESS in 9m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a1cdfef63df44ac83d968b2b2a24aa5 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f9eaec7d7cdb4ddca7e8d5bab098322e : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa8a4c6e5659440cb3cd1e9ff1f1815c : SUCCESS in 32m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4ad00bddc1124228addb79446edd23a9 : SUCCESS in 1h 06m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77b26290d0354907bbc75941d2b2046b : SUCCESS in 46m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6ea55e832f284cef99ffd995988cf5f6 : SUCCESS in 1h 08m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c3b56fd9155849ddade2ea2a4c4cf560 : SUCCESS in 19m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25d0ef054f564c34936e877de76e9615 : SUCCESS in 14m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/052282d2c93c436cbfb574f4bc7d70a4 : SUCCESS in 1h 48m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dbe73e4672fc40c8b0e1595203531532 : SUCCESS in 34m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16b9ca278ae74c5f816f8cf2765175f0 : SUCCESS in 27m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6a4af17f79774e8fb042117dcbb18840 : SUCCESS in 27m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e15dc6d3361747b29e6393638551873d : SUCCESS in 16m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c02262b8495f43e8afa14e340c16ce76 : SUCCESS in 11m 44s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"3d4f112adfe7ed1a885e21f7c30d5edc0f456761","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 23:17:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c9cf6a26072f02827beca1fbbfd577064b875145","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-18 01:22:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f1170e93915466396ae0cabc82aa915 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4b0997a2337427a95c24fc2187c49f3 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c31e9cecc00a4fa0b1b51e6d3cdb173f : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb5c12f0b3f34943be891730d3a88f36 : SUCCESS in 8m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a096ebd624f24fb4a8ad1d0ae397e113 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b4873777592d4ab6ae85cf7522c91ba5 : SUCCESS in 11m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04349e3cff1042ce8b69f64a91bf4aa4 : SUCCESS in 33m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b3140a34b49f4d33b6769d622f630ba3 : SUCCESS in 1h 06m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9da243851c9e422a98d3242a51c93b4e : SUCCESS in 46m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1e06a7a3ab964cffb7cebae0075b043d : SUCCESS in 1h 08m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/438f40125bcb41cfab664c651bb5ef42 : SUCCESS in 20m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3253fa66a13e4b92a4e61f5c07180ded : SUCCESS in 14m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/09e1d6fa8f2f49e7a4468549ffdbd35e : SUCCESS in 2h 00m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fb084f6eaec74692b3ac9674f3644076 : SUCCESS in 37m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c0abea542e940d58beeb8323b04425a : SUCCESS in 27m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/352cd282bee14798a53cb4a38b5f1170 : SUCCESS in 26m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d5e3f69342de4b53a79e73789670c0cf : SUCCESS in 13m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63171a6f8706423cab7a9349643f1501 : SUCCESS in 10m 32s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"cdc47fea94a570292423a89eec9bbc9b907f8eaa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-18 01:22:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"02f5f2bcb18127b5ce06d1e0e2442a366dfd8063","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-18 01:24:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9830b566a84e949926b8d215bdd4da : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81d6fa9a07c4f20d6bc244799d149101d0f62dd3","revisions":{"2e5eda52fe4ea998b00760886aabe51970aad7b1":{"kind":"REWORK","_number":1,"created":"2021-09-17 15:24:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/809769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/809769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/809769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/809769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/809769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/809769/1"}}},"commit":{"parents":[{"commit":"f1212d0f07cd127c4a618126178a9edfb44160e6","subject":"Merge \"Support setting docker registry credentials to the empty string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1212d0f07cd127c4a618126178a9edfb44160e6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-17 15:24:28.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-17 15:24:28.000000000","tz":120},"subject":"CI: Pin molecule to known working version","message":"CI: Pin molecule to known working version\n\nChange-Id: Ifd75f704177751ec740475bcce8a9f7b8a1578f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e5eda52fe4ea998b00760886aabe51970aad7b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e5eda52fe4ea998b00760886aabe51970aad7b1"}]},"branch":"refs/heads/master"},"81d6fa9a07c4f20d6bc244799d149101d0f62dd3":{"kind":"REWORK","_number":2,"created":"2021-09-17 16:04:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/809769/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/809769/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/809769/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/809769/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/809769/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/809769/2"}}},"commit":{"parents":[{"commit":"f1212d0f07cd127c4a618126178a9edfb44160e6","subject":"Merge \"Support setting docker registry credentials to the empty string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1212d0f07cd127c4a618126178a9edfb44160e6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-17 15:47:34.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-17 15:47:34.000000000","tz":120},"subject":"CI: Avoid molecule 3.5.x","message":"CI: Avoid molecule 3.5.x\n\nMolecule 3.5.1 was released by accident [1] and is breaking our\nkayobe-tox-molecule job. Avoid 3.5.x releases for now.\n\n[1] https://github.com/ansible-community/molecule/discussions/3240\n\nChange-Id: Ifd75f704177751ec740475bcce8a9f7b8a1578f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81d6fa9a07c4f20d6bc244799d149101d0f62dd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81d6fa9a07c4f20d6bc244799d149101d0f62dd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
