)]}'
{"id":"openstack%2Fkayobe~811097","triplet_id":"openstack%2Fkayobe~master~I7049eae6518f818f9e180dfdb6f515d527644808","project":"openstack/kayobe","branch":"master","topic":"bifrost-firewalld","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-10-19 14:43:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7049eae6518f818f9e180dfdb6f515d527644808","subject":"Prevent Bifrost from using firewalld","status":"MERGED","created":"2021-09-27 09:40:34.000000000","updated":"2021-10-19 14:45:48.000000000","submitted":"2021-10-19 14:43:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"811097-bifrost-firewalld","meta_rev_id":"a4cbdc6afae98395c3110194e06e5c9227df7041","_number":811097,"virtual_id_number":811097,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-19 14:43:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-19 12:14:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-19 12:15:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-19 09:56:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-27 11:42:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-04 10:06:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-19 12:14:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"670367b64d9f0e82158244f08c9c09a2d3a3e2fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-27 09:40:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e00a93bf2cd1f44fe14e1dbf69e5cd44dea9754a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-27 09:40:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ed33f7c42bf2dc30efba7c79033fdae7563445f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:42:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b7cb04d7dee4926bd3956b6115c0524 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7643b237cdf459eaf777e8a32c9f981 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ad9473e8e594097a956e59f4eda244f : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d72f236e9868466281d4f5fa154fcafc : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24395617c4784a368c30cb8cd8595d75 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee55ebf27a94cc1acc4d58ba8e623cf : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30528db46bbd4584b9c6b91324e0211f : SUCCESS in 5m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/39824d708239449987143baa17bc6f6a : SUCCESS in 7m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a008a45d86c4283ba6de8f35d1aaf7a : SUCCESS in 12m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eb1143dd10c742578ce5e95209d8ee89 : SUCCESS in 28m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/152bddf952e940218ed9998dca2d793e : SUCCESS in 1h 10m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b822335487d64494a062511fa9fb3813 : SUCCESS in 46m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/02e630ea0b6749bfbf502d32cdd16c5f : SUCCESS in 1h 13m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b6f2e9340dde4fad81eceaea2b4583c3 : SUCCESS in 20m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbbe24f0d3a9404fb58bcc0992ffd537 : SUCCESS in 13m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/98847f718c444f60b6b7af3cb2e729ee : SUCCESS in 2h 00m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c94fd99e9b914e2dbe4f145c22891d1f : SUCCESS in 35m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18b897a6b59c4ddd9d9f0e4e8850e13f : SUCCESS in 33m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9bf92f98420347cab4ca28de5e82f9d7 : SUCCESS in 29m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/12c86b21ad5241d6a7ea29f7a0eb8412 : SUCCESS in 16m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd0e9ba43a09486b8231ba30b96abcb7 : SUCCESS in 14m 13s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":2},{"id":"540551693b759b40f9d7183812af10ce8c1900f5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 10:06:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"41b6ff918bd9757dd6907e1c03e76134c24e23b1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-05 08:06:37.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"d56b2ebd0e928c2f54f6eccda6700794fe880acb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-19 09:55:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"957ba06b5230e2dae72b399fa064474db54a3800","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-19 09:56:05.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"8b31105866a9055fcda4b495fe912d46d662cb00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 11:54:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a6bd2995b794f039ee47c5c183a5859 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7afab758d0f74ac4aa72d9fc211dc4e0 : SUCCESS in 12m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/363b0272c91448f791789d794522ce45 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bfdeea8a311481b8575f3bad0094ada : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04225dfafdad4b0f850520d5d303627c : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b26be0e04a497db0d4af10c2b272d1 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52db7a61d9c94b5aa057d398bb2c51a1 : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84e8f94aa2124236b5641406d11b0972 : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/44a90137f19f4981a65f09d2a9493317 : SUCCESS in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fdccaeb4fd8b44d5be8b4ebfd64c99a7 : SUCCESS in 33m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e69257a8fcae4d4c8d0e30643f43b032 : SUCCESS in 1h 02m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8c62887bad64068beb32a4116b1f441 : SUCCESS in 34m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4f30f7fa396c4dae93ef6592af7d8fed : SUCCESS in 1h 13m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eb34d9f8befe4684bffd083efd407a47 : SUCCESS in 24m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/731f34dac2a64d41a37e13b95d388554 : SUCCESS in 17m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b53646fe4d5c462b8f3ad25f41ba2264 : SUCCESS in 1h 53m 19s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f788617d803c418aa39725ad28a8edaf : SUCCESS in 1h 26m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/461d7b934c1e478d93f663aae775df96 : SUCCESS in 38m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebb0227a345148e69662a300b32bc7b6 : SUCCESS in 30m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7f189188a4cc40b3a256598073618ba3 : SUCCESS in 32m 00s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/160f62c74d9b413eb44c2b14b7b17c66 : SUCCESS in 26m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/21d65bd2e52d41e5940b5fe122c08fab : SUCCESS in 15m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96d42b8179aa456d9aa4936b9c16c03f : SUCCESS in 12m 46s","accounts_in_message":[],"_revision_number":3},{"id":"c4388050bfd597ee0947438f02386d5f991fe128","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-19 12:14:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"781ffcb934f296c824ce2831c95832351b972e9b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-19 12:15:01.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d6d44ca68bf7f35341c25009a5566a628ea2696b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 12:18:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f0da33993b4901d96219d60de5a705d72e0d5d54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 14:43:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/543e6b443a3b492ba40b39808f7079da : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ecaf1b16d2f4764aaa62d309a0f7615 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df003074d6d645398caae86e2f274e6c : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e286dc2fb33497c9940325a36a6986e : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fabebaa094df4a3e85ecb6916b504001 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fafd936dcacc4789b14ba6eebe8ebaa9 : SUCCESS in 7m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9987fb48623d4e88ba1a2fdc2d259fc5 : SUCCESS in 12m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/814a1a51389743c69d5396002aea79cc : SUCCESS in 26m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0ed8dd87af4c4cbfbab752b80012d132 : SUCCESS in 1h 13m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f53bc5c756134ffcbd276b516a5e88ea : SUCCESS in 39m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ff0fb394dfb44f819e0549c98016db6e : SUCCESS in 1h 06m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/67386ff0ded24b7ab1e1b76ec3dfe563 : SUCCESS in 18m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b545a84f62e4c7b8f40259454016522 : SUCCESS in 14m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/28229caac601421e8c6445cf06d2fa16 : SUCCESS in 2h 21m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4c104ba81d674dd1872717b278478a9a : SUCCESS in 35m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67cb6e0b728d410185964f095ed7fd74 : SUCCESS in 31m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4adbc2391ba34fc29f4200129cc36dad : SUCCESS in 28m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ea0983e42e9241378da00f51329a444f : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6247447f4cb047aaa74f65c758828f3f : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":3},{"id":"25e9fe94450d9f154b8965cfdcca74df0204e513","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 14:43:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a4cbdc6afae98395c3110194e06e5c9227df7041","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 14:45:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a85b4fad314406a3eca8d9424b6afb : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fe9cd37db524aa7835e0ceea1bcee18 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9f6c912b3418b21c06f53860e786b5b56daddff6","revisions":{"192f8fef28beaa04f5a38681e771651e6c517fa2":{"kind":"REWORK","_number":1,"created":"2021-09-27 09:40:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/97/811097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/811097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/811097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/811097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/811097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/811097/1"}}},"commit":{"parents":[{"commit":"51ca557a74560492dfc6808cf5af27350c7ccc12","subject":"Merge \"Add missing quotes to Docker registry TLS example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51ca557a74560492dfc6808cf5af27350c7ccc12"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-27 09:40:29.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-27 09:40:29.000000000","tz":120},"subject":"Prevent Bifrost from using firewalld","message":"Prevent Bifrost from using firewalld\n\nThis is to avoid conflicting with iptables rules configured on the seed\nhost by Kayobe.\n\nA new variable kolla_bifrost_use_firewalld is introduced to configure\nwhether Bifrost uses firewalld.\n\nChange-Id: I7049eae6518f818f9e180dfdb6f515d527644808\nStory: 2009252\nTask: 43442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/192f8fef28beaa04f5a38681e771651e6c517fa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/192f8fef28beaa04f5a38681e771651e6c517fa2"}]},"branch":"refs/heads/master"},"694b10c48f21cd15f97cac56fe355bcc2c92bf94":{"kind":"REWORK","_number":2,"created":"2021-09-27 09:40:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/97/811097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/811097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/811097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/811097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/811097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/811097/2"}}},"commit":{"parents":[{"commit":"51ca557a74560492dfc6808cf5af27350c7ccc12","subject":"Merge \"Add missing quotes to Docker registry TLS example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51ca557a74560492dfc6808cf5af27350c7ccc12"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-27 09:40:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-27 09:40:49.000000000","tz":120},"subject":"Prevent Bifrost from using firewalld","message":"Prevent Bifrost from using firewalld\n\nThis is to avoid conflicting with iptables rules configured on the seed\nhost by Kayobe.\n\nA new variable kolla_bifrost_use_firewalld is introduced to configure\nwhether Bifrost uses firewalld.\n\nChange-Id: I7049eae6518f818f9e180dfdb6f515d527644808\nStory: 2009252\nTask: 43442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/694b10c48f21cd15f97cac56fe355bcc2c92bf94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/694b10c48f21cd15f97cac56fe355bcc2c92bf94"}]},"branch":"refs/heads/master"},"9f6c912b3418b21c06f53860e786b5b56daddff6":{"kind":"REWORK","_number":3,"created":"2021-10-19 09:55:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/97/811097/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/811097/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/811097/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/811097/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/811097/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/811097/3"}}},"commit":{"parents":[{"commit":"96a9d861cf49ffdd0561fd67caa52cd3f5b63304","subject":"Merge \"Add missing apt playbook for infrastructure VMs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96a9d861cf49ffdd0561fd67caa52cd3f5b63304"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-27 09:40:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-19 09:55:37.000000000","tz":120},"subject":"Prevent Bifrost from using firewalld","message":"Prevent Bifrost from using firewalld\n\nThis is to avoid conflicting with iptables rules configured on the seed\nhost by Kayobe.\n\nA new variable kolla_bifrost_use_firewalld is introduced to configure\nwhether Bifrost uses firewalld.\n\nChange-Id: I7049eae6518f818f9e180dfdb6f515d527644808\nStory: 2009252\nTask: 43442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f6c912b3418b21c06f53860e786b5b56daddff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f6c912b3418b21c06f53860e786b5b56daddff6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
