)]}'
{"id":"openstack%2Fkayobe~811410","triplet_id":"openstack%2Fkayobe~master~I7a4530f4f63ddb37aa30a617db5944b97bc3e17f","project":"openstack/kayobe","branch":"master","topic":"ansible-4.x","hashtags":[],"change_id":"I7a4530f4f63ddb37aa30a617db5944b97bc3e17f","subject":"Bump up Ansible supported versions","status":"MERGED","created":"2021-09-28 12:27:16.000000000","updated":"2021-09-30 10:06:00.000000000","submitted":"2021-09-30 10:04:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"811410-ansible-4.x","meta_rev_id":"a2d025395c29e84d8ebf8fa6f64b1e8b6db7ec9e","_number":811410,"virtual_id_number":811410,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 10:04:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-30 07:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-30 07:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-09-30 07:59:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-30 07:57:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-28 15:03:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-29 05:33:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-09-29 08:38:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-09-30 07:43:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-09-30 07:59:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"c3416b0a804ce64060322cab11157925961efaec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-28 12:27:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5da39bbc87e11029d60f96f7d259f02b3f350803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-28 13:01:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8179cd5199dee303770c12916684ae099bbec87c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 15:03:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e16d4eec1b464aca967a71822883fb23 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c9f8837a3c413598f28598fdde04ed : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a1b1853535844e6b22b2f8b6816a608 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c60b6770360a44d6aab62bad90d04852 : SUCCESS in 8m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/386f3caa94d6449f8b2d9a2c77e286db : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce74007cfee48b89c1b04c9e6160727 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb527568c2f2408496b838ca0aa7cfbb : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/51388e0351554dfc94b0de94e6a97e2a : SUCCESS in 8m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9ae453ec34a492885124a93195fb714 : SUCCESS in 12m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/13fd7289e119468291a22f9ac2dfcbe6 : FAILURE in 9m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1cb35c272dfc402797ff57025c5bd327 : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6a6e5b11637402793c273975c67e0c1 : SUCCESS in 57m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/437642fcfbed4d6aa99a6d8d7b2c609b : SUCCESS in 1h 07m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/64c9e6c52e964351894b42ece2aea9fe : SUCCESS in 25m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93de1074e002497d858a94a0ce855767 : SUCCESS in 17m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba23a75667774825b4c60ffeac8e9691 : SUCCESS in 1h 58m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dcab64d5a0e44c02b4c32437de30d461 : SUCCESS in 43m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b3acd3fd4eb4f7ca01d4443849699e8 : SUCCESS in 31m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0eab9ff617ad4f9ea98b10d03f92e3ac : SUCCESS in 33m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f9d6dcf03fbc405ba92215458a6615d4 : SUCCESS in 16m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3014b3771ba42539dc0aacc3c05c76d : SUCCESS in 12m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ca8869490325683c8ff4220d974af154b2dd6eea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-28 15:19:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ac0355871fe35dd93ed4f981890bfcd440bdef9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 17:25:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/136eb3f933684e1cb1dcb6925a19dfdf : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f8efe280384d6d8098dc6b53c4af39 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dcd1676745e48e19d14a2fc7794ff3a : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1057bf1704bd492dae0724dcd773b2e3 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bebe3f49a1e94959a463f2d811531858 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18adf0d4058b4f218d41da3d586e32cb : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b23e1e3c7fef4b1ba8fae259d3386684 : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/24ba6e15c6fc4f7e8c8b2ce4d3c217fb : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a60a4986ddd24e8797f0df654dbfb14c : SUCCESS in 12m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/90b085d272e9420c88b7f87b9035ee55 : FAILURE in 8m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7a13f27586f04d5d9db8d9272f87a86d : SUCCESS in 1h 07m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7439a3af6e5a454ebbd390a7741d145d : SUCCESS in 49m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/56efd4645cc840ecb58ecc8a7c8c0863 : SUCCESS in 1h 10m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c4566b26315347a4a7c6ff99b2a1ad5c : SUCCESS in 19m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a696699d9e44c0e9fd08106292d62b0 : SUCCESS in 16m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/aa5c85147c924d59869df10de4271005 : SUCCESS in 2h 03m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2e408d9bd96746049aa8c3cf7fbebb18 : SUCCESS in 49m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96c8bb4b9627405386aab750a8cd0c45 : SUCCESS in 31m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/747f0d809a0d460ca98fe991767768bc : SUCCESS in 32m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a096f72075894610bcf4f9458120dc96 : SUCCESS in 14m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d9c1a332f2c4399ba1548c9aa6f8ddc : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":3},{"id":"09e9ae824c7574cd058313df4fcfd718d4ae5512","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-28 20:22:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d4cc56496d9b021d4ea0fa13761e503fe1db2207","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-28 20:45:22.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6f7e1e8c3595b91f9da2cf0c29b4b2b9f65eaa0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 21:23:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a5c6a867ca64026b4855ff6523ebd20 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92b682082c694779b6dd4d5eed263ad6 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ef5816c00df41ecb4e0252f8310de93 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f342e912e0344ff99557566435a0f567 : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91fe6f748b4a47d9b974f622396dd16d : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b73138749941c4890423e350e261e7 : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b95d2d488a6e494381468831ef0986bf : SUCCESS in 4m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1f963815f3642db8f403d6c96c95871 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":4},{"id":"7871639774778a57a90a753d7e538e69d3156fdc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-29 05:33:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"06cc7fe5a4cfec683120d8f5931ebd1fdb19b39c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 05:54:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"28a7f5d43f5cda0d2b70caf5d2c890e12a0a3124","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 05:56:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a6f799eb28359d4ee82dedf763e5a92f721f9f08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 05:57:08.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"acda9470ad43ff31745b1fbacc19ccc35601f4ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 07:31:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f76cdb563fcd4b4eb58c66acc68b65a9 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91aaf83b303249849abf08ae651ec8f8 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77a616117f444eaa96e24db3a6eb6926 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/877d32d82ba74784b493d218b511eab4 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/709d18cee56e4a419439db4244f0d157 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b9c539ba474514ae4acaf6d5728e85 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b23d622332d42e59bb6f6fe1cea037a : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ea7ec8c21f441b1b22d03073beb0927 : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/278f02db31f2499ab9de3434c0315a08 : SUCCESS in 12m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c77f6117ea54c4bace2851ebf1625e0 : SUCCESS in 30m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a519b37c57da4bf8b17a8bfd27d88d6d : SUCCESS in 1h 10m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0b36550b3324af698eba1d629a00411 : SUCCESS in 51m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/04bc9f6da73c4016beecceefd895965a : SUCCESS in 1h 19m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/265c5d1fd6e048d2ba069e8bb988d62d : SUCCESS in 20m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4adf204abcf445c292c03706aa35d724 : SUCCESS in 16m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d8b07567aa50499faad87308e912d653 : FAILURE in 1h 28m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5e9a23b5a10b443892d059382657a565 : SUCCESS in 40m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8051c675b0ea4003ac041372a06b1d40 : SUCCESS in 29m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e15399816c92484db44b9bd3a10fdc9c : SUCCESS in 35m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/be969010cda0446ca5e694d2442022c1 : SUCCESS in 14m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/213fc003204e48f19f042a224e54fcdd : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":6},{"id":"979b1550ec172e76053f5d8236a0169c68e73f9f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-29 08:38:04.000000000","message":"Patch Set 6:\n\nUpgrading directly from ansible-2.10, ansible-3, or ansible-base-2.10 to ansible-4 or greater with pip is known to cause problems.  Please uninstall the old version found at...","accounts_in_message":[],"_revision_number":6},{"id":"d7f20878e93cd69dc65d5886855c95aac79a85ba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 11:06:43.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6:\n\u003e \n\u003e Upgrading directly from ansible-2.10, ansible-3, or ansible-base-2.10 to ansible-4 or greater with pip is known to cause problems.  Please uninstall the old version found at...\n\nI\u0027ll propose something.","accounts_in_message":[],"_revision_number":6},{"id":"478446d4437fd8dd7ee357480cf287b4d608838b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 15:18:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0e0530008afc6f8f52d291138b4242c4c493a498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 16:46:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e96b7d9aecd4e9ebe89f27167874c72 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/763d153d8a744aa8bb8dcba612e5bdfe : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40d7cf5629b04334ba342a1615818872 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ca85f6a59564a3998172ad568ea6744 : SUCCESS in 7m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a7bb7e3ac1a42fc9ad5f040e4e58a4d : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d131cb4fe6f45a8b48d49dfe4daaf0b : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7796a67d24f84c6caa6f96001baf43ce : SUCCESS in 5m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c28262d616574f609785501b04a93657 : SUCCESS in 8m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8327f3ec8874439cb619017d5f9ab494 : SUCCESS in 12m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0566dab5b98242a48cc1b92d1049442f : SUCCESS in 32m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7923845c2646407e85f3ad6a6d4bbe10 : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1952a5c903b64a729f037a62bf634df1 : SUCCESS in 49m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4506cb3305ec4c26b5d24ad400b0d19a : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/451f8f54e1b741ea87f6a31e7b7540fc : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92c9b78cff6342c390e3df9031c4c75e : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/69842a8aa7c5423fbbaaf39f4cc740be : FAILURE in 1h 19m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0089c1c291ba49dea81240d4e57252fc : SUCCESS in 37m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/638bf0cf327146fabfa262bdfc2a9aae : SUCCESS in 32m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/640f6476ad444fe491257937ada65329 : SUCCESS in 31m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a820b8d90453424e926511727076e646 : SUCCESS in 18m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60e48fd03076422597af0f3d8035f629 : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":7},{"id":"ab14ea74b3c89254cad458369cceb4f26fee821b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 16:56:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"bd22cd00c5935716f5853b6114b92eb5c058cd47","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 17:07:40.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2f9b11f8c63d3da8f4bae8a1ee178496cb292a9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 19:22:31.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/740d25abdd1c4dccb00a5cc7938ab669 : SUCCESS in 1h 02m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4b332aebbb7c48cf9bf24f53df1271c6 : SUCCESS in 2h 12m 26s","accounts_in_message":[],"_revision_number":9},{"id":"35d74f590a8d239117084538a7fbd77853f728c5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 20:25:06.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ff25b623d1c4dcfd406180c76e454bd81d4276d8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 20:31:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d15501d0dedb20b2fe6dbd02c9a8adfde239766a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 22:21:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28b2c0240222433f85c7b8c0155de950 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72ddfeb8b11d476883c9b1aa82ae63da : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb8c21bb9bd94d379bc700a56365c89a : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4566de78d8d448748a748cb2513e8c53 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea888715a7e64e649274509cbc4a98ee : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa26618195474c15ab2ca4e973cd72e2 : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a3961f658f643f29225b7bda2bde909 : SUCCESS in 4m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9e02d765e9284b4cbdb93f145ef593d8 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/95ecf637017c4bc9b497c47b91421574 : SUCCESS in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8820d31e64664111916b32f619b95c63 : SUCCESS in 29m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e6e90ad8e3904aba8fd2e1bb3fc5fe63 : SUCCESS in 1h 09m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58146d4ec3284b69810841383ad66fcb : SUCCESS in 50m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e1b8ba05f6324e2494f8cc43923f8da9 : SUCCESS in 1h 11m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b6efc5449d4146a18dfef601324f73b1 : SUCCESS in 20m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b934a3742cb645aeae97b3e32bd7de2b : SUCCESS in 14m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c48036c6a72c4f6cb5713e01108502c6 : SUCCESS in 1h 49m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ce4487220c4d494ea648e8e3add3f185 : SUCCESS in 45m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/502bf64f2e8241e6aa24eff37d47f7a1 : SUCCESS in 31m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b5ca9d6865e34c2c88a3e69e5a06e0b7 : SUCCESS in 27m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3f288edbf21345338837c19678268b99 : SUCCESS in 17m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7690cba156574a7791e8808d50d03834 : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":11},{"id":"e2708f077f7b0868564fe5b8ff1ddeb2d067d5c2","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-30 06:52:01.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"403bc8549591927e2af148800b138939f8932e40","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-30 07:43:54.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"72adfccbf91394ebe4727de295a3219f59633c01","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-30 07:57:37.000000000","message":"Patch Set 11: Review-Priority+1","accounts_in_message":[],"_revision_number":11},{"id":"cd873887406d25e4a74a7e927cd4fd3b0cea6f3d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 07:59:08.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"dfae74ec2d72fa777b1b118188dd21f9d2ccf733","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 08:00:06.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"5053094117ad6fe86d3c78ee0233833b94fc37f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:04:26.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ac3afbc52f34572bcd6b30c21f29b7c : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20ca7f9bd9404fd59f6e84db168f9b57 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6629157ae94c43708292f0526ad58d14 : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e00cf125fe34e4fab1d40c56e09d6fb : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46b4325b9fe749deb434e2c61d9979dc : SUCCESS in 5m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/903bb394967849b59bdb003627a800d3 : SUCCESS in 7m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/16d9c13dc7954ce89a4c88879e43445a : SUCCESS in 13m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91a15da0f8164715b655a70c258973ee : SUCCESS in 35m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1bfdda85acf34338b7df3151c1307e16 : SUCCESS in 1h 09m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ec8e062129446af826135554bb0f812 : SUCCESS in 49m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d1005b05286e4d82b55ebdf840f03fd6 : SUCCESS in 1h 10m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b706562baa9841179571d2241558ebf9 : SUCCESS in 21m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73b07ecc5b7c42e0be56bd96c9d4e13d : SUCCESS in 17m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0e8442d9459b419f8769ebd4bba78be3 : SUCCESS in 1h 53m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d165dc7aefd34c07a75cae9668830679 : SUCCESS in 38m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74894796e7064681b5a654649192aa03 : SUCCESS in 36m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/20108bddf2884f91a46f5af1e8624757 : SUCCESS in 29m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/48852db993d54a918b2483501ecb75c5 : SUCCESS in 15m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91007be23d0a48489237f02a9e0f9cf0 : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":11},{"id":"b90fbbe6a431fa183d4386ac4e6d9b3fabbba27b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:04:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"a2d025395c29e84d8ebf8fa6f64b1e8b6db7ec9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:06:00.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cece4296733544799b2950c6e568ed8b : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21a2ce90015048d8be2ccad6b3623d31 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"0f2b1042133b2450d3ef1ea8e16c328f1f8583e2","revisions":{"3c4f699494cb5e6d70f1a9e27af327936ef1ee50":{"kind":"REWORK","_number":1,"created":"2021-09-28 12:27:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/10/811410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/811410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/811410/1"}}},"commit":{"parents":[{"commit":"f9e31f0b540474e7257ab10fa9642388f46305ee","subject":"Merge \"CI: Fix bashate ignores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9e31f0b540474e7257ab10fa9642388f46305ee"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-28 12:24:18.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-28 12:24:18.000000000","tz":120},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 4.x\n(ansible-core 2.11.x) and minimum to 2.10. This synchronises Kayobe with\nKolla Ansible (see change I8b9212934dfab3831986e8db55671baee32f4bbd).\n\nChange-Id: I7a4530f4f63ddb37aa30a617db5944b97bc3e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c4f699494cb5e6d70f1a9e27af327936ef1ee50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c4f699494cb5e6d70f1a9e27af327936ef1ee50"}]},"branch":"refs/heads/master"},"a5681591031e7672f0fcf901131e1ec248ad56c6":{"kind":"REWORK","_number":2,"created":"2021-09-28 13:01:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/10/811410/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/811410/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/811410/2"}}},"commit":{"parents":[{"commit":"f9e31f0b540474e7257ab10fa9642388f46305ee","subject":"Merge \"CI: Fix bashate ignores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9e31f0b540474e7257ab10fa9642388f46305ee"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-28 12:59:04.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-28 12:59:04.000000000","tz":120},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 4.x\n(ansible-core 2.11.x) and minimum to 2.10. This synchronises Kayobe with\nKolla Ansible (see change I8b9212934dfab3831986e8db55671baee32f4bbd).\n\nUses of the docker_image are updated to provide the source parameter,\nwhich is now required.\n\nChange-Id: I7a4530f4f63ddb37aa30a617db5944b97bc3e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5681591031e7672f0fcf901131e1ec248ad56c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5681591031e7672f0fcf901131e1ec248ad56c6"}]},"branch":"refs/heads/master"},"985840ec38ccde6d60f2e5ade7a53ba3c8dc3511":{"kind":"REWORK","_number":3,"created":"2021-09-28 15:19:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/10/811410/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/811410/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/811410/3"}}},"commit":{"parents":[{"commit":"f9e31f0b540474e7257ab10fa9642388f46305ee","subject":"Merge \"CI: Fix bashate ignores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9e31f0b540474e7257ab10fa9642388f46305ee"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-28 13:42:01.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-28 13:42:01.000000000","tz":120},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 4.x\n(ansible-core 2.11.x) and minimum to 2.10. This synchronises Kayobe with\nKolla Ansible (see change I8b9212934dfab3831986e8db55671baee32f4bbd).\n\nUses of the docker_image are updated to provide the source parameter,\nwhich is now required.\n\nUse docker_image_info instead of docker_image_facts.\n\nChange-Id: I7a4530f4f63ddb37aa30a617db5944b97bc3e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/985840ec38ccde6d60f2e5ade7a53ba3c8dc3511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/985840ec38ccde6d60f2e5ade7a53ba3c8dc3511"}]},"branch":"refs/heads/master"},"46a870926efee58405e087de369ae2d425b0204c":{"kind":"REWORK","_number":4,"created":"2021-09-28 20:22:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/10/811410/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/811410/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/811410/4"}}},"commit":{"parents":[{"commit":"f9e31f0b540474e7257ab10fa9642388f46305ee","subject":"Merge \"CI: Fix bashate ignores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9e31f0b540474e7257ab10fa9642388f46305ee"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-28 20:22:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-28 20:22:08.000000000","tz":120},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 4.x\n(ansible-core 2.11.x) and minimum to 2.10. This synchronises Kayobe with\nKolla Ansible (see change I8b9212934dfab3831986e8db55671baee32f4bbd).\n\nUses of the docker_image are updated to provide the source parameter,\nwhich is now required.\n\nUse docker_image_info instead of docker_image_facts.\n\nChange-Id: I7a4530f4f63ddb37aa30a617db5944b97bc3e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46a870926efee58405e087de369ae2d425b0204c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46a870926efee58405e087de369ae2d425b0204c"}]},"branch":"refs/heads/master"},"69c9b5d4d78e62c5b18020f902fc4c62fba9bdc5":{"kind":"REWORK","_number":5,"created":"2021-09-29 05:54:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/10/811410/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/811410/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/811410/5"}}},"commit":{"parents":[{"commit":"f9e31f0b540474e7257ab10fa9642388f46305ee","subject":"Merge \"CI: Fix bashate ignores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9e31f0b540474e7257ab10fa9642388f46305ee"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 05:52:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 05:52:12.000000000","tz":120},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 4.x\n(ansible-core 2.11.x) and minimum to 2.10. This synchronises Kayobe with\nKolla Ansible (see change I8b9212934dfab3831986e8db55671baee32f4bbd).\n\nUses of the docker_image are updated. We must now provide the source\nparameter. When source is build (for molecule), we replace force by\nforce_source and force_tag and move path and dockerfile under the build\nparameter.\n\nUse docker_image_info instead of docker_image_facts.\n\nChange-Id: I7a4530f4f63ddb37aa30a617db5944b97bc3e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69c9b5d4d78e62c5b18020f902fc4c62fba9bdc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69c9b5d4d78e62c5b18020f902fc4c62fba9bdc5"}]},"branch":"refs/heads/master"},"65820b6701ccb2a7a783599da6addcf68731923d":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-09-29 05:57:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/10/811410/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/811410/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/811410/6"}}},"commit":{"parents":[{"commit":"f9e31f0b540474e7257ab10fa9642388f46305ee","subject":"Merge \"CI: Fix bashate ignores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9e31f0b540474e7257ab10fa9642388f46305ee"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 05:56:51.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 05:56:51.000000000","tz":120},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 4.x\n(ansible-core 2.11.x) and minimum to 2.10. This synchronises Kayobe with\nKolla Ansible (see change I8b9212934dfab3831986e8db55671baee32f4bbd).\n\nUses of docker_image are updated. We must now provide the source\nparameter. When source is build (for molecule), we replace force by\nforce_source and force_tag and move path and dockerfile under the build\nparameter.\n\nUse docker_image_info instead of docker_image_facts.\n\nChange-Id: I7a4530f4f63ddb37aa30a617db5944b97bc3e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65820b6701ccb2a7a783599da6addcf68731923d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65820b6701ccb2a7a783599da6addcf68731923d"}]},"branch":"refs/heads/master"},"c948d9e274dcc38f4f2c8f5ae8b082cc660e4256":{"kind":"REWORK","_number":7,"created":"2021-09-29 15:18:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/10/811410/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/811410/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/811410/7"}}},"commit":{"parents":[{"commit":"f09faa43d1a1d38b30967cce948af6b60c34be70","subject":"CI: Disable libvirt debug logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09faa43d1a1d38b30967cce948af6b60c34be70"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 15:18:24.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 15:18:24.000000000","tz":120},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 4.x\n(ansible-core 2.11.x) and minimum to 2.10. This synchronises Kayobe with\nKolla Ansible (see change I8b9212934dfab3831986e8db55671baee32f4bbd).\n\nUses of docker_image are updated. We must now provide the source\nparameter. When source is build (for molecule), we replace force by\nforce_source and force_tag and move path and dockerfile under the build\nparameter.\n\nUse docker_image_info instead of docker_image_facts.\n\nHandle update of ansible inside kolla-ansible virtualenv and document\nhow to do it for the kayobe virtualenv.\n\nChange-Id: I7a4530f4f63ddb37aa30a617db5944b97bc3e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c948d9e274dcc38f4f2c8f5ae8b082cc660e4256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c948d9e274dcc38f4f2c8f5ae8b082cc660e4256"}]},"branch":"refs/heads/master"},"44973f970564bdbc0a838d852fa34453d4606e1e":{"kind":"REWORK","_number":8,"created":"2021-09-29 16:56:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/10/811410/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/811410/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/811410/8"}}},"commit":{"parents":[{"commit":"f09faa43d1a1d38b30967cce948af6b60c34be70","subject":"CI: Disable libvirt debug logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09faa43d1a1d38b30967cce948af6b60c34be70"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 16:56:18.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 16:56:18.000000000","tz":120},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 4.x\n(ansible-core 2.11.x) and minimum to 2.10. This synchronises Kayobe with\nKolla Ansible (see change I8b9212934dfab3831986e8db55671baee32f4bbd).\n\nUses of docker_image are updated. We must now provide the source\nparameter. When source is build (for molecule), we replace force by\nforce_source and force_tag and move path and dockerfile under the build\nparameter.\n\nUse docker_image_info instead of docker_image_facts.\n\nHandle update of ansible inside kolla-ansible virtualenv and document\nhow to do it for the kayobe virtualenv.\n\nChange-Id: I7a4530f4f63ddb37aa30a617db5944b97bc3e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44973f970564bdbc0a838d852fa34453d4606e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44973f970564bdbc0a838d852fa34453d4606e1e"}]},"branch":"refs/heads/master"},"76c88cfa7cb2ddbcae9da704aaa84dd26403a0e6":{"kind":"REWORK","_number":9,"created":"2021-09-29 17:07:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/10/811410/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/811410/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/811410/9"}}},"commit":{"parents":[{"commit":"f09faa43d1a1d38b30967cce948af6b60c34be70","subject":"CI: Disable libvirt debug logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09faa43d1a1d38b30967cce948af6b60c34be70"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 17:07:31.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 17:07:31.000000000","tz":120},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 4.x\n(ansible-core 2.11.x) and minimum to 2.10. This synchronises Kayobe with\nKolla Ansible (see change I8b9212934dfab3831986e8db55671baee32f4bbd).\n\nUses of docker_image are updated. We must now provide the source\nparameter. When source is build (for molecule), we replace force by\nforce_source and force_tag and move path and dockerfile under the build\nparameter.\n\nUse docker_image_info instead of docker_image_facts.\n\nHandle update of ansible inside kolla-ansible virtualenv and document\nhow to do it for the kayobe virtualenv.\n\nChange-Id: I7a4530f4f63ddb37aa30a617db5944b97bc3e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76c88cfa7cb2ddbcae9da704aaa84dd26403a0e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76c88cfa7cb2ddbcae9da704aaa84dd26403a0e6"}]},"branch":"refs/heads/master"},"b8e0c817c3116f39ebfa592bc10c025a7f07cde8":{"kind":"REWORK","_number":10,"created":"2021-09-29 20:25:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/10/811410/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/811410/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/811410/10"}}},"commit":{"parents":[{"commit":"4e14f9e9c58eaccf01b8b38b6d318a9da37fcdc9","subject":"Merge \"Support Ansible collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e14f9e9c58eaccf01b8b38b6d318a9da37fcdc9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 20:22:58.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 20:22:58.000000000","tz":120},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 4.x\n(ansible-core 2.11.x) and minimum to 2.10. This synchronises Kayobe with\nKolla Ansible (see change I8b9212934dfab3831986e8db55671baee32f4bbd).\n\nUses of docker_image are updated. We must now provide the source\nparameter. When source is build (for molecule), we replace force by\nforce_source and force_tag and move path and dockerfile under the build\nparameter.\n\nUse docker_image_info instead of docker_image_facts.\n\nHandle update of ansible inside kolla-ansible virtualenv and document\nhow to do it for the kayobe virtualenv.\n\nChange-Id: I7a4530f4f63ddb37aa30a617db5944b97bc3e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8e0c817c3116f39ebfa592bc10c025a7f07cde8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8e0c817c3116f39ebfa592bc10c025a7f07cde8"}]},"branch":"refs/heads/master"},"0f2b1042133b2450d3ef1ea8e16c328f1f8583e2":{"kind":"REWORK","_number":11,"created":"2021-09-29 20:31:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/10/811410/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/811410/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/811410/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/811410/11"}}},"commit":{"parents":[{"commit":"4e14f9e9c58eaccf01b8b38b6d318a9da37fcdc9","subject":"Merge \"Support Ansible collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e14f9e9c58eaccf01b8b38b6d318a9da37fcdc9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 20:31:33.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 20:31:33.000000000","tz":120},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 4.x\n(ansible-core 2.11.x) and minimum to 2.10. This synchronises Kayobe with\nKolla Ansible (see change I8b9212934dfab3831986e8db55671baee32f4bbd).\n\nUses of docker_image are updated. We must now provide the source\nparameter. When source is build (for molecule), we replace force by\nforce_source and force_tag and move path and dockerfile under the build\nparameter.\n\nUse docker_image_info instead of docker_image_facts.\n\nHandle update of ansible inside kolla-ansible virtualenv and document\nhow to do it for the kayobe virtualenv.\n\nChange-Id: I7a4530f4f63ddb37aa30a617db5944b97bc3e17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f2b1042133b2450d3ef1ea8e16c328f1f8583e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f2b1042133b2450d3ef1ea8e16c328f1f8583e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
