)]}'
{"id":"openstack%2Fkayobe~811501","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ic047e80aa382148b009b1c03581b417d17ec2598","project":"openstack/kayobe","branch":"stable/victoria","topic":"libvirt-logging-stable/victoria","hashtags":[],"change_id":"Ic047e80aa382148b009b1c03581b417d17ec2598","subject":"CI: Disable libvirt debug logs","status":"MERGED","created":"2021-09-29 11:55:27.000000000","updated":"2021-09-30 14:44:04.000000000","submitted":"2021-09-30 14:42:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"811501-libvirt-logging-stable/victoria","cherry_pick_of_change":811562,"cherry_pick_of_patch_set":1,"meta_rev_id":"fc9f4c630fdaf24f3bbda78dd1ca8f754dbb7d56","_number":811501,"virtual_id_number":811501,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 14:42:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-30 07:31:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-29 12:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-30 07:31:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-29 11:55:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-09-29 17:30:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a97ac482a78cca5641cde4945562bdd85f61859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-29 11:55:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f2fce5fb7a729f90ed74fb85cea992fdecfd2b4b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 12:04:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b86972de75e59f1829106a4effe4c296f4f83b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 17:30:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3da53b52701f487eb1ed3255c55d2f02 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/919aafff615e4fcc99de57cc3935db3f : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fdd7142e8bf4c09aa42a4d846ae2760 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/056041852e2144aeb0185d859f39b04c : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9428e1d9ab644e7bb1b29c8f2a002adb : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/259cd967201b4e099029a3b5cbc6e0d5 : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/67a52e5a6e884ca39a624a244648e8a2 : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b6bb847a37c94e108bed07143b6394e4 : SUCCESS in 17m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a7d7cb1f8f244c0f99f4db85a8cf63f4 : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5c76254ec3374bcd9362b9441156a8a1 : SUCCESS in 58m 02s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c00ed9cfdedf48b1a55eb3526490887b : SUCCESS in 59m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0bd939bb702b40a394fa68d013ebe208 : SUCCESS in 1h 05m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/64c8783ee2b34eba9ce02448ae695aea : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fa105975975e452d85d35fc6cfed27a8 : SUCCESS in 18m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7cfeebd4f03846de9a02eba6fe449a08 : SUCCESS in 2h 08m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e40b4ac333f44730b8512f2406d4c211 : POST_FAILURE in 2h 54m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6d3eb02174a94ee983921f60c9eef480 : SUCCESS in 31m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/48049a4b79ae4b35bf15208627e06178 : SUCCESS in 28m 23s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5a2777ecf7a148b485e9add2208b48c9 : SUCCESS in 25m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c098d80ece4c4acfad4b449ba7b3451b : SUCCESS in 25m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1b277d36b6f54eb4bfe06e0d58fa7148 : SUCCESS in 13m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f0420fd380604de69f082611c3853556 : SUCCESS in 13m 38s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"311e9c228d64b8b88a0e36a92d9e1bc3bbe0241f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-29 19:53:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9c00f2f1b742a149914c7674ec8d5e8780ad58d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 00:03:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd736d719ac447c8a8b748fc180c4de1 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9745d41dbc44e4492b9aad4bea27440 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7610d3897200426cbac109589d2b6d1f : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5346ad80c8a4763bfab50d3085c169f : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81fccc3f3051430c99498209716e31a1 : SUCCESS in 8m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/201ea6f84dd443318bf7ad5160f1bb89 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a317b1dcfcee46eb9161fda80621ef30 : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5775dc953fa747ef9fca2a8e1ae6884d : SUCCESS in 19m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fe0d1d7a223743429a134ac7a82515f2 : SUCCESS in 1h 06m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7910ed034fec4608a7c671733ca33747 : SUCCESS in 1h 02m 21s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d62b123dc1ec43ee92f0571cc7d53ce3 : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/01f294ae69ea4b88872235ae318a2601 : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/56163893cb0b4bb4ae57056c448aa258 : SUCCESS in 17m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b7f6a4647ab142fabef94eec005e830a : SUCCESS in 18m 17s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0f6be0ec04d743ed847b9c3ba985ef93 : SUCCESS in 2h 01m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/37e7ae2324574e65893626461f1b487c : POST_FAILURE in 2h 39m 28s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a2a3348c43ea47bfbbf8273c9865660f : SUCCESS in 27m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ff58b428762e4fd784a6f33009e56bbe : SUCCESS in 31m 52s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ed9a020df03645d0878263e44b202bd3 : SUCCESS in 28m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e37d14fcc0474e8c83fea5a5b6b25ccc : SUCCESS in 26m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4ee361babf274eedb4a3db1bd9978402 : SUCCESS in 11m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8ef5e98fea964bac91e93bcebfb01038 : SUCCESS in 15m 09s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"cf3d9afa19e4de8d6697b19c967afc20a1d6bf31","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 07:31:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"025df4e5482b1e896a95f36e6ac5df0e8255cb3e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 07:31:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f47603ba2c7f3c60bc5a879f99f26dd5f847603f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 08:01:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bba62014a434341af5155769f10e486 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21c2ed379d574e2e8ba035ccce9fba97 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/595d979b74d540a989f4bf8691b1921a : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b7cad52562a49c49bcb39cbff8e546d : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14d37f282b9f4d16aea40f55ae4051cf : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/705cc693bc324fdfbd9c1c54f920b121 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ef254040f214a76a1b37ebddb4b21d6 : SUCCESS in 8m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d06d4921d8fa4d878a9b01127f803665 : SUCCESS in 18m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5aed71c559db4cc885a655c5296f01aa : FAILURE in 14m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6dddfa566bf742edacd206ba62ddaa42 : FAILURE in 14m 47s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ac7686d182c24e718cf1e0f3fbf58900 : FAILURE in 14m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/724a9cf993aa401481ab74af4ede7e1a : FAILURE in 16m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a19c1e88c7ab42e399f75eeffbed4a0f : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e66787de1e3e4652b4fb5a379610648e : SUCCESS in 16m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e1f4378ddaeb4ac88911e909b5012252 : FAILURE in 15m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d910b422aaf0423689c2a0f54d83faac : FAILURE in 14m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f89bc3980a75427a8630730064e57e0e : FAILURE in 12m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fe92b2b4a7624e7e99fec1884f71f719 : FAILURE in 12m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8ad0559e9e1549dd894a2eecef30906b : FAILURE in 12m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/524bf32b143f49629828e1601eb3024b : SUCCESS in 25m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/143a205206ce4e33b0022d501ad5d25f : SUCCESS in 12m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0b9c707669564a3782fc336ed0a7b150 : SUCCESS in 12m 27s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"66c669d2b17f5111eabe8765cc4d90f6ef15ec2e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-30 08:02:32.000000000","message":"Patch Set 1:\n\nrecheck\n\n502 Proxy Error","accounts_in_message":[],"_revision_number":1},{"id":"647b36d0569577c075ab303dcb749f1873e73a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 12:09:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c28a286802a43d690b64697419859be : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdda35a04412454f9ba61a56f41c051f : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f675b629be4e4a82beb398644a56f61f : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c79d8a526ae410fa633fa9f59c415bd : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d85934be941c41eb9e8da77786bdea02 : SUCCESS in 7m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dbcac7a4c97c4146a412c5c1215c84f8 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b5f1ac6d5840419eb7aac9f7dda31a4f : SUCCESS in 9m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75da0c3b088b441381529b2e867c811d : SUCCESS in 17m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9d245476ebca4e91a5d463b23dbb2396 : SUCCESS in 58m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7f7efa0a173046ee816fc93e9ea1b761 : SUCCESS in 57m 29s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/77be975268c14eeb83d27ce2970177c8 : SUCCESS in 59m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dcaef0c470fa49e7a3a7c4071e5de035 : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/abd6d024f7e4436d8f82170d34493299 : SUCCESS in 16m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e1ba95ffb50c45afa45d9b4974d8ff39 : SUCCESS in 15m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/176c056496054629a13f7b26e51a3a7c : SUCCESS in 2h 16m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/df329d0da2114db284eee20671ce20cb : SUCCESS in 2h 49m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4ef91f6ff022444cb5f7139ae5241ec9 : SUCCESS in 30m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9e20de63a4d846a8bc77e3b8ad7f8e55 : SUCCESS in 30m 11s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/99a519ebea3e44f7a08626d39ed08ae6 : SUCCESS in 25m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f165f60a470a44b8aeb1c7d4afb08467 : SUCCESS in 25m 45s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8b3c32391ae940ec9d59761dcee9f738 : SUCCESS in 11m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/07ab30ed08874c8c933b9cd166b45054 : SUCCESS in 13m 09s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"90847fece4457708818049d53dadad6f61d9077e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 12:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"559bfae3dfde9d6b9570eced2ad43f7bcf28a8c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 14:42:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba675a17ecab4454aa623085294e6cca : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4772b4985f8a4272926520ce4156551f : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aac990b6068342b489dc8ad6f243a572 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994aa8e0e645481b97793b932879c965 : SUCCESS in 7m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f25fe5fca1a34fc49f974fe5fce93ff8 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a24a5b6a93e14178808313514b150d1e : SUCCESS in 9m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8cdf6ce60503494ab1738f805b5601a1 : SUCCESS in 19m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/60124cac8ff34b4580e2184dd25b8886 : SUCCESS in 1h 03m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c1681e14a7a84c8db51925da6b3acb1e : SUCCESS in 1h 03m 10s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f6e1fe09e67c4d44a2a05f491e4dec59 : SUCCESS in 1h 06m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7c7a49e3f9014d81b38b0ac59eff1106 : SUCCESS in 1h 05m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1214284f4c0c433795bc896b0769d253 : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6e4943f49b0a45f1a9476a213f469472 : SUCCESS in 17m 30s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/790b000c850a45b989504fe4473f9d1c : SUCCESS in 2h 26m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d8620c4fdc6f4fce9a6aad6819b0625c : SUCCESS in 2h 27m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9b80afbb44d84cbfa54a82ab5c4568e9 : SUCCESS in 28m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/58f7c3b9399f467a8fb56b446af36b94 : SUCCESS in 30m 44s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4353a4e8e6ef45d094d4261c896cfc7a : SUCCESS in 29m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b1a1ff0ddcd64371aea47283d0ae2128 : SUCCESS in 30m 22s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/417c6318573946108acf40ba1274614c : SUCCESS in 13m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/08e9c3ea45f54b1abe37088c43f12631 : SUCCESS in 12m 28s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"b235f06a97ad8595c2601b4b16e8a30e9833e39f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 14:42:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fc9f4c630fdaf24f3bbda78dd1ca8f754dbb7d56","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 14:44:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c94d27bc5ba486ca1348d716c21141f : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b588b8054d3ceb2e3af1924fa342124ef6c213d2","revisions":{"b588b8054d3ceb2e3af1924fa342124ef6c213d2":{"kind":"REWORK","_number":1,"created":"2021-09-29 11:55:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/811501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/811501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/811501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/811501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/811501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/811501/1"}}},"commit":{"parents":[{"commit":"0c7e4fe0a26de0879a0cad2bafd92b64f5de0f41","subject":"Merge \"Add missing quotes to Docker registry TLS example\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c7e4fe0a26de0879a0cad2bafd92b64f5de0f41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-29 07:54:10.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-29 11:55:27.000000000","tz":0},"subject":"CI: Disable libvirt debug logs","message":"CI: Disable libvirt debug logs\n\nLibvirt debug logs can grow very large, particularly on Ubuntu (we have\nseen up to 1.1GB). Disable libvirt debug logging using variable\nnova_libvirt_logging_debug which was added to Kolla Ansible during the\nWallaby cycle (see change I90bfd1b300ad3202ea4d139fda6d6beb44c5820f) and\nbackported to Victoria and Ussuri.\n\nChange-Id: Ic047e80aa382148b009b1c03581b417d17ec2598\n(cherry picked from commit f09faa43d1a1d38b30967cce948af6b60c34be70)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b588b8054d3ceb2e3af1924fa342124ef6c213d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b588b8054d3ceb2e3af1924fa342124ef6c213d2"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
