)]}'
{"id":"openstack%2Fkayobe~811784","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I5ce3c734433e1682ee942867505468c57440e689","project":"openstack/kayobe","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-10-21 15:47:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5ce3c734433e1682ee942867505468c57440e689","subject":"Make setup module arguments configurable","status":"MERGED","created":"2021-09-30 07:46:41.000000000","updated":"2021-10-21 15:50:49.000000000","submitted":"2021-10-21 15:47:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"811784","cherry_pick_of_change":811783,"cherry_pick_of_patch_set":2,"meta_rev_id":"688f0cb8ee9d8f2fc9d7ce5a02ebbec41f2d7285","_number":811784,"virtual_id_number":811784,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-21 15:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-01 14:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-20 15:13:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-20 15:13:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-30 10:42:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 15:13:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"7a14d041868c1e5d6902e2d7e3c06b22a6afcd11","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 07:46:41.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.\n\nThe following files contain Git conflicts:\n* doc/source/configuration/reference/ansible.rst","accounts_in_message":[],"_revision_number":1},{"id":"05aef4c7a2a1a5d9cd72201ec37a350156639f3b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 08:50:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3bb85c503e5f1ea870f2e5c966afacef35b09464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:42:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/09acce76c42849e1ae323bbd2a025280 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a24e8d3f67f44f6c87a0623a792783fc : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d5953a921084d2b95e750e905963151 : FAILURE in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2d3d7b9c1bf43769dd3a2081f0ac9b0 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8028f2f6c3e84bd29e685fd2586585a7 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af13abc86154290a3f2efe450e7e996 : FAILURE in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdce04382ca94d9f8e3bbf0577aea607 : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5288fa7cecd747d4b322bfe03a05302f : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5da4f7ab513e460ba1ef8e197b28d53c : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cebc211182ee4627add3701d49dfa28d : SUCCESS in 20m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/edbca59799cf4fee8e6a44641a4c59e3 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/004b7247e9964ff78c2737db97a826af : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dadfb653f10a418480c49dfaf66cb735 : SUCCESS in 1h 50m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2c4ac5f11e8b4f47ba5f518cd2ab8fb0 : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b346a60089764b5cbef2ed6e634f465c : SUCCESS in 28m 10s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6c3eac0fc6094cdebe675fdbe69eb5d4 : SUCCESS in 10m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"77283533eaaeb9bf599d2b827d1c5aa79cd81fd6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-30 14:08:55.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"61595ec4dd4622bab9c5b5c87988c24f5b0346db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 16:21:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fd91668151594fa585909c4f2cb128ff : SUCCESS in 3m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cbc12578766430e9efcba8b1bd4a6d9 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/596c399e44cb4447b3634a0b60301a9f : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8e18300efcb4d4eb39a85fc10ea6f84 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e99f0bc6821b4f19bbf4b576fbb61cb3 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1580de0fd9448ff9aad401ab8106de6 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9708f87991b945f38237bcb8ae4398be : SUCCESS in 4m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4f42538a94ea4447b095426844cbd75f : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7711655c414c4ffbbe4ee6a6604a3bc3 : SUCCESS in 10m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7de35d7968cd41f48b57c81e2c99f6a0 : SUCCESS in 18m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c8b2be2ab6ec4fc79f3920b69b729a66 : SUCCESS in 1h 11m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a29aec4f8a644ca8ad605834817617c4 : SUCCESS in 17m 09s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0db9c8d5d6b646339dfea2640b258be3 : SUCCESS in 2h 09m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8f1082200b6544afb78d1092f2dc6cef : SUCCESS in 30m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fd63304ea7f1442692c6b6e7c841a377 : SUCCESS in 31m 21s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1bccde0fd8954653834b7da778f24503 : SUCCESS in 11m 48s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"e09162d25ffed67e30b33450642b8856196da40d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-01 08:07:32.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"00b6821a3386437cc667da0ce4ee580e07f6031f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-01 08:08:30.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ae3332d3f8c87b714c55f530137ab816ea66cbfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 10:09:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e5658fc2cd6e4feea0b7cb35560b2f43 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad38f98df74540b4a770fe150be9c54e : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a1e383dee1a4fe88725f09eccdf8ae0 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14169b6935414042b38ec14ad9211e40 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8cf2bbda7f804e5f9866d663e4a4d805 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2973b3e569394671865a0e5077b281a6 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a53866cc6f704025bdab9afb79d36e6a : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d795cee44ab34f6cb027b7d19386ab83 : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce2fb28fe1444d94b6ccfaa3708f82f9 : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e0e96624fa704269a74badf67eed83bf : SUCCESS in 17m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b0ab1d6cf69848efbb2b27cd10d20f59 : SUCCESS in 1h 33m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5365332ae1444a55a25052ce70753fb8 : SUCCESS in 16m 09s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/46f168ef29a14ed28110bf638f540c1f : SUCCESS in 1h 56m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/89792d0cbf314195a06cd3b06419d0cd : SUCCESS in 27m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3af96d28e5784e01a4961fda0c1371fb : SUCCESS in 27m 23s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c704258882dc428285d19ae0f1fa8006 : SUCCESS in 13m 43s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":4},{"id":"064ca89143a1e7decffe934ac24aea9ea8132924","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-01 14:07:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9f3387ba62a37e9cb6b2b0c843cc5da1169833b5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 15:13:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c02f6b8476301a3f152ca1fa9b6d477c47d16e98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 15:15:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e57ed04c82b797502409e9da40dc53378e7d4ea6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 21:42:23.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e2d3ebbfd85945d4acf71fdf689e248b : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ae4f0f80cf4c8291147b54b388a466 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ac2f4a2a033463599e5cd7cf74cbea7 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/389a88868ba34ad0aba0220132ee28fa : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f30c7682fa54b419a805919f2b3dbaf : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a701541bc0e340ddb9802770d188cfaf : SUCCESS in 5m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/06358e640a384a9db2e59827b28a7b61 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/694e0e2e35e0480dbcecd9bd352f151f : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/809422b894294edf823c7f5077a4cfd7 : SUCCESS in 17m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/afdebc1972a7452fa3dd904fa6727d52 : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a56966eab7414b7bb936300e6534a876 : SUCCESS in 15m 09s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/60652a9327504ea1b34cb1641680b032 : RETRY_LIMIT in 1h 21m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c311f605da5148a386340b32244c87cd : SUCCESS in 27m 39s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d119ad9b63824408937cac2244572e37 : SUCCESS in 29m 09s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/16ab88f7b02542ac8630430ed38d0732 : SUCCESS in 13m 39s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":4},{"id":"fc7dabfbf04bc3529d52f9142d58f0576284063a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 07:31:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eb5d8b26d3267cd4f63bd56e0595551d48f2f7e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 09:41:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/890546df8da845a7989769783bdcc7b8 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35c1fb2c0fdc4d4fa3bd06594bd2d777 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec379e0ea1cd4203870367ad79b92369 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a215e41d27e94026adb2be4670890637 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8365fee0e2bc427cab05f24da6efc827 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0956d22313e640aab6c1bf0d762a9c19 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8832439b47a48b3ad161a8e4a7c65f0 : SUCCESS in 5m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5a09d4ebefb84ba385d4a227320df321 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/078f5134ccd846a39dc599acc15b2863 : SUCCESS in 10m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/48ec83d7e4524e5fb83f8bb160b1cef3 : SUCCESS in 17m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/73802e778cd54cdbb9323822cb46167a : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/db8838b6e6b243789e7e566f66bb99e1 : SUCCESS in 18m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8345a908853d45ea848c279d051de01e : SUCCESS in 2h 04m 51s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e7d469172e9a424d82bf7659571b03e9 : SUCCESS in 27m 45s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1f68756b2bea44a2a28663a74cd0fc33 : SUCCESS in 30m 44s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a0aaa9ac2cfe4db5870a48922b95447b : SUCCESS in 13m 09s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"011a205701ccfcb79b56f6d8485c03978dc4d276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 09:42:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"49b1adcad56aac4ce7f01767a67815aafe18c7e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:47:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/54f9faaf91844506b3763611305c4618 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bec4a3b6d0d04cf9bfb43f21bd3bdf20 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f131454c54743ab9c92798a8793c89f : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e9925a954534f6daed6a9ea5b9d461f : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b221a3d9f4104b91bf017280329f3601 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c60890f2bd7b4f3a9aaa5fbbd14c1b5d : SUCCESS in 5m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a055b7ea2c84c10a0fe2a27ea23d1fa : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/38b4ce32d4f5490e893b7f83f3b56e7a : SUCCESS in 8m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bd3b12589328482fbab37053a4aa1144 : SUCCESS in 17m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/dee4185d522b4d439122149445101e63 : SUCCESS in 1h 14m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/297ad728d9684c4b83a86150a97befa1 : SUCCESS in 16m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cbe098593cc24a1caf7f2333fa381d18 : SUCCESS in 2h 00m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/589ab362efbb436ea19532c9907cbc6c : SUCCESS in 36m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7d36edcebd1841fb858b1ad2640763bf : SUCCESS in 29m 25s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c1cb246a9e794910910b5965e18e1a50 : SUCCESS in 11m 34s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"d60f0b9105ed6e542aa82cdb2707da066df951c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:47:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"688f0cb8ee9d8f2fc9d7ce5a02ebbec41f2d7285","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:50:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4895c17a8f254657b2194919368006a6 : SUCCESS in 2m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c9a9cb981ce4c39b4fb3d88713892d3 : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e74d562948cc37e35cbe910a82d4778cea965a89","revisions":{"ab711e584bbaa746d67c80ca57a84a840b4ae2d0":{"kind":"REWORK","_number":1,"created":"2021-09-30 07:46:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/811784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/811784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/811784/1"}}},"commit":{"parents":[{"commit":"31af98f2b01ff57f4c61311398442ace4e4a832d","subject":"Use ansible_facts to reference facts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31af98f2b01ff57f4c61311398442ace4e4a832d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-18 13:51:44.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 07:46:41.000000000","tz":0},"subject":"Make setup module arguments configurable","message":"Make setup module arguments configurable\n\nAnsible facts can have a large impact on the performance of the Ansible\ncontrol host. This patch introduces some control over which facts are\ngathered (kayobe_ansible_setup_gather_subset) and which facts are stored\n(kayobe_ansible_setup_filter). By default we do not change the default\nvalues of these arguments to the setup module. The flexibility of these\narguments is limited, but they do provide enough for a large performance\nimprovement in a typical moderate to large OpenStack cloud.\n\nIn particular, the large complex dict fact for each interface has a\nlarge effect, and on an OpenStack controller or hypervisor there may be\nmany virtual interfaces. We can use the kayobe_ansible_setup_filter\nvariable to help:\n\n    kayobe_ansible_setup_filter: \u0027ansible_[!qt]*\u0027\n\nThis causes Ansible to collect but not store facts matching that\npattern, which includes the virtual interface facts. Currently we are\nnot referencing other facts matching the pattern within Kayobe.\nNote that including the \u0027ansible_\u0027 prefix causes meta facts module_setup\nand gather_subset to be filtered, but this seems to be the only way to\nget a good match on the interface facts. To work around this, we use\nansible_facts rather than module_setup to detect whether facts exist in\nthe cache.\n\nThe exact improvement will vary, but has been reported to be as large as\n18x on systems with many virtual interfaces.\n\nThis change also introduces a new command to gather facts for Kayobe \u0026\nKolla Ansible on demand, \u0027kayobe overcloud facts gather\u0027. This can be\nused to populate a fact cache.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/794610\nStory: 2007993\nTask: 42586\n\nChange-Id: I5ce3c734433e1682ee942867505468c57440e689\n(cherry picked from commit 2648f48746fa01a5215e1fd9c79a0e7821414841)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab711e584bbaa746d67c80ca57a84a840b4ae2d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab711e584bbaa746d67c80ca57a84a840b4ae2d0"}]},"branch":"refs/heads/stable/ussuri"},"c931ba88242bd95a3a0b9822e5da0cfac373a41c":{"kind":"REWORK","_number":2,"created":"2021-09-30 08:50:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/811784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/811784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/811784/2"}}},"commit":{"parents":[{"commit":"31af98f2b01ff57f4c61311398442ace4e4a832d","subject":"Use ansible_facts to reference facts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31af98f2b01ff57f4c61311398442ace4e4a832d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-18 13:51:44.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 08:50:00.000000000","tz":0},"subject":"Make setup module arguments configurable","message":"Make setup module arguments configurable\n\nAnsible facts can have a large impact on the performance of the Ansible\ncontrol host. This patch introduces some control over which facts are\ngathered (kayobe_ansible_setup_gather_subset) and which facts are stored\n(kayobe_ansible_setup_filter). By default we do not change the default\nvalues of these arguments to the setup module. The flexibility of these\narguments is limited, but they do provide enough for a large performance\nimprovement in a typical moderate to large OpenStack cloud.\n\nIn particular, the large complex dict fact for each interface has a\nlarge effect, and on an OpenStack controller or hypervisor there may be\nmany virtual interfaces. We can use the kayobe_ansible_setup_filter\nvariable to help:\n\n    kayobe_ansible_setup_filter: \u0027ansible_[!qt]*\u0027\n\nThis causes Ansible to collect but not store facts matching that\npattern, which includes the virtual interface facts. Currently we are\nnot referencing other facts matching the pattern within Kayobe.\nNote that including the \u0027ansible_\u0027 prefix causes meta facts module_setup\nand gather_subset to be filtered, but this seems to be the only way to\nget a good match on the interface facts. To work around this, we use\nansible_facts rather than module_setup to detect whether facts exist in\nthe cache.\n\nThe exact improvement will vary, but has been reported to be as large as\n18x on systems with many virtual interfaces.\n\nThis change also introduces a new command to gather facts for Kayobe \u0026\nKolla Ansible on demand, \u0027kayobe overcloud facts gather\u0027. This can be\nused to populate a fact cache.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/794610\nStory: 2007993\nTask: 42586\n\nChange-Id: I5ce3c734433e1682ee942867505468c57440e689\n(cherry picked from commit 2648f48746fa01a5215e1fd9c79a0e7821414841)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c931ba88242bd95a3a0b9822e5da0cfac373a41c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c931ba88242bd95a3a0b9822e5da0cfac373a41c"}]},"branch":"refs/heads/stable/ussuri"},"ff9840ee0468234ad82aab708b0905d918a0d9d1":{"kind":"REWORK","_number":3,"created":"2021-09-30 14:08:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/811784/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/811784/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/811784/3"}}},"commit":{"parents":[{"commit":"31af98f2b01ff57f4c61311398442ace4e4a832d","subject":"Use ansible_facts to reference facts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31af98f2b01ff57f4c61311398442ace4e4a832d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-18 13:51:44.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-30 14:08:44.000000000","tz":0},"subject":"Make setup module arguments configurable","message":"Make setup module arguments configurable\n\nAnsible facts can have a large impact on the performance of the Ansible\ncontrol host. This patch introduces some control over which facts are\ngathered (kayobe_ansible_setup_gather_subset) and which facts are stored\n(kayobe_ansible_setup_filter). By default we do not change the default\nvalues of these arguments to the setup module. The flexibility of these\narguments is limited, but they do provide enough for a large performance\nimprovement in a typical moderate to large OpenStack cloud.\n\nIn particular, the large complex dict fact for each interface has a\nlarge effect, and on an OpenStack controller or hypervisor there may be\nmany virtual interfaces. We can use the kayobe_ansible_setup_filter\nvariable to help:\n\n    kayobe_ansible_setup_filter: \u0027ansible_[!qt]*\u0027\n\nThis causes Ansible to collect but not store facts matching that\npattern, which includes the virtual interface facts. Currently we are\nnot referencing other facts matching the pattern within Kayobe.\nNote that including the \u0027ansible_\u0027 prefix causes meta facts module_setup\nand gather_subset to be filtered, but this seems to be the only way to\nget a good match on the interface facts. To work around this, we use\nansible_facts rather than module_setup to detect whether facts exist in\nthe cache.\n\nThe exact improvement will vary, but has been reported to be as large as\n18x on systems with many virtual interfaces.\n\nThis change also introduces a new command to gather facts for Kayobe \u0026\nKolla Ansible on demand, \u0027kayobe overcloud facts gather\u0027. This can be\nused to populate a fact cache.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/794610\nStory: 2007993\nTask: 42586\n\nChange-Id: I5ce3c734433e1682ee942867505468c57440e689\n(cherry picked from commit 2648f48746fa01a5215e1fd9c79a0e7821414841)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff9840ee0468234ad82aab708b0905d918a0d9d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff9840ee0468234ad82aab708b0905d918a0d9d1"}]},"branch":"refs/heads/stable/ussuri"},"e74d562948cc37e35cbe910a82d4778cea965a89":{"kind":"REWORK","_number":4,"created":"2021-10-01 08:08:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/811784/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/811784/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/811784/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/811784/4"}}},"commit":{"parents":[{"commit":"31af98f2b01ff57f4c61311398442ace4e4a832d","subject":"Use ansible_facts to reference facts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31af98f2b01ff57f4c61311398442ace4e4a832d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-18 13:51:44.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-01 08:08:24.000000000","tz":0},"subject":"Make setup module arguments configurable","message":"Make setup module arguments configurable\n\nAnsible facts can have a large impact on the performance of the Ansible\ncontrol host. This patch introduces some control over which facts are\ngathered (kayobe_ansible_setup_gather_subset) and which facts are stored\n(kayobe_ansible_setup_filter). By default we do not change the default\nvalues of these arguments to the setup module. The flexibility of these\narguments is limited, but they do provide enough for a large performance\nimprovement in a typical moderate to large OpenStack cloud.\n\nIn particular, the large complex dict fact for each interface has a\nlarge effect, and on an OpenStack controller or hypervisor there may be\nmany virtual interfaces. We can use the kayobe_ansible_setup_filter\nvariable to help:\n\n    kayobe_ansible_setup_filter: \u0027ansible_[!qt]*\u0027\n\nThis causes Ansible to collect but not store facts matching that\npattern, which includes the virtual interface facts. Currently we are\nnot referencing other facts matching the pattern within Kayobe.\nNote that including the \u0027ansible_\u0027 prefix causes meta facts module_setup\nand gather_subset to be filtered, but this seems to be the only way to\nget a good match on the interface facts. To work around this, we use\nansible_facts rather than module_setup to detect whether facts exist in\nthe cache.\n\nThe exact improvement will vary, but has been reported to be as large as\n18x on systems with many virtual interfaces.\n\nThis change also introduces a new command to gather facts for Kayobe \u0026\nKolla Ansible on demand, \u0027kayobe overcloud facts gather\u0027. This can be\nused to populate a fact cache.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/794610\nStory: 2007993\nTask: 42586\n\nChange-Id: I5ce3c734433e1682ee942867505468c57440e689\n(cherry picked from commit 2648f48746fa01a5215e1fd9c79a0e7821414841)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e74d562948cc37e35cbe910a82d4778cea965a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e74d562948cc37e35cbe910a82d4778cea965a89"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
