)]}'
{"id":"openstack%2Fkayobe~812271","triplet_id":"openstack%2Fkayobe~master~If09c4840bbe51168a8b94be0db85e6610c21aa82","project":"openstack/kayobe","branch":"master","topic":"story/2009270","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-10-19 12:34:27.000000000","reason":"Change was submitted"},"10910":{"account":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"last_update":"2021-10-19 12:34:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If09c4840bbe51168a8b94be0db85e6610c21aa82","subject":"Ubuntu systemd-networkd: VLAN ifname heuristics","status":"MERGED","created":"2021-10-03 21:13:44.000000000","updated":"2021-10-19 12:36:04.000000000","submitted":"2021-10-19 12:34:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812271-story/2009270","meta_rev_id":"fed1050ae19005ba86d5d1f6137ea6f4750851c8","_number":812271,"virtual_id_number":812271,"owner":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-19 12:34:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-19 09:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-19 10:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-19 10:25:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-03 23:12:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-04 09:18:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-19 10:25:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"2bd242bc34d506198204396f1ce8eafc3c2d71e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2021-10-03 21:13:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb17aabc4be6a24488e09d8ae08e4967dacbd1d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 23:12:29.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46db5cfd803d4c29bc45c518e68b6356 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/490469948462497fb45b2af1d6127646 : FAILURE in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0914baec8954ba988b5e364280edd41 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40226a7b41bf46f0a66d566378618c05 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e94d99382714c35af211d3758693806 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ff860ce21a4d55825e71df7d6fb15b : SUCCESS in 12m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c036098c462645a2b67d2b386ef8a39a : SUCCESS in 6m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b03d66949dcf4bf497b79b630ec360a4 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a7ba18db716e4469bd488a28ceb110c4 : SUCCESS in 26m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5aeadf4a4c9e4738a0a6bdb266b09b6c : SUCCESS in 1h 03m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09e8c478ef764f51b775e06964f788f7 : FAILURE in 8m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b065a44f675c4170a5df34a80321e4ab : SUCCESS in 1h 10m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e7900b08a8fd4c1a825385b43aa72004 : SUCCESS in 18m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b716b56902d42778d4960dbd971997b : FAILURE in 11m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4276de3e05a34934a9062c115f555fed : SUCCESS in 1h 57m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7c7b642ba1004a49988226f29bd3a9f5 : SUCCESS in 36m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47a216fbcca04d2cafbc92e8de319a06 : FAILURE in 10m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7ecc46dbbaa9470ab670a377df937cff : SUCCESS in 29m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9a2b78e1315a46c48f19b465d6b79ecc : SUCCESS in 14m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1be11a53e62442df8d9581939a710abb : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4fde2732619aa3126577d942a3788681e5aa9244","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 09:18:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f136ad978f6e70a9988deb5cf1668c63bb3c8ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 11:24:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5806c7a8e0d343b0a2015f2759335696 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0442fe8bb94e6fa3d4f1a260dafb5c : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b41c0ce37a24433af54e26bf83e7e69 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18e48a08452a4ba9b1b3905027e04097 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da09c96401fc425aa6ddba8a6c86ab3c : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f952ed18f848be88db80d8766429da : SUCCESS in 9m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85fc6572bbbc40889a8407706ef028fa : SUCCESS in 7m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73ee47062e654a03bcd75881394205cc : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b208286482a74495bcae07e6cc303b75 : SUCCESS in 31m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/71b4b6bedb1b4f949625938e56e9245a : SUCCESS in 1h 11m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f5aa0e7c3004df9bf29c3e15272b5a7 : SUCCESS in 52m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f67ee0ebb49c4d7c807652cbd8c16769 : SUCCESS in 1h 19m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ad141e4a6c93448d936084b468160c09 : SUCCESS in 23m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1050ec6754e45fe800277e5cfa1cdfd : SUCCESS in 15m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/67cf52102078455a8060180b2d24be6c : SUCCESS in 2h 04m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2087c5ce54ac4991a05552808098ffa7 : SUCCESS in 47m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b452c363cd2942fabcb15f78d006e898 : SUCCESS in 39m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a017928f7c464eaca12e89db98fde1ca : SUCCESS in 39m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c7d3abefb3834688a9adae3eda25fef9 : SUCCESS in 15m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/732af7e7a7de45c7953095f12a8c7852 : SUCCESS in 14m 36s","accounts_in_message":[],"_revision_number":2},{"id":"5410e3230e85190965c6c8f3a48abeff8223f099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 11:49:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ac8ed8f7350dc125de51f89b0e8355c74216674","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 13:46:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/183b248f768a4af6ad7a4f638ad9dbf5 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27c03a5b681f4be9a10907b02465526f : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5103b75ae494892999662afa5b2bed1 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5614aaee76924713a48a91d9f79c35f7 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/546d719deffd42cbac7d19817fa985be : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c5f6c2b020a4e1caa2c4ce8c2b08e92 : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/607f49ec38c946f99b5c176cc21caf5d : SUCCESS in 4m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/54e1c8b689c5440bba96451baf115bfd : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1adf595c4e4145e3980572dff463b6c5 : SUCCESS in 12m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5ea4448dee045449c4d2c46d057a13a : SUCCESS in 33m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9437e3e04a8747b7a6710c84e0fb399b : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1e103858e3e417ab47c4199e33be126 : SUCCESS in 46m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9499b341f6824ab99e75f04ce9749c67 : SUCCESS in 1h 09m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/216f707727104cfea78f7a25bf9751c4 : SUCCESS in 23m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/904451ed075f4bd2a12803b8754c995d : SUCCESS in 17m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9fffbf1344b64c64b6c24260fdbdd3a7 : SUCCESS in 1h 55m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc3c4e65abfe4be288a9eb93f2a6f620 : SUCCESS in 1h 26m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3c14b894619f4dd49ce916a2cccfd895 : SUCCESS in 38m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca41afb276aa4cbcaf10c9e47774e129 : SUCCESS in 31m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e5e3221932794ff5835f7fc32ab1af70 : SUCCESS in 31m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3d5ae266a134be7acb0a81e35c908eb : SUCCESS in 26m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6ac09e65d6f14219b81ae69c905d455f : SUCCESS in 16m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a31ea87af7d1473f91aa82789420ce20 : SUCCESS in 14m 13s","accounts_in_message":[],"_revision_number":3},{"id":"2e2b0cb8c6c2d1cb5a20da4a43568633a72b9d9b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-19 09:58:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"758230d22d9003ea208ffd497e8eec3f71800e0e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-19 10:25:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"73f9c480b3963ad8f9d9797e70d30c008a553b1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 10:26:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5cef55a1767f1487e19c50f9982a8e8bdc032154","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 12:34:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d171decc5acc4353abaab1e0cbb5a6b5 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d953e553a9542e1afc0080ed3bebce2 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2fee16caf014a49b59e12108258461e : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de001b80bd1f4d61a0b8e23d8c82e5ff : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1c63ea42a994da984af5754e7749402 : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/acf6d857afbf48fab688ae5c61e8efd8 : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/33c18faf1c19454a983a38e57cf885c2 : SUCCESS in 12m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0105bcef5bc34650aa762bae6c165350 : SUCCESS in 30m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/99f5cdbb933a40a6b62a2d610b47f25e : SUCCESS in 1h 02m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d245dec961a54af69ae4b6ce756b927d : SUCCESS in 48m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c6cf3004e9894523bd51a412ddf804a4 : SUCCESS in 1h 41m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0abe4edd81604c838955e78bc526f2a5 : SUCCESS in 20m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38ea464edaa04465b8ed1eeeee1ac7cc : SUCCESS in 16m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3980cf1d25474e15868f01ea43140859 : SUCCESS in 2h 04m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cbf4c0bf3bb84bf9bef1bb62a6f48acc : SUCCESS in 34m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70791f753c9040f3bb7e664e459499d5 : SUCCESS in 36m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cc996fc48cc240c19aa64c15ea5936f9 : SUCCESS in 31m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5f53df3f0af24eee9f1d144187dc7686 : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43aa30c2ba6b46ef956b953c82873ed1 : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":3},{"id":"2732deacd978eb7c85ec7a41cddb1415e1a432dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 12:34:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fed1050ae19005ba86d5d1f6137ea6f4750851c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 12:36:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2042324a837e4d47a682af5efaea3619 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df91ff642fec48a3924246fd7abd17f2 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1b647acaf650a0ed593d222f453e722867d9757c","revisions":{"274f9e73278b6f24fc65b345368a2f904cd4c835":{"kind":"REWORK","_number":1,"created":"2021-10-03 21:13:44.000000000","uploader":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"ref":"refs/changes/71/812271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/812271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/812271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/812271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/812271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/812271/1"}}},"commit":{"parents":[{"commit":"b8abf706d4d8a40f09e7e8c343710bbf4d5404d9","subject":"Build and deploy source type images by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8abf706d4d8a40f09e7e8c343710bbf4d5404d9"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2021-10-03 20:58:52.000000000","tz":60},"committer":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2021-10-03 20:58:52.000000000","tz":60},"subject":"Ubuntu systemd-networkd: VLAN ifname heuristics","message":"Ubuntu systemd-networkd: VLAN ifname heuristics\n\nAdd a regular expression match to verify that VLAN\ninterfaces are named like VLAN interfaces, enabling\nnon-VLAN interfaces to exist on a network where the\nVLAN ID is defined.\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\nStory: 2009270\nTask: 43511\nChange-Id: If09c4840bbe51168a8b94be0db85e6610c21aa82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/274f9e73278b6f24fc65b345368a2f904cd4c835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/274f9e73278b6f24fc65b345368a2f904cd4c835"}]},"branch":"refs/heads/master"},"307e14ede523e841600d4126e997635a3f227c05":{"kind":"REWORK","_number":2,"created":"2021-10-04 09:18:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/812271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/812271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/812271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/812271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/812271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/812271/2"}}},"commit":{"parents":[{"commit":"b8abf706d4d8a40f09e7e8c343710bbf4d5404d9","subject":"Build and deploy source type images by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8abf706d4d8a40f09e7e8c343710bbf4d5404d9"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2021-10-03 20:58:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-04 09:18:21.000000000","tz":60},"subject":"Ubuntu systemd-networkd: VLAN ifname heuristics","message":"Ubuntu systemd-networkd: VLAN ifname heuristics\n\nAdd a regular expression match to verify that VLAN\ninterfaces are named like VLAN interfaces, enabling\nnon-VLAN interfaces to exist on a network where the\nVLAN ID is defined.\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\nStory: 2009270\nTask: 43511\nChange-Id: If09c4840bbe51168a8b94be0db85e6610c21aa82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/307e14ede523e841600d4126e997635a3f227c05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/307e14ede523e841600d4126e997635a3f227c05"}]},"branch":"refs/heads/master"},"1b647acaf650a0ed593d222f453e722867d9757c":{"kind":"REWORK","_number":3,"created":"2021-10-04 11:49:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/812271/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/812271/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/812271/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/812271/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/812271/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/812271/3"}}},"commit":{"parents":[{"commit":"b8abf706d4d8a40f09e7e8c343710bbf4d5404d9","subject":"Build and deploy source type images by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8abf706d4d8a40f09e7e8c343710bbf4d5404d9"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2021-10-03 20:58:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-04 11:49:27.000000000","tz":60},"subject":"Ubuntu systemd-networkd: VLAN ifname heuristics","message":"Ubuntu systemd-networkd: VLAN ifname heuristics\n\nAdd a regular expression match to verify that VLAN\ninterfaces are named like VLAN interfaces, enabling\nnon-VLAN interfaces to exist on a network where the\nVLAN ID is defined.\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\nStory: 2009270\nTask: 43511\nChange-Id: If09c4840bbe51168a8b94be0db85e6610c21aa82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b647acaf650a0ed593d222f453e722867d9757c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b647acaf650a0ed593d222f453e722867d9757c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
