)]}'
{"id":"openstack%2Fkayobe~812301","triplet_id":"openstack%2Fkayobe~master~I478d6304c3f8d207bc7062a0909147a169003027","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-10-20 13:07:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I478d6304c3f8d207bc7062a0909147a169003027","subject":"CI: enable DNF tests on CentOS Stream 8","status":"MERGED","created":"2021-10-04 08:53:23.000000000","updated":"2021-10-20 13:10:33.000000000","submitted":"2021-10-20 13:07:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812301","meta_rev_id":"0ffff8c19b2bd7015ff732ef684e9aedf930df5f","_number":812301,"virtual_id_number":812301,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-20 13:07:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-20 10:58:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-05 15:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-20 10:58:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-04 10:53:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-05 15:54:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-10-20 10:58:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"714652db32d572d3a57827390341be2f2590091b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-04 08:53:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04a08bc0f887f45b8e196f08effe84085fc206a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 10:53:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a1ac5ced4de4f6caa9d83ab1cb3789d : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a4244b06a7b441ea6ec5cfb032aa2f2 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c074d001e864459993e9841a3b80b2a : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c60fcbdd135a4cc99a6a3604b136bd93 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf3fe9013a714785a059e1192b73d94d : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c41c64a4254a2eade0a813e2a88fc6 : SUCCESS in 10m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19cbeffdfdf94272aab7f0486c87fb29 : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e9b79bddf5e74bdb81fe6c18fb146394 : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bd96f39673a747d6a88d013d9287d37d : SUCCESS in 30m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/725159d40bd5448ca9f39c1bf93a82b4 : SUCCESS in 1h 12m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/759d4266cd2f46fda4309750b1f5628b : SUCCESS in 50m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/77696e3c59254e21907af5a2c3891059 : SUCCESS in 1h 12m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e6f37c556ebe42c899853d77154ec8bb : SUCCESS in 20m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0be93648a02c44ae883eb7f7361c07c0 : SUCCESS in 15m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dabe6fc49dff49f58a2145abd01ec711 : SUCCESS in 1h 58m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a4d2c94417aa4938a24142404b8745d9 : SUCCESS in 35m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbc6203e5bf64ef48dc8736179d7f08a : SUCCESS in 56m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9e16498f390d40859cf1d90230f6fa6c : SUCCESS in 31m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c923b747b8294f39a76173cef4fbc40f : SUCCESS in 15m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9884326de72d4f91a6400e6e0965346f : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"71f28c1b5a7c5aa799396429ca4d07ccbb0d0aa6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-05 15:54:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2fec080e6a419b74454a7862261f1af9259473a3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-20 10:58:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1ea15ef78d3098137ddbad3e809414aa2378088","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 10:59:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d83ffb8a18ee84378a112ad47a1f3e86a078033d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 13:07:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2249a3a027d6426abf974c7b51ea8364 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd94af39434d42aca3291d42899f15f4 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c0a99d2674f440480c29ddde085c808 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c854d4f05040929d6e97b4be8a9fad : SUCCESS in 9m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a2bf1d98bc3f4fb1b138b1576841e4a6 : SUCCESS in 6m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/999a663eb3e546c18e40d313274ac5ef : SUCCESS in 14m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/99e96dc69f2a4b21a759eeab46f309ae : SUCCESS in 30m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b83ed1ed1d8d457cb3ae4b38ef2425d2 : SUCCESS in 1h 07m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bb8ce0e6dbb44e49b7b2cb74fc203e4 : SUCCESS in 48m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/53f4f00d86ff4c26b0d2aa403f6e259e : SUCCESS in 1h 10m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a1b2786acc2b401d8f027a7320c3342e : SUCCESS in 24m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df525b895bde481c83a86be10ad9e80d : SUCCESS in 16m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/55b0d7b329154c7695454e19b39355e9 : SUCCESS in 2h 03m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/165f9412c3a44e4fad962eb7b9d1fd5a : SUCCESS in 40m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e19e2e8424004f0c962cabf5b22ef7bb : SUCCESS in 33m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a655095344004fc585aed5889df20963 : SUCCESS in 33m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/96a1adc4d2ad4261a2e4e4dfc74e34f0 : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f73380413d80469588d15080b7971b3d : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f62de5817254afa871b0c3f207056d3adf21857c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 13:07:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0ffff8c19b2bd7015ff732ef684e9aedf930df5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 13:10:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71aad847c054a25af3fda9e98c23806 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a69a4f4db9f66978d9b593b0f510b1046fa969b","revisions":{"6a69a4f4db9f66978d9b593b0f510b1046fa969b":{"kind":"REWORK","_number":1,"created":"2021-10-04 08:53:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/812301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/812301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/812301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/812301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/812301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/812301/1"}}},"commit":{"parents":[{"commit":"f09faa43d1a1d38b30967cce948af6b60c34be70","subject":"CI: Disable libvirt debug logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09faa43d1a1d38b30967cce948af6b60c34be70"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-04 08:48:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-04 08:51:43.000000000","tz":60},"subject":"CI: enable DNF tests on CentOS Stream 8","message":"CI: enable DNF tests on CentOS Stream 8\n\nThese were inadvertently skipped following the move to CentOS Stream,\ndue to matching the distro name on CentOS Linux.\n\nBackport to Victoria.\n\nChange-Id: I478d6304c3f8d207bc7062a0909147a169003027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a69a4f4db9f66978d9b593b0f510b1046fa969b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a69a4f4db9f66978d9b593b0f510b1046fa969b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
