)]}'
{"id":"openstack%2Fkayobe~812698","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Iea1daff70fca5cf49f4e7f44af71a900678bb5c9","project":"openstack/kayobe","branch":"stable/wallaby","topic":"support_apt_proxy-stable/wallaby","hashtags":[],"change_id":"Iea1daff70fca5cf49f4e7f44af71a900678bb5c9","subject":"Add support for apt proxy setting","status":"MERGED","created":"2021-10-06 19:16:39.000000000","updated":"2021-10-08 02:13:31.000000000","submitted":"2021-10-08 02:11:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812698-support_apt_proxy-stable/wallaby","cherry_pick_of_change":799629,"cherry_pick_of_patch_set":13,"meta_rev_id":"9b64f9e373c0af19ec4b7b62044f2049d10446d7","_number":812698,"virtual_id_number":812698,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-08 02:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-07 09:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-07 09:32:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-07 09:32:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33694,"name":"Skylar Tristan Kelty","display_name":"Skylar Kelty","email":"skylar.kelty@kcl.ac.uk","username":"skylarkelty"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33694,"name":"Skylar Tristan Kelty","display_name":"Skylar Kelty","email":"skylar.kelty@kcl.ac.uk","username":"skylarkelty"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-06 19:16:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":33694,"name":"Skylar Tristan Kelty","display_name":"Skylar Kelty","email":"skylar.kelty@kcl.ac.uk","username":"skylarkelty"},"state":"REVIEWER"},{"updated":"2021-10-06 21:06:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-07 09:21:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-07 09:32:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-10-07 12:41:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"6ca7f164fd83e68b3f0123059c599174a46a637b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-06 19:16:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ca6164d44b5b23c4da160479d74bb275427c03ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 21:06:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74e1eada29b64ab9924537cb81cb5538 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/741e6130db8c4325a1df3a77f90c0164 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6035781200541949c7022ee3c350fd2 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1026f0c8cfab45efb69d1ff9f0d1a394 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9e5e3d7dd5e4ce59a7c3364762e634a : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9c1d97b27b646f990d1ce8fbc162077 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d66639e917e481e941befc5c86d1038 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/52d99aaf69044001bed7faa589d530cb : SUCCESS in 7m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c78100eb3454b6e9dfbdfb01470a65c : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a630e9b77d42427694b995702c6191df : SUCCESS in 19m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7696947e2d7341e1bbd90e15fbb13890 : SUCCESS in 1h 03m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e70edbf3e08e42a79062757340de4d6b : SUCCESS in 47m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7e170c523d35454692cb8ed67af89bca : SUCCESS in 1h 00m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1424580264734779b07aa661d00c9c31 : SUCCESS in 16m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88da78f9b19840f7a481a97a804521eb : SUCCESS in 15m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/779ac4e68e6b40b0904c7c7d2e5bd1e2 : SUCCESS in 1h 46m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a9d755e9432e41b18696fbaeaea18b17 : SUCCESS in 39m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70438f8696c14372b76728451617d369 : SUCCESS in 27m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/aa1012ffe7f84e85b792f7f3ec31791a : SUCCESS in 26m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dee2d19c67294c63936accd70e354709 : SUCCESS in 14m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c59708e0bcd451b8f6783334e0075a6 : SUCCESS in 11m 54s","accounts_in_message":[],"_revision_number":1},{"id":"14d42d5d03dac6837e4beed69e7d79c2857be1e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-07 09:21:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"639470e313b598b1e5c871b66cd61c56d95261b4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-07 09:32:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e85f6d562bb799f59772429d6991fbc277570c4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 09:38:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43676dc32788af865154599a020b0bc49812e1f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 12:40:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5aa503278c843858a117912def8a233 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a35aa17f9a854f20a302543cef267585 : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed826149cf0041bea0058fb3536f0506 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c00a386aef64fc086862a8b01fc4f5b : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef191ecdf455469789a5328c9a1278eb : SUCCESS in 5m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/263519e0b55b4fdba14a42f32299aedd : SUCCESS in 7m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a7c33bc2deda4fa99c28b18a0efa7423 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/24d2ee531cd34f9b98385fdcd80284a9 : SUCCESS in 18m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c9ac51e001ed4424bd5473c62936345b : SUCCESS in 1h 11m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ed5cc0effcc4349979269623fbaf6ce : SUCCESS in 46m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/28f6668843ea4634b2f84bb442fc792e : SUCCESS in 1h 16m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/922536746a3e4ecca2deafb5822c7cad : SUCCESS in 20m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/198520b142054c7a8947775482de94c4 : SUCCESS in 20m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3032e3272d604556a198b3b3b7007fa4 : SUCCESS in 1h 46m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3a7a87b61dc545f0878c042b23d29dd0 : SUCCESS in 38m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96c32e5089ac4d8085591ee664b007c6 : SUCCESS in 33m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6590d057039c4519bdb44030a089457e : FAILURE in 20m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/561651afcc5c47e9a7bce0ddc4fe884d : SUCCESS in 16m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f8a6221e25644839b20153a2f56b425 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"04feaaea6a4b0c30c71c3e38b13e073642852bb2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-07 12:41:54.000000000","message":"Patch Set 1:\n\nrecheck (mirror issue)","accounts_in_message":[],"_revision_number":1},{"id":"ba408d44a33bf93af6b5da6f751274cfd1645559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 14:34:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3682101aa6e4e54a9326e3b9f61ed05 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcd6770ff2614859a951591aa45ec1cb : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0adce260b28140809ed308e4b487213e : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f371574dd1a42528c1d3d4fa0ad3fc2 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5be4b3a9db494ba2b42f8bada341fc54 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/406efc81ab0449e2a5077f9590af8874 : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44afb39e05d34d7b8cd6a89ad4f0666b : SUCCESS in 4m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1dd4da0dae9b483ebc8cc85e4cfc2797 : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fe7a5fabbe3e4baa860f30c6190a691c : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0b6d4d8e3af1493bbdcf4233d99aaeb8 : SUCCESS in 19m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/873e273e7ac54444b2e07a034f0fa0fc : SUCCESS in 56m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4263de834ca4abd88205a9413b0b4c8 : SUCCESS in 50m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/662f9cda49934ac089cdc13b7a03a18e : SUCCESS in 1h 13m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3a745ecad7a0470d8d78421f26f85b1f : SUCCESS in 18m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7202a8f8a60845bcb23ac2231020d609 : SUCCESS in 14m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/972c08d3138c4b039ed86992bdf6443f : SUCCESS in 1h 51m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/83181505bd854a859643e879a9dc353a : SUCCESS in 35m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e15425f6d82443d1b6f81385d5c9acad : SUCCESS in 26m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e777feea2ea942868dbdb5f897750c44 : SUCCESS in 30m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a30ee8934e094c9fa4bde00d6e0927a0 : SUCCESS in 14m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7295ec1d9b3a49dba0d5def67f538f21 : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"98aae1835ef63a7153748062fb79d822338f8ace","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 14:35:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3f0aebcad7a4bdfca4b924dae883bcaba284f38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 17:30:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/040da6e498af40e7bbd6c64869262efc : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bba7da431354e01987123a13b44d599 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e7ed554a13445828c928d01d2e8cbfd : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb09bca240e04d789ca28f3d1fd20dfa : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1905babc030b4c1192b8af8a6abe5448 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a05b01fadc24e6f9fbeea9d5ec9920b : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/addf37cd0450405e86c35ab57fc4915b : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/24c615f637ae43c5b3fc987671530a33 : SUCCESS in 18m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4256d077d2fa43b1b520ce7a6e84ccb6 : SUCCESS in 1h 07m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf1bb425f0df47b7877c6f6b6dfdc6c9 : SUCCESS in 47m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/460f8e51da2d445780347bedb5547ee1 : SUCCESS in 1h 08m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/65cd994f0bd947799e82978c41ec9532 : SUCCESS in 20m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f8f4707f28d47c78d66f40ec9c8e9cc : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c1e8dc67dcdd4c62b59228d66a7a7379 : FAILURE in 1h 28m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9d16e730f35f4c24be74f880628109b3 : SUCCESS in 34m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/756bc3f0cf084ff2a5a846a0eff51e83 : SUCCESS in 35m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e9ff511c2f68472b9c8ee105f3153908 : SUCCESS in 30m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2536d04d00fe495f8756d421980cc616 : SUCCESS in 15m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c887c48f0224d5da197090ffb23b6c5 : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3e10477391e52ec28c6799783d8625ffc190fe4a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-07 20:08:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f9d691ec8f377d53efdef545e4c05dcdc1d85c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 22:43:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcc380da357f4f2ea3922e038756f440 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c3fff1dcb92436da1146f73ece069b9 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b99c9a8a236d404780d07ea31b9d8cfd : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a8f0a967c2445deb2a5725b2e432c5a : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da570ff599d6458bbe8923721b7668a8 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b1a4515972b40cca7aa2688d10e64be : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1eb1ac270bcd4e0699ebaba9703b4c9a : SUCCESS in 6m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d0ac313068434d00884f453adf9eaf5e : SUCCESS in 8m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b8dbbb0ae0948cf8d497a9953a6b5e9 : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/29b227772ccc46b78beab15eac4f82d0 : SUCCESS in 18m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0e9e932ef65c470d9b47c8a27b89ad47 : SUCCESS in 1h 08m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9c95e74a8e348d9b0ff1c48966e6491 : SUCCESS in 50m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9f10882993bf4e278c17917f77e24b76 : SUCCESS in 1h 07m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8c01383b4b7d4d70988b198795368589 : SUCCESS in 23m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a2031248d044e9e9d09dbd751d4d76d : SUCCESS in 15m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/96c220eae0044ee1ba232fd06cb88b7f : SUCCESS in 1h 43m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/86db60648efe4ba4bf11c378ecba2aa6 : SUCCESS in 42m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9dee5e53eb740afb8a4bfe31334b506 : SUCCESS in 33m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/343c14a04304407195df907acb02241c : SUCCESS in 25m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1f89d47043604f079fd7874a851a80f5 : SUCCESS in 18m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92c2f29992004a08aa8abcf9f93e881f : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a64e30e7ed1ec468b762a50153cc3c2366cfe9b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 22:43:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e28b629287c031be5528a6bdca1494acc056957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 02:11:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09ae0fe4d20e45f7b260944a59e526c4 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c2529bd10f642dea6818c5a58523a94 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/128503ef2e7647ba9807dc64d0975f28 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d5b68ccdae04c8d8499d3ad609bf5f0 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3884149a33694498bc052f7b508bf3ba : SUCCESS in 5m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0b87c1157ff849bf944bf4c73354f467 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ec2e78da21574833b2edadd583cc1d18 : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a0cb355f2491406a87b5ec0858b85784 : SUCCESS in 18m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/998bc6edf252499484ba3758982d1bda : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f84a0b2b2e04020aea72d447dc1ccc3 : SUCCESS in 49m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/432ebe7595f642f4a3ae8eed93b80f6d : SUCCESS in 1h 10m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/64d5b1f3e72e431abe6702e29791d167 : SUCCESS in 21m 46s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b87875b258b460cb7e43e27e74b6c28 : SUCCESS in 14m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8844cc242bdc45a7b7fd3bacde6a3645 : SUCCESS in 1h 46m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/350365e3ec504c80afc231fcaaa65272 : SUCCESS in 41m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c79f06ba6c149ea9bc220ed921af91d : SUCCESS in 35m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dc386154abed46c7a554999c002d2d8b : SUCCESS in 30m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0589489f5a914ea4bf3219dc6b1cc091 : SUCCESS in 14m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65e6d1b734b04d2fae81dfe335223fec : SUCCESS in 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3d1ba7f29dd52e1b5aa7a979fbe0fac081a1ec3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 02:11:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9b64f9e373c0af19ec4b7b62044f2049d10446d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 02:13:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d71da42eda4521bfcead538ae14aa7 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35b3c4292500482093ba0d8b41c7e65a : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2164c6b2226777c7712c4f2ece82293b64a1ef52","revisions":{"2164c6b2226777c7712c4f2ece82293b64a1ef52":{"kind":"REWORK","_number":1,"created":"2021-10-06 19:16:39.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/98/812698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/812698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/812698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/812698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/812698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/812698/1"}}},"commit":{"parents":[{"commit":"0854692a7ec1f9605891ae316354709bb0c18ae0","subject":"CI: Fix bashate ignores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0854692a7ec1f9605891ae316354709bb0c18ae0"}]}],"author":{"name":"Skylar Kelty","email":"skylar.kelty@kcl.ac.uk","date":"2021-07-06 10:32:46.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-06 19:16:39.000000000","tz":0},"subject":"Add support for apt proxy setting","message":"Add support for apt proxy setting\n\nAdd support for configuring apt\u0027s proxy setting on Ubuntu hosts.\n\nChange-Id: Iea1daff70fca5cf49f4e7f44af71a900678bb5c9\nStory: 2009035\nTask: 42782\n(cherry picked from commit f24b3176ebb80d77b1a954a828100a2eded2e7ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2164c6b2226777c7712c4f2ece82293b64a1ef52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2164c6b2226777c7712c4f2ece82293b64a1ef52"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
