)]}'
{"id":"openstack%2Fkayobe~812947","triplet_id":"openstack%2Fkayobe~master~I5d49279fa9a2a263d84c6f468c00d4c8538c538c","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"34076":{"account":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"last_update":"2021-10-11 11:25:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d49279fa9a2a263d84c6f468c00d4c8538c538c","subject":"Update Kolla images namespace to openstack.kolla","status":"MERGED","created":"2021-10-07 10:21:25.000000000","updated":"2021-10-11 11:28:27.000000000","submitted":"2021-10-11 11:25:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812947","meta_rev_id":"2b7e7c2be4449a7bc011fbf383e5303b993c4988","_number":812947,"virtual_id_number":812947,"owner":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-11 11:25:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-08 14:21:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-08 14:22:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-10-11 09:20:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 10:30:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-10-07 12:46:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-07 13:58:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-08 14:21:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"888095399348ba78c3ad169f6226a8ad8fe2f115","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2021-10-07 10:21:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79d1b28bd9325b4a24e4a7ee79eb36dc5b8360bb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-07 10:30:22.000000000","message":"Patch Set 1:\n\nOops, good catch.","accounts_in_message":[],"_revision_number":1},{"id":"67122a093e8f723e74aabe093dd5d63ba94b035c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 12:46:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/485ea1626c6a41899343592b5908636e : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b7ac6405e5c456196f3ce270822cc9d : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91f3fcf022e34f318539e6fca7447d8f : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23a54da973654d399973d95930396056 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f54f5cc62bfa4239ba9201e86d05702b : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5869788db28d4a9387c346eead8ddb2d : SUCCESS in 9m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef140b4ae8b84f1f948da5fa06f8b19c : SUCCESS in 7m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1103fe6832d24b2c847d3b107a00d487 : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c9217811f2f948ceafead138ab02479b : SUCCESS in 29m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4959198bbe9c4c2ba0ba952db60116a0 : SUCCESS in 1h 02m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27096d88aef447f9b0650bff6f02604b : SUCCESS in 54m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c3395ad005f9487d9a923ab68c51b160 : SUCCESS in 1h 15m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5b53acc425ff41ac81744e4552601f6e : SUCCESS in 20m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/460b3144ec9743aa9c87abe8f31af968 : SUCCESS in 16m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f65efd6b3d784040ae26cc9ad97bd786 : SUCCESS in 2h 14m 10s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eb56439249e43d6adcc2240f5c1e5d9 : SUCCESS in 1h 23m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d6fd821d9d2942cca0470287de79a3e7 : SUCCESS in 38m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f71ec12db13459cb01300395a741087 : SUCCESS in 32m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4004175510c24f3fa5d111b88da799f2 : SUCCESS in 37m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/484d6b70c46f4518b7ca54031b83f9a2 : SUCCESS in 29m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f7e7082d7227411791d0be04860c42d8 : SUCCESS in 13m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f3159b2ba4f4578919320be4b1fb8fd : SUCCESS in 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8da1b6fe0bf35666af253d5534f44665e94062a2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-07 13:58:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c07bb59be28dcddc91ec7a47972936855d563302","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2021-10-07 14:13:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3d50ba3eab5b6ff62a5b5a9afbb7537d43ee170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2021-10-07 14:14:56.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"762644eb6015535e78a8fa32569f7c2bf19a89ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 16:36:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c8518ebad5e4395894d0ae766e13c1a : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/462ff85a157b479a9f9187b2caa6f9c1 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72f1de2c96c047748c6518ff93daa039 : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0482b065071493d8012244f4e0edd6d : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f9edab7caee43739a7f90695b73b13b : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466cbfb2bff04366b454f63da6f539db : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5ee675cd93d49848c391093687f2e58 : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6deba3f3359f4a7b80734bef7693bfa8 : SUCCESS in 7m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f40e6edb85346d3b412b2dda68cc51d : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad671b778633481180c54a2f8bbc65e2 : SUCCESS in 30m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/44da9a5f67fb45e1adbaec22bc26b664 : SUCCESS in 1h 06m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0ed79d3de934fd28f0ac0c9bb0cec12 : SUCCESS in 52m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8c6de83e39d64d21b17d680499f85e4e : SUCCESS in 1h 11m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8fd9a53a1acc495d9ccd8b80776ef086 : SUCCESS in 22m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bf957ceb90a4b8eb22ebce9d8ad615f : SUCCESS in 15m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5d3de3b5ba3c4ee799a2eadf65d00f53 : SUCCESS in 2h 13m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e37a39240a9411aa9895d23570abe54 : SUCCESS in 1h 25m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/153bd0bfc2c04d4bb92a3c51228feac5 : SUCCESS in 47m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48a0dc11aad040b783e56ad296a7eb22 : SUCCESS in 29m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b5c4539cc37043cc9be0e0dc78483bab : SUCCESS in 30m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74bed1d97eda40ddb1df523ab91568c3 : SUCCESS in 24m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e59d0b6f50c64dbc88e85dd2413b2ce9 : SUCCESS in 16m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1f25cdee5d04ae7887dc7fa13e08267 : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":3},{"id":"456ec13f81fadb5fc57fa27c36e2a754c364da9a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-08 14:21:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3223c6a04719ec4d8e115680135adefe21c3137c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-08 14:22:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"58e77e2a8cc4c527fc46b5561fc6f2c36ee1e19a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-11 09:20:13.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ed848db72e7e17af1a8dba98a4686bf26dd0ae1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 09:20:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1e689accd7ea873865b51fcc020bea64d8d44ec6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 11:25:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e884a2ecc1342b08adb7c829b6d1e64 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40a997560c8d4750994c8bd0bcc76dba : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bac4b7b6ea946f39e5e0251cbdb8550 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc57b1c20d4f497dab01b4b2b8b4a36f : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9949a555785460daf4655a99faf940a : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ad74a651015427caaf62dc8d358a891 : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3601726505b0485c80e363ffe660e1a2 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/789375d38a354362a1f0725ccbe6343f : SUCCESS in 30m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/70defde6229b496081e834e524651929 : SUCCESS in 1h 08m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/944535a8f8324ca48645533fa0bfbc9f : SUCCESS in 44m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2a0128c1a996472686fda5ab1f66197c : SUCCESS in 1h 11m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ed334e73e9834432ad90be9af97ab54a : SUCCESS in 22m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d739ae2530674c14ba55a1c42a34e433 : SUCCESS in 16m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1707f0d0a300439ea7c58c61e04ed953 : SUCCESS in 1h 56m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7e115e4bcc3a4d3789992d050d8162e6 : SUCCESS in 35m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06d26c8692af49c8aa766be70c981eb0 : SUCCESS in 32m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/10b6862faceb49aab8eb2aea22b9783a : SUCCESS in 29m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/672818d252c941f7a6b52bd3ebb96cf0 : SUCCESS in 15m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb7f8a22a6b54805a7ec3920ef879209 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":3},{"id":"f628b73fa7c2022cff3d254a5c43a65b1c48fcbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 11:25:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2b7e7c2be4449a7bc011fbf383e5303b993c4988","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 11:28:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40cb958aaa44f4aaf628de56aaf58c4 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f28bbb1be96f421b9e18e5a06dc0ce7d : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b1b6358f7976372187ac7d5329607c49ad0481d8","revisions":{"2fe67d59e302444e5132fe23dc099b685403403f":{"kind":"REWORK","_number":1,"created":"2021-10-07 10:21:25.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/47/812947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/812947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/812947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/812947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/812947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/812947/1"}}},"commit":{"parents":[{"commit":"9b93b504a3d518bdc4bd6bfe49d913675cf00786","subject":"Merge \"Fix IPA builder version typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b93b504a3d518bdc4bd6bfe49d913675cf00786"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2021-10-07 10:14:24.000000000","tz":0},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2021-10-07 10:21:23.000000000","tz":0},"subject":"Update Kolla images Docker namespace to openstack.kolla","message":"Update Kolla images Docker namespace to openstack.kolla\n\nDocker namespace for Kolla images is changed from kolla\nto openstack.kolla to align with If5a26b1ba4bf35bc29306c24f608396dbf5e3371\n\nChange-Id: I5d49279fa9a2a263d84c6f468c00d4c8538c538c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fe67d59e302444e5132fe23dc099b685403403f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fe67d59e302444e5132fe23dc099b685403403f"}]},"branch":"refs/heads/master"},"72896919916c4c4e0a699714e9d210395737769f":{"kind":"REWORK","_number":2,"created":"2021-10-07 14:13:29.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/47/812947/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/812947/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/812947/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/812947/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/812947/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/812947/2"}}},"commit":{"parents":[{"commit":"9b93b504a3d518bdc4bd6bfe49d913675cf00786","subject":"Merge \"Fix IPA builder version typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b93b504a3d518bdc4bd6bfe49d913675cf00786"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2021-10-07 10:14:24.000000000","tz":0},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2021-10-07 14:09:28.000000000","tz":0},"subject":"Update Kolla images Docker namespace to openstack.kolla","message":"Update Kolla images Docker namespace to openstack.kolla\n\nDocker namespace for Kolla images is changed from kolla\nto openstack.kolla to align with If5a26b1ba4bf35bc29306c24f608396dbf5e3371\n(Added to reno)\n\nChange-Id: I5d49279fa9a2a263d84c6f468c00d4c8538c538c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72896919916c4c4e0a699714e9d210395737769f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72896919916c4c4e0a699714e9d210395737769f"}]},"branch":"refs/heads/master"},"b1b6358f7976372187ac7d5329607c49ad0481d8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-10-07 14:14:56.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/47/812947/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/812947/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/812947/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/812947/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/812947/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/812947/3"}}},"commit":{"parents":[{"commit":"9b93b504a3d518bdc4bd6bfe49d913675cf00786","subject":"Merge \"Fix IPA builder version typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b93b504a3d518bdc4bd6bfe49d913675cf00786"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2021-10-07 10:14:24.000000000","tz":0},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2021-10-07 14:14:32.000000000","tz":0},"subject":"Update Kolla images namespace to openstack.kolla","message":"Update Kolla images namespace to openstack.kolla\n\nDocker namespace for Kolla images is changed from kolla\nto openstack.kolla to align with If5a26b1ba4bf35bc29306c24f608396dbf5e3371\n(Added to reno)\n\nChange-Id: I5d49279fa9a2a263d84c6f468c00d4c8538c538c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1b6358f7976372187ac7d5329607c49ad0481d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1b6358f7976372187ac7d5329607c49ad0481d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
